)]}'
{"id":"openstack%2Fovsdbapp~729363","triplet_id":"openstack%2Fovsdbapp~stable%2Fstein~I9c7994d50066adf02b0066cb8efb1beb61d2e1f0","project":"openstack/ovsdbapp","branch":"stable/stein","hashtags":[],"change_id":"I9c7994d50066adf02b0066cb8efb1beb61d2e1f0","subject":"Add indexed column support to ovsdbapp","status":"ABANDONED","created":"2020-05-19 19:19:17.000000000","updated":"2020-11-02 20:47:49.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"ca1dc895070983be89409395eac7ddb2328a6f3c","_number":729363,"virtual_id_number":729363,"owner":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2020-05-31 08:35:34.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-05-19 21:43:33.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"all":[{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":-1,"date":"2020-06-01 03:36:47.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-19 21:43:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-05-31 08:35:34.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"}],"messages":[{"id":"8340f77919d0a84caae48d0f7b5c6943d6620e71","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-05-19 19:19:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"38c6b036d291fe3169a0814bb3a3af39667aab9c","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-05-19 19:29:47.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"9fcc6e6f01a4ca7be28aa91789cb6a23532f4935","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-05-19 19:47:44.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"33180ea6a48ad7cb9c77c060681d281106255587","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-19 21:43:33.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3ba8b8372e25482386178164ed10ef9c : SUCCESS in 4m 28s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d648782fe469438d84bdce5ad96a4c3d : SUCCESS in 5m 21s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/70a1d44f49a04203a88a13556504f80e : SUCCESS in 4m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8f67dc9a4db9417ab6be9a9134a49303 : SUCCESS in 4m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0e83659ea7c64d9ea9de327ff78d07b5 : SUCCESS in 7m 29s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/957acec91d8e481ab43fb24209c8091b : SUCCESS in 7m 03s\n- ovsdbapp-tempest-dsvm-networking-ovn-ovs-release https://zuul.opendev.org/t/openstack/build/71c65903f94745a399d47f04e1c9ea59 : SUCCESS in 1h 54m 10s","accounts_in_message":[],"_revision_number":3},{"id":"b11b85bc1d01e269323523b51e4a182921c7d53b","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-05-31 08:35:34.000000000","message":"Patch Set 3:\n\nplease do not approve until https://bugs.launchpad.net/neutron/+bug/1881424 is resolved without need to revert","accounts_in_message":[],"_revision_number":3},{"id":"7b4e05f6984d46597a88521041a1c3df7b459791","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-06-01 03:36:47.000000000","message":"Patch Set 3: Workflow-1","accounts_in_message":[],"_revision_number":3},{"id":"ca1dc895070983be89409395eac7ddb2328a6f3c","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-11-02 20:47:49.000000000","message":"Abandoned\n\nnot going to backport this to stein, even disabled by default","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"5525338bec4f6287771e21e106df4f4440ac4132","revisions":{"32070b63da5a7fef5cca1d752f1cc20cbb20bc5a":{"kind":"REWORK","_number":1,"created":"2020-05-19 19:19:17.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/63/729363/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/63/729363/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/63/729363/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/63/729363/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/63/729363/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/63/729363/1"}}},"commit":{"parents":[{"commit":"1cb33fa2d04d565683911fcbd4731e585bd5d5b3","subject":"Add \"inactivity_probe\" parameter to \"enable_connection_uri\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/1cb33fa2d04d565683911fcbd4731e585bd5d5b3"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-05-07 21:34:38.000000000","tz":0},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-05-19 19:19:01.000000000","tz":-300},"subject":"Add indexed column support to ovsdbapp","message":"Add indexed column support to ovsdbapp\n\nAdd support for python-ovs multi-column indexes. This is currently\ndisabled by default, though in a simple test of creating 10k\nLogical_Switches the time went from \"I give up after 10 mins\" to\n4 seconds. The only reason I didn\u0027t enable it by default was in\ncase users want to tailor which columns they index.\n\nThere are slight API changes w/ Backend.ovsdb_connection\nproperties added and start_connection becoming an instance method\ninstead of a classmethod. But since ovsdb_connection was the\nitem being stored on the class and it still is via an overridable\nproperty, users shouldn\u0027t be broken by the change.\n\nThe backport disables auto_index by default\n\nDepends-On: https://review.opendev.org/#/c/727232/\nChange-Id: I9c7994d50066adf02b0066cb8efb1beb61d2e1f0\n(cherry picked from commit db0cdd9a4ffd64d99a6b012b5d0bf2c5bd4f96f4)\n(cherry picked from commit 766414a23d78edc09bd152771a11d9107ca494ff)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/32070b63da5a7fef5cca1d752f1cc20cbb20bc5a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/32070b63da5a7fef5cca1d752f1cc20cbb20bc5a"}]},"branch":"refs/heads/stable/stein"},"6cde4067140650045913dad42ca54b6c9acb3b02":{"kind":"REWORK","_number":2,"created":"2020-05-19 19:29:47.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/63/729363/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/63/729363/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/63/729363/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/63/729363/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/63/729363/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/63/729363/2"}}},"commit":{"parents":[{"commit":"1cb33fa2d04d565683911fcbd4731e585bd5d5b3","subject":"Add \"inactivity_probe\" parameter to \"enable_connection_uri\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/1cb33fa2d04d565683911fcbd4731e585bd5d5b3"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-05-07 21:34:38.000000000","tz":0},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-05-19 19:29:35.000000000","tz":-300},"subject":"Add indexed column support to ovsdbapp","message":"Add indexed column support to ovsdbapp\n\nAdd support for python-ovs multi-column indexes. This is currently\ndisabled by default, though in a simple test of creating 10k\nLogical_Switches the time went from \"I give up after 10 mins\" to\n4 seconds. The only reason I didn\u0027t enable it by default was in\ncase users want to tailor which columns they index.\n\nThere are slight API changes w/ Backend.ovsdb_connection\nproperties added and start_connection becoming an instance method\ninstead of a classmethod. But since ovsdb_connection was the\nitem being stored on the class and it still is via an overridable\nproperty, users shouldn\u0027t be broken by the change.\n\nThe backport disables auto_index by default\n\nDepends-On: https://review.opendev.org/#/c/727232/\nChange-Id: I9c7994d50066adf02b0066cb8efb1beb61d2e1f0\n(cherry picked from commit db0cdd9a4ffd64d99a6b012b5d0bf2c5bd4f96f4)\n(cherry picked from commit 766414a23d78edc09bd152771a11d9107ca494ff)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/6cde4067140650045913dad42ca54b6c9acb3b02"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/6cde4067140650045913dad42ca54b6c9acb3b02"}]},"branch":"refs/heads/stable/stein"},"5525338bec4f6287771e21e106df4f4440ac4132":{"kind":"REWORK","_number":3,"created":"2020-05-19 19:47:44.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/63/729363/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/63/729363/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/63/729363/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/63/729363/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/63/729363/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/63/729363/3"}}},"commit":{"parents":[{"commit":"1cb33fa2d04d565683911fcbd4731e585bd5d5b3","subject":"Add \"inactivity_probe\" parameter to \"enable_connection_uri\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/1cb33fa2d04d565683911fcbd4731e585bd5d5b3"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-05-07 21:34:38.000000000","tz":0},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-05-19 19:47:35.000000000","tz":-300},"subject":"Add indexed column support to ovsdbapp","message":"Add indexed column support to ovsdbapp\n\nAdd support for python-ovs multi-column indexes. This is currently\ndisabled by default, though in a simple test of creating 10k\nLogical_Switches the time went from \"I give up after 10 mins\" to\n4 seconds. The only reason I didn\u0027t enable it by default was in\ncase users want to tailor which columns they index.\n\nThere are slight API changes w/ Backend.ovsdb_connection\nproperties added and start_connection becoming an instance method\ninstead of a classmethod. But since ovsdb_connection was the\nitem being stored on the class and it still is via an overridable\nproperty, users shouldn\u0027t be broken by the change.\n\nThe backport disables auto_index by default\n\nDepends-On: https://review.opendev.org/#/c/727232/\nChange-Id: I9c7994d50066adf02b0066cb8efb1beb61d2e1f0\n(cherry picked from commit db0cdd9a4ffd64d99a6b012b5d0bf2c5bd4f96f4)\n(cherry picked from commit 766414a23d78edc09bd152771a11d9107ca494ff)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/5525338bec4f6287771e21e106df4f4440ac4132"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/5525338bec4f6287771e21e106df4f4440ac4132"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
