)]}'
{"id":"openstack%2Fovsdbapp~729969","triplet_id":"openstack%2Fovsdbapp~stable%2Fstein~I17c24fd88adecbea09c71741c0105df1d0240e66","project":"openstack/ovsdbapp","branch":"stable/stein","topic":"bug/1871104-stable/stein","hashtags":[],"change_id":"I17c24fd88adecbea09c71741c0105df1d0240e66","subject":"Accept a Row as a result in AddCommand","status":"MERGED","created":"2020-05-21 14:10:49.000000000","updated":"2020-05-21 18:35:24.000000000","submitted":"2020-05-21 18:35:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"729969-1590086124252-2e69b6b6","meta_rev_id":"b82663af4a94e77a15bc1d3aa4eee87a25beedd0","_number":729969,"virtual_id_number":729969,"owner":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-21 18:35:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2020-05-21 16:24:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"all":[{"value":0,"date":"2020-05-21 15:03:17.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2020-05-21 16:24:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-21 15:03:17.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2020-05-21 18:35:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e6ea305876342716567c62f021d375171f83fe73","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-05-21 14:10:49.000000000","message":"Patch Set 1: Cherry Picked from branch stable/train.","accounts_in_message":[],"_revision_number":1},{"id":"39021a36c96dab563cdd5f1647ce350346c3618e","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-05-21 14:11:00.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/rocky as commit ba9ece12823de487353887c1da6c76c19165dfea","accounts_in_message":[],"_revision_number":1},{"id":"784f790b1e9af14d5944a7f082aaaa5934344f12","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-21 15:45:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/60643fa8dcf54aee866b4dba55a3cebc : SUCCESS in 4m 11s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/664f0d743d2848b1b7210dff4664799b : SUCCESS in 7m 36s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/f31b04c722bb410c846a220acc9569c4 : SUCCESS in 4m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/38f8d6a5d40545c7aa153c45608a79db : SUCCESS in 5m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d9a7f4c540ba4f069a904214bb08e708 : SUCCESS in 7m 11s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/1fe1017e66a2408aae40c5205a41cd8d : SUCCESS in 9m 46s\n- ovsdbapp-tempest-dsvm-networking-ovn-ovs-release https://zuul.opendev.org/t/openstack/build/9ecda22cdab9414cb52d7dbc80119f46 : SUCCESS in 1h 26m 47s","accounts_in_message":[],"_revision_number":1},{"id":"82ee92f23d7ae3b70517aac87cb3b848c47220b8","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-05-21 16:24:05.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nninja merging because I need to make releases and this is identical to the train which merged a while ago and tests are passing.","accounts_in_message":[],"_revision_number":1},{"id":"f4ec97d31cdbb97adae570d74888a394722eabb7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-21 16:24:18.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c1ec13b754ff8abba6bf91d7f3563e20a38def86","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-21 18:35:24.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f8fa8c06781b4fba9df8db1aff301023 : SUCCESS in 4m 54s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4c2559f7a8ce42dd93c8ae8727cab498 : SUCCESS in 7m 38s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/010dafb104ec4e3fbd8e03362f00d971 : SUCCESS in 4m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7b717b0ae95f4d2e9996c323e901fa66 : SUCCESS in 5m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/477fb31c55ad4a6da03b9320dd0b0f59 : SUCCESS in 7m 04s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/aa031092476e4e18bcf33d0f7edd6030 : SUCCESS in 6m 35s\n- ovsdbapp-tempest-dsvm-networking-ovn-ovs-release https://zuul.opendev.org/t/openstack/build/13496146f6374e97a864465879042457 : SUCCESS in 1h 17m 43s","accounts_in_message":[],"_revision_number":1},{"id":"33ee915ef4d553c488e1236d4265fa1647aec625","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-21 18:35:24.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1e538a2d714d0a1ae8944450f07a024c98ca73ec","revisions":{"1e538a2d714d0a1ae8944450f07a024c98ca73ec":{"kind":"REWORK","_number":1,"created":"2020-05-21 14:10:49.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/69/729969/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/69/729969/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/69/729969/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/69/729969/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/69/729969/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/69/729969/1"}}},"commit":{"parents":[{"commit":"1cb33fa2d04d565683911fcbd4731e585bd5d5b3","subject":"Add \"inactivity_probe\" parameter to \"enable_connection_uri\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/1cb33fa2d04d565683911fcbd4731e585bd5d5b3"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-04-07 15:52:20.000000000","tz":-300},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-05-21 14:10:49.000000000","tz":0},"subject":"Accept a Row as a result in AddCommand","message":"Accept a Row as a result in AddCommand\n\nMake AddCommand a little more lenient and accept Row/RowView objs\nas results.\n\nConflicts:\n      ovsdbapp/backend/ovs_idl/command.py\n\nChange-Id: I17c24fd88adecbea09c71741c0105df1d0240e66\nRelated-Bug: 1871104\n(cherry picked from commit 0338c0bb5cd6368edc14bff8e64f980d4a6286ac)\n(cherry picked from commit fd0629f140fe34d24b567ec6d6a10a833f826ce9)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/1e538a2d714d0a1ae8944450f07a024c98ca73ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/1e538a2d714d0a1ae8944450f07a024c98ca73ec"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
