)]}'
{"id":"openstack%2Fovsdbapp~729970","triplet_id":"openstack%2Fovsdbapp~stable%2Frocky~I17c24fd88adecbea09c71741c0105df1d0240e66","project":"openstack/ovsdbapp","branch":"stable/rocky","topic":"bug/1871104-stable/stein-stable/rocky","hashtags":[],"change_id":"I17c24fd88adecbea09c71741c0105df1d0240e66","subject":"Accept a Row as a result in AddCommand","status":"MERGED","created":"2020-05-21 14:11:00.000000000","updated":"2020-05-21 18:35:24.000000000","submitted":"2020-05-21 18:35:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"729970-1590086124948-7d9b3224","meta_rev_id":"9eff68d0dfac0948ee63f048335113b378b27a1e","_number":729970,"virtual_id_number":729970,"owner":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-21 18:35:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2020-05-21 16:24:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"all":[{"value":0,"date":"2020-05-21 15:03:20.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2020-05-21 16:24:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-21 15:03:20.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2020-05-21 18:35:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"016d8f458062cc465c37225ec66a5a6a5571e95a","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-05-21 14:11:00.000000000","message":"Patch Set 1: Cherry Picked from branch stable/stein.","accounts_in_message":[],"_revision_number":1},{"id":"0f7cd440dad81a0bdd248c9441bf9aa9506d9108","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-05-21 14:11:08.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/queens as commit cdf619e7aeee817f74db6fff55d085c7fda4728f","accounts_in_message":[],"_revision_number":1},{"id":"cb6f1041ae61d13044b9baf58b7ca9ebb591ca93","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-21 15:50:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/98bb0a96ab134f4888410f6032fc25c0 : SUCCESS in 6m 12s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8909109393ed4349a6d34e4d7ad8a03e : SUCCESS in 5m 05s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/9401bb4a603b4970b667581dc89f2d25 : SUCCESS in 7m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d90177adc3894ef9b52ed4a1085ddc54 : SUCCESS in 8m 53s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/c021673ca0f64356aada65efbc979906 : SUCCESS in 7m 12s\n- ovsdbapp-tempest-dsvm-networking-ovn-ovs-release https://zuul.opendev.org/t/openstack/build/ed05e2ae84774ef1bbd1202c07c1c62d : SUCCESS in 1h 25m 43s","accounts_in_message":[],"_revision_number":1},{"id":"f6fb0d0365e18105af3dd5723336273432e2f128","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-05-21 16:24:40.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nninja merging because I need to make releases and this is identical to the train patch which merged a while ago and tests are passing.","accounts_in_message":[],"_revision_number":1},{"id":"205cda5c64d23dfb70df425c2f8f2ed470061fa0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-21 16:24:52.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ea83b26a6b2874d0c26e904710ef5b26724b9c66","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-21 18:35:24.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/12f3365aea7d463bacd147222a3b1a6a : SUCCESS in 4m 45s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/43e6f5710c294a89b8d213d032ec7c74 : SUCCESS in 6m 34s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/b074a67fb8a24f77b2a815c8430f752b : SUCCESS in 5m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a64f863f80974a12bc8f9887af2eaa44 : SUCCESS in 7m 48s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/e30819b19b394d89991c5b47faf5323f : SUCCESS in 7m 46s\n- ovsdbapp-tempest-dsvm-networking-ovn-ovs-release https://zuul.opendev.org/t/openstack/build/03f32ab4d14b46d3bc52dc902962e711 : SUCCESS in 1h 30m 47s","accounts_in_message":[],"_revision_number":1},{"id":"a126e5a0f3577533f4c363e77c89ad78063c81b6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-21 18:35:24.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ba9ece12823de487353887c1da6c76c19165dfea","revisions":{"ba9ece12823de487353887c1da6c76c19165dfea":{"kind":"REWORK","_number":1,"created":"2020-05-21 14:11:00.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/70/729970/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/70/729970/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/70/729970/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/70/729970/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/70/729970/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/70/729970/1"}}},"commit":{"parents":[{"commit":"1bc0f9f1715a9051492948314aa69259e822768a","subject":"Add \"inactivity_probe\" parameter to \"enable_connection_uri\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/1bc0f9f1715a9051492948314aa69259e822768a"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-04-07 15:52:20.000000000","tz":-300},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-05-21 14:11:00.000000000","tz":0},"subject":"Accept a Row as a result in AddCommand","message":"Accept a Row as a result in AddCommand\n\nMake AddCommand a little more lenient and accept Row/RowView objs\nas results.\n\nConflicts:\n      ovsdbapp/backend/ovs_idl/command.py\n\nChange-Id: I17c24fd88adecbea09c71741c0105df1d0240e66\nRelated-Bug: 1871104\n(cherry picked from commit 0338c0bb5cd6368edc14bff8e64f980d4a6286ac)\n(cherry picked from commit fd0629f140fe34d24b567ec6d6a10a833f826ce9)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/ba9ece12823de487353887c1da6c76c19165dfea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/ba9ece12823de487353887c1da6c76c19165dfea"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
