)]}'
{"id":"openstack%2Fovsdbapp~730559","triplet_id":"openstack%2Fovsdbapp~master~Idfea423764768d5a35c8b5b8f73286cd4a19a974","project":"openstack/ovsdbapp","branch":"master","topic":"reno-openstackdocstheme","hashtags":[],"change_id":"Idfea423764768d5a35c8b5b8f73286cd4a19a974","subject":"Switch to newer openstackdocstheme and reno versions","status":"MERGED","created":"2020-05-25 06:50:07.000000000","updated":"2020-06-08 14:58:09.000000000","submitted":"2020-06-08 14:55:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"730559-1591628122314-df6bd473","meta_rev_id":"1b363ba204d2f4ae23901f95b2965fef3ff2e2cf","_number":730559,"virtual_id_number":730559,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-08 14:55:22.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":2,"date":"2020-05-25 13:39:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":2,"date":"2020-06-08 13:01:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":1,"date":"2020-05-29 09:00:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":1,"date":"2020-06-08 13:01:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-25 13:39:31.000000000","updated_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"},{"updated":"2020-05-29 09:00:39.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-06-08 13:01:56.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2020-06-08 14:55:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"50862a85b055c35c7c9349fd03aa34ace65bc184","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-05-25 06:50:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3bd81ad889d2c5af14ba1092bde25dcc581f34e9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-25 08:35:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/80f8a4333b5f48078462d89a4fe02d2d : SUCCESS in 2m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6392f31313024b7890058b55f572b9dd : SUCCESS in 4m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f59d8ba08f6a4dfa8dbfc579ef8e616a : SUCCESS in 5m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0322fecb9a81472c8b0bd95b446f83da : SUCCESS in 4m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7b42b46810aa4e559b74b15721134580 : SUCCESS in 4m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2a992329e10f4340b565ece06a41e760 : SUCCESS in 4m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6bb600765dcf4ca7bbb4f7ec033bea5d : SUCCESS in 5m 56s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/bc5ec604d4304e558838cc615d6dcef5 : SUCCESS in 6m 00s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/fb50330e039a42a4b589eab86db92f24 : SUCCESS in 1h 44m 28s","accounts_in_message":[],"_revision_number":1},{"id":"f3fe61623991acd2c1078def0b759b697e5bfcd8","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-05-25 13:27:04.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"20c8ce23859b68ba108ba48695f9644303906a5d","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-05-25 13:33:44.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"a3800131c139da64f241f9d4c9fcfb96b5760121","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-05-25 13:34:02.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b0cf7e2a57e5a73bb2fe9967c8a0a38dfc2452f6","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-05-25 13:39:31.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"c009d432af839f5f2323e07327acb82c78b64dad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-25 15:29:06.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/60ed690a29274df0a3f8c07480aa7391 : SUCCESS in 2m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2dace82d09a140b3bc10f01d10c335f3 : SUCCESS in 3m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/16bc22d996aa4e278695171567afa84b : SUCCESS in 4m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/69dcebc9ce204e98b61d7acd7b2f8d88 : SUCCESS in 3m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0f640c12509d4e61bdbcb3005cbbb4ad : SUCCESS in 4m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9988f7a9b8fc434082141b808ec0e3d0 : SUCCESS in 3m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8fddad864544b00a53ac91aaca86015 : SUCCESS in 5m 03s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/de8b8b21fddb4708833266dfe25a52e9 : SUCCESS in 6m 41s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/1900acea8aa44f9ca560d831746ce4b1 : SUCCESS in 1h 50m 11s","accounts_in_message":[],"_revision_number":2},{"id":"b3238412ce5e8442150bc4e01939b80e8e2f2917","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-05-29 09:00:39.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"8342884d352e337be023ebf5a2c15798c6966c99","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-06-08 13:01:56.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"584d4b3840e7118bd0a75af53164f24ea90e85f7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-08 13:02:07.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"8759fda6651e0b4fb6ad2b03786db30c11cdd382","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-08 14:55:22.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"a32dccb92048b679346e5826e94e226ce87373d4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-08 14:55:22.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/5c4dd0081db5477a92ec9b55435b5851 : SUCCESS in 3m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dae544bed5f1434e8446c3bb548f8d25 : SUCCESS in 4m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/63f98f7bbc7543e4a85c78a59833c657 : SUCCESS in 4m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/36bdd251fd4142eead011d5366f0d78d : SUCCESS in 4m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/40f5858a0b6547578626b8a652bf502e : SUCCESS in 4m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f5e322cdaf1492aba74362e7a84be7e : SUCCESS in 5m 57s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/8ebc83e764f249ccbc3d418f0b65055d : SUCCESS in 6m 59s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/77158a7293bf4ce59eeb2ba31385d657 : SUCCESS in 1h 51m 47s","accounts_in_message":[],"_revision_number":2},{"id":"b14d324a3e475bce5062eca1a2003298fb0a8f0c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-08 14:58:09.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/13edd071aa934f1d8992df08f0b34501 : SUCCESS in 2m 22s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ac54f148f05beb50585f4b76984cff802126406d","revisions":{"f532e9d9869681e5a154d8b4657bc3102484578d":{"kind":"REWORK","_number":1,"created":"2020-05-25 06:50:07.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/59/730559/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/59/730559/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/59/730559/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/59/730559/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/59/730559/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/59/730559/1"}}},"commit":{"parents":[{"commit":"7736bac7dc093abbcb6990cf84d46c2fb6beeb9d","subject":"Merge \"Add indexed column support to ovsdbapp\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/7736bac7dc093abbcb6990cf84d46c2fb6beeb9d"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-05-25 06:47:41.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-05-25 06:49:32.000000000","tz":120},"subject":"Switch to newer openstackdocstheme and reno versions","message":"Switch to newer openstackdocstheme and reno versions\n\nSwitch to openstackdocstheme 2.2.1 and reno 3.1.0 versions. Using\nthese versions will allow especially:\n* Linking from HTML to PDF document\n* Allow parallel building of documents\n* Fix some rendering problems\n\nUpdate Sphinx version as well.\n\nSet openstackdocs_pdf_link to link to PDF file. Note that\nthe link to the published document only works on docs.openstack.org\nwhere the PDF file is placed in the top-level html directory. The\nsite-preview places the PDF in a pdf directory.\n\nDisable openstackdocs_auto_name to use \u0027project\u0027 variable as name.\n\nChange pygments_style to \u0027native\u0027 since old theme version always used\n\u0027native\u0027 and the theme now respects the setting and using \u0027sphinx\u0027 can\nlead to some strange rendering.\n\nRemove docs requirements from lower-constraints, they are not needed\nduring install or test but only for docs building.\n\nopenstackdocstheme renames some variables, so follow the renames\nbefore the next release removes them. A couple of variables are also\nnot needed anymore, remove them.\n\nSee also\nhttp://lists.openstack.org/pipermail/openstack-discuss/2020-May/014971.html\n\nChange-Id: Idfea423764768d5a35c8b5b8f73286cd4a19a974\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/f532e9d9869681e5a154d8b4657bc3102484578d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/f532e9d9869681e5a154d8b4657bc3102484578d"}]},"branch":"refs/heads/master"},"ac54f148f05beb50585f4b76984cff802126406d":{"kind":"REWORK","_number":2,"created":"2020-05-25 13:33:44.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/59/730559/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/59/730559/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/59/730559/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/59/730559/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/59/730559/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/59/730559/2"}}},"commit":{"parents":[{"commit":"7736bac7dc093abbcb6990cf84d46c2fb6beeb9d","subject":"Merge \"Add indexed column support to ovsdbapp\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/7736bac7dc093abbcb6990cf84d46c2fb6beeb9d"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-05-25 06:47:41.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","date":"2020-05-25 13:33:41.000000000","tz":0},"subject":"Switch to newer openstackdocstheme and reno versions","message":"Switch to newer openstackdocstheme and reno versions\n\nSwitch to openstackdocstheme 2.2.1 and reno 3.1.0 versions. Using\nthese versions will allow especially:\n* Linking from HTML to PDF document\n* Allow parallel building of documents\n* Fix some rendering problems\n\nUpdate Sphinx version as well.\n\nSet openstackdocs_pdf_link to link to PDF file. Note that\nthe link to the published document only works on docs.openstack.org\nwhere the PDF file is placed in the top-level html directory. The\nsite-preview places the PDF in a pdf directory.\n\nDisable openstackdocs_auto_name to use \u0027project\u0027 variable as name.\n\nChange pygments_style to \u0027native\u0027 since old theme version always used\n\u0027native\u0027 and the theme now respects the setting and using \u0027sphinx\u0027 can\nlead to some strange rendering.\n\nRemove docs requirements from lower-constraints, they are not needed\nduring install or test but only for docs building.\n\nopenstackdocstheme renames some variables, so follow the renames\nbefore the next release removes them. A couple of variables are also\nnot needed anymore, remove them.\n\nSee also\nhttp://lists.openstack.org/pipermail/openstack-discuss/2020-May/014971.html\n\nChange-Id: Idfea423764768d5a35c8b5b8f73286cd4a19a974\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/ac54f148f05beb50585f4b76984cff802126406d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/ac54f148f05beb50585f4b76984cff802126406d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
