)]}'
{"id":"openstack%2Fovsdbapp~731020","triplet_id":"openstack%2Fovsdbapp~master~I0250c7bd51c63fea534c3b6b55bb5bb053637b19","project":"openstack/ovsdbapp","branch":"master","hashtags":[],"change_id":"I0250c7bd51c63fea534c3b6b55bb5bb053637b19","subject":"Add MAC_Binding index lookup","status":"MERGED","created":"2020-05-27 02:44:01.000000000","updated":"2020-07-22 19:26:46.000000000","submitted":"2020-07-22 19:25:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"731020-1595445934478-bdb66bb5","meta_rev_id":"10d98eb90b87a909a03f3e42b401781db24ada42","_number":731020,"virtual_id_number":731020,"owner":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"date":"2020-07-22 16:05:17.000000000","_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-22 19:25:34.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2020-07-22 17:50:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-07-22 16:07:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2020-07-22 17:50:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-22 16:07:36.000000000","updated_by":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"reviewer":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"state":"REVIEWER"},{"updated":"2020-07-22 17:50:21.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2020-07-22 19:25:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d8a9c848d1e1280b8bb2984ce94d6f3c66bac169","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-05-27 02:44:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4e3e661a1a1577ae70e5e0f78736c8c064d3e56c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-27 04:50:01.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6ae5954dd2d14f8d90531011f284186b : SUCCESS in 3m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/744288eab29b4f29ae1506edcdc2e97b : FAILURE in 3m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9a08960ac0d84914a60173ad3fcd882d : SUCCESS in 4m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f5a600c3c5924129a36610f49ba1633e : SUCCESS in 4m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/39254c91c81140a29c9e454c7347392e : SUCCESS in 4m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/350700ff784a4a6986a158351303b31b : SUCCESS in 8m 19s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/a94123a067654148ae05213747ab3314 : SUCCESS in 7m 05s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/932159479f5546d8b456ca80f51e3f2e : SUCCESS in 2h 04m 55s\n\nWarning:\n  Comments left for invalid file ovsdbapp/backend/ovs_idl/idlutils.py","accounts_in_message":[],"_revision_number":1},{"id":"8a5a8123cc0eb8000f64efcd6dbba719db1bd619","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-05-28 15:21:55.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"beb8eda85b18da784676167c8e72a34991fb171e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 17:04:52.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/68c078e1ead740779a283af6a81b5f86 : SUCCESS in 6m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7d4e05d2c4c3454080cc353bcb9095be : SUCCESS in 6m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fe031cfb8e6444ba9bc3aad02786f550 : SUCCESS in 6m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1ed0753583b647d48e191dcc3e490099 : SUCCESS in 7m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c6f159ca420e4c9c9d788dbca10a6b59 : SUCCESS in 5m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6925f43b169349498b0c716ed273eb55 : SUCCESS in 6m 59s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/6634d8ecc6ff4dee903a6b0a73cf8071 : SUCCESS in 7m 33s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/6f65a9e0227342e9aa0f1e967b10042e : SUCCESS in 1h 41m 14s","accounts_in_message":[],"_revision_number":2},{"id":"3eb72bf51e9a81242dd7dc7ed53d1415a56b57e6","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-07-15 13:46:18.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"bc88eb7ca94866963aefcaac23da7982ae5e564a","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-07-15 14:59:09.000000000","message":"Patch Set 3:\n\nHi Terry, why do we need this new value? Just asking.","accounts_in_message":[],"_revision_number":3},{"id":"c8b46089176828ede9b0de4f1e49c23689b65ced","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-15 15:23:04.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1a1bafd0fa824a228d37f4e5a55ba4ac : SUCCESS in 5m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eea3e559d6414462aba804d79738706b : SUCCESS in 5m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/514f7a10ecce40a5ba92cb8ca5ed1a83 : SUCCESS in 6m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1df507cec15748d5ac556816755c467c : SUCCESS in 5m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/60c1d991072649dd9af1529ac2298f7e : SUCCESS in 9m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1522bd90d9b0467b83bc174e614a2857 : SUCCESS in 9m 37s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/06d9c2fd14b84fb89fef4d033859ce85 : SUCCESS in 9m 23s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/f6e75e805d9846c1843d6154352e97ad : SUCCESS in 1h 29m 34s","accounts_in_message":[],"_revision_number":3},{"id":"318602fa37449e3a2b55cc29512c823f7dfdc26f","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-07-22 16:04:48.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"5a7abb453e8d52caef36e9de28b4eb6f0d6694e9","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-07-22 16:05:17.000000000","message":"Patch Set 4:\n\nRodolfo: updated the commit message with the answer to your very good question. :)","accounts_in_message":[],"_revision_number":4},{"id":"db2d18391bd889044297e917540dd5abb30a9730","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2020-07-22 16:07:36.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"b189662b8f9529213a4958dfd0429583af24a97b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-22 17:32:34.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8a0f1c78ccef41e3a1245a0fe52166ca : SUCCESS in 4m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eadfa61eeca84cc2a7d7f78615231570 : SUCCESS in 4m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/925b0ac6f4154f9590444614f2dd834c : SUCCESS in 4m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a799815f915645ed84481954c4ab20f3 : SUCCESS in 6m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/df5462bdd4564ee6a8699b71101adbcf : SUCCESS in 5m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fff3b9321cee4400b37c65a7a70420cf : SUCCESS in 4m 54s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/fa513bcb681544ba8fb79f8176ea2a40 : SUCCESS in 6m 11s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/38b021c2163f48f295ea83f855fbb871 : SUCCESS in 1h 21m 21s","accounts_in_message":[],"_revision_number":4},{"id":"6c84582bfde11e6e4be851b8fcf6566e90b70672","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-07-22 17:50:21.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"c73767d74bb2c9c0ffa27c4f221134f7db951834","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-22 17:50:45.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"b05ce776f05741f9c719e06976ad52da9b845b83","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-22 19:25:34.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1569f1162ad24a5b890e607201f9624b : SUCCESS in 8m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c850c70be90043c1a316e84942220b8b : SUCCESS in 7m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b3fa7f1057934524bd4fd9ef633ec88b : SUCCESS in 3m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a5cc461a00944dca8c6dddeac69451a0 : SUCCESS in 4m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f33f55c263aa465098db2b580f511431 : SUCCESS in 5m 28s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/8d5e4a3bedd0455ebd2d115f4796a914 : SUCCESS in 5m 55s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/0898b7e8629e4af1a40e9800f2c36174 : SUCCESS in 1h 28m 33s","accounts_in_message":[],"_revision_number":4},{"id":"2c3ce9868ad72d554cfd91ba39732dd95bfe8713","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-22 19:25:34.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"10d98eb90b87a909a03f3e42b401781db24ada42","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-22 19:26:46.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ac33c45ccf04c10b45d7c31294375ed : SUCCESS in 52s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"469095b4472fef6d3e2e1f1ef4e86314b90548b0","revisions":{"71ea4064b4cfaa68cce39d7f6b870bb1f17fb5ec":{"kind":"REWORK","_number":1,"created":"2020-05-27 02:44:01.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/20/731020/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/20/731020/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/20/731020/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/20/731020/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/20/731020/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/20/731020/1"}}},"commit":{"parents":[{"commit":"4fd0ce7e8868a76df7a1cd922005e731a8d2527e","subject":"Handle index lookups in DbFindCommand","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/4fd0ce7e8868a76df7a1cd922005e731a8d2527e"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-05-27 02:41:54.000000000","tz":0},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-05-27 02:43:51.000000000","tz":0},"subject":"Add MAC_Binding index lookup","message":"Add MAC_Binding index lookup\n\nChange-Id: I0250c7bd51c63fea534c3b6b55bb5bb053637b19\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/71ea4064b4cfaa68cce39d7f6b870bb1f17fb5ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/71ea4064b4cfaa68cce39d7f6b870bb1f17fb5ec"}]},"branch":"refs/heads/master"},"770729098ff4b596e774e3ec8eb1da3905882553":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-05-28 15:21:55.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/20/731020/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/20/731020/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/20/731020/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/20/731020/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/20/731020/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/20/731020/2"}}},"commit":{"parents":[{"commit":"8a5b23527ee90ccd2592a3565dd43bfc7af3fd9c","subject":"Handle index lookups in DbFindCommand","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/8a5b23527ee90ccd2592a3565dd43bfc7af3fd9c"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-05-27 02:41:54.000000000","tz":0},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-05-28 15:21:55.000000000","tz":0},"subject":"Add MAC_Binding index lookup","message":"Add MAC_Binding index lookup\n\nChange-Id: I0250c7bd51c63fea534c3b6b55bb5bb053637b19\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/770729098ff4b596e774e3ec8eb1da3905882553"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/770729098ff4b596e774e3ec8eb1da3905882553"}]},"branch":"refs/heads/master"},"3841134da86af57f45580c87f4c0c78060bc6400":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-07-15 13:46:18.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/20/731020/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/20/731020/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/20/731020/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/20/731020/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/20/731020/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/20/731020/3"}}},"commit":{"parents":[{"commit":"7bd45d00d3e483dcedd7541ffc87a0310d7663a8","subject":"Merge \"drop mock from lower-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/7bd45d00d3e483dcedd7541ffc87a0310d7663a8"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-05-27 02:41:54.000000000","tz":0},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-07-15 13:46:18.000000000","tz":0},"subject":"Add MAC_Binding index lookup","message":"Add MAC_Binding index lookup\n\nChange-Id: I0250c7bd51c63fea534c3b6b55bb5bb053637b19\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/3841134da86af57f45580c87f4c0c78060bc6400"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/3841134da86af57f45580c87f4c0c78060bc6400"}]},"branch":"refs/heads/master"},"469095b4472fef6d3e2e1f1ef4e86314b90548b0":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2020-07-22 16:04:48.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/20/731020/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/20/731020/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/20/731020/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/20/731020/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/20/731020/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/20/731020/4"}}},"commit":{"parents":[{"commit":"7bd45d00d3e483dcedd7541ffc87a0310d7663a8","subject":"Merge \"drop mock from lower-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/7bd45d00d3e483dcedd7541ffc87a0310d7663a8"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-05-27 02:41:54.000000000","tz":0},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-07-22 16:04:42.000000000","tz":0},"subject":"Add MAC_Binding index lookup","message":"Add MAC_Binding index lookup\n\nThis is required because MAC_Binding does not have a \"simple\" one-column\nindex in the OVSDB schema, so it will not have an autocreated index.\nNeutron code does a db_find of the MAC_Binding entry by the \"ip\" column,\nso adding this entry allows us to do an indexed lookup.\n\nChange-Id: I0250c7bd51c63fea534c3b6b55bb5bb053637b19\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/469095b4472fef6d3e2e1f1ef4e86314b90548b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/469095b4472fef6d3e2e1f1ef4e86314b90548b0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
