)]}'
{"id":"openstack%2Fovsdbapp~743985","triplet_id":"openstack%2Fovsdbapp~stable%2Fstein~Ie1b3b162e18090ae1cc90998b5a41d74b9ef2876","project":"openstack/ovsdbapp","branch":"stable/stein","topic":"pep8-error-stable/ussuri-stable/train-stable/stein","hashtags":[],"change_id":"Ie1b3b162e18090ae1cc90998b5a41d74b9ef2876","subject":"Update sphinx reqs to match global","status":"MERGED","created":"2020-07-30 14:18:50.000000000","updated":"2020-08-05 14:49:53.000000000","submitted":"2020-08-05 14:49:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"743985-1596638993707-88ec3396","meta_rev_id":"02ab1622e137977822b0117388a8fa4768a65940","_number":743985,"virtual_id_number":743985,"owner":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-05 14:49:53.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2020-08-05 13:15:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-08-04 06:55:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2020-08-05 13:15:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-04 06:55:33.000000000","updated_by":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"reviewer":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"state":"REVIEWER"},{"updated":"2020-08-05 13:15:39.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2020-08-05 14:49:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e78a8adb8d70a5d59a266791327c2d6860772de9","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-07-30 14:18:50.000000000","message":"Patch Set 1: Cherry Picked from branch stable/train.","accounts_in_message":[],"_revision_number":1},{"id":"9ab0ca8b8d42d9e442f5fca04d322c709549c20c","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2020-07-30 14:20:35.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d36a44c183bd26fb613fc65588ca6bfad3251dec","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-07-30 14:47:10.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"27c7f2d211e935256a2a0a72a8c1f978c9773001","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-30 17:06:22.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ea26ab0bc46b46d38708363d535c0db6 : SUCCESS in 5m 06s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4e82e7c2d24a4d6bbf39473db891e8e1 : SUCCESS in 3m 31s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/530ba8669f394b9a9b4cd973021a6846 : SUCCESS in 4m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/956bccdab4e04286a47bda55b17c80ce : SUCCESS in 4m 16s\n- requirements-check https://zuul.opendev.org/t/openstack/build/c845e126d43e477c8b22f15a2ae259f8 : FAILURE in 3m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/21bb9655bcab482d8b1279be7501a3cc : SUCCESS in 4m 37s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/4d7ee60f6318473e9e837363383ba6e6 : SUCCESS in 4m 56s\n- ovsdbapp-tempest-dsvm-networking-ovn-ovs-release https://zuul.opendev.org/t/openstack/build/678f12116e844a2a86b54ac9f3047e4d : SUCCESS in 1h 10m 01s","accounts_in_message":[],"_revision_number":1},{"id":"f842dddfa76a2ae1138e829e3b1813c6e88aa5f6","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-08-03 14:03:57.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"6b8ffe32b4ccad1ce8c69f59f599c59f0e97964d","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-08-03 14:53:22.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"ea91a28033f95e7b5b51ec738a7051c9e7e2a460","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2020-08-03 15:01:49.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"cea564d47a86b9c8f6c2eedbba7cdf94eed321be","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2020-08-03 15:01:57.000000000","message":"Patch Set 3:\n\nFingers crossed :)","accounts_in_message":[],"_revision_number":3},{"id":"afbf3debf6ba5fad0ad2f602881f4803270920e7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-03 16:23:14.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c63f7e39ca56470db13c13972f3cb057 : SUCCESS in 4m 06s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0d8bfb887c8e47f28cd48606fcc6ee1e : SUCCESS in 3m 18s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/52f0048308094d8e835ebe4caee537d7 : SUCCESS in 3m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/315c0dcaf2d74ad9af4b9072593571b6 : SUCCESS in 3m 59s\n- requirements-check https://zuul.opendev.org/t/openstack/build/2c97ed41eb4e4e028cf56d12a7438fa2 : FAILURE in 2m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b95df176839142ffb4ca2d1ab6a674d9 : SUCCESS in 4m 26s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/b267c625b52c47ba8ee92081743ace88 : SUCCESS in 5m 46s\n- ovsdbapp-tempest-dsvm-networking-ovn-ovs-release https://zuul.opendev.org/t/openstack/build/7c19861444b648f994a846b6f2178a22 : SUCCESS in 1h 19m 57s","accounts_in_message":[],"_revision_number":3},{"id":"5e74c3d340c1a73b98d4887452c41427289a68cc","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-08-03 16:31:34.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"b054afda395594aabb6284cd59621d450321b209","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-03 17:40:25.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed61a2dffe194666b96f776ecc9d56a7 : SUCCESS in 4m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d2a993e2bb3d499dbf42fdd385291a44 : SUCCESS in 5m 11s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/fdd49eb0da6a4068af3feb3899ee3a1f : SUCCESS in 5m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9946c218fe7e48acbc2e951a548e9520 : SUCCESS in 3m 33s\n- requirements-check https://zuul.opendev.org/t/openstack/build/4cdd3ebad6ec4d0f8463065e9dacbdc9 : SUCCESS in 4m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/93b8be8ed7ea4596a706f7dd3015298c : SUCCESS in 6m 19s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/a104684c4b09476ea3942f179eda2a9f : SUCCESS in 5m 01s\n- ovsdbapp-tempest-dsvm-networking-ovn-ovs-release https://zuul.opendev.org/t/openstack/build/2283efb8eb0c497db1ba5e9eccee27b8 : SUCCESS in 1h 04m 32s","accounts_in_message":[],"_revision_number":4},{"id":"7560d774d5d3ef9da2fa5d18c29be345023e4f23","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2020-08-04 06:55:33.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"b76a70197924132d06d0403c8290d877601dfedf","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-08-05 13:15:39.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"00fb9e2b0ff1f3dde3d2fe3ba09b04dd30e38546","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-05 13:15:50.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"b029341c6de69f7b902831f0fc07d1e66842a57d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-05 14:49:53.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b4973c76f2d4453199a8cc588ae0afb8 : SUCCESS in 6m 09s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0237dfe63f9b4606a90ff8d904eb8460 : SUCCESS in 6m 09s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/3ac216968c1d4989bfe5b65461441827 : SUCCESS in 4m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/48fc411a4c0840ee86162e4898b5b7ac : SUCCESS in 4m 42s\n- requirements-check https://zuul.opendev.org/t/openstack/build/37312e6e4f074d9a8b3ba3d5762f0873 : SUCCESS in 4m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2115a585b89c4f4d892c2473e3e97c55 : SUCCESS in 6m 20s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/a7fc7fece1094bacb86930f58561f988 : SUCCESS in 6m 51s\n- ovsdbapp-tempest-dsvm-networking-ovn-ovs-release https://zuul.opendev.org/t/openstack/build/8bcbdd8542c34c4a89934780d134444d : SUCCESS in 1h 26m 32s","accounts_in_message":[],"_revision_number":4},{"id":"49cd6a73aa39347e61f658e0946a29160138a9cb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-05 14:49:53.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"7c785866cf978dfaeb06afd0335acd8fbc21b459","revisions":{"dab76ce12ad5743eb206807bcfd1e9bb1b979bc3":{"kind":"REWORK","_number":1,"created":"2020-07-30 14:18:50.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/85/743985/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/85/743985/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/85/743985/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/85/743985/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/85/743985/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/85/743985/1"}}},"commit":{"parents":[{"commit":"1e538a2d714d0a1ae8944450f07a024c98ca73ec","subject":"Accept a Row as a result in AddCommand","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/1e538a2d714d0a1ae8944450f07a024c98ca73ec"}]}],"author":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2020-07-07 15:14:58.000000000","tz":120},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-07-30 14:18:50.000000000","tz":0},"subject":"Fix pep8 job","message":"Fix pep8 job\n\nNew versions of isort broke pylint. This patch fixes it at 4.3.21.\n\nChange-Id: Ie1b3b162e18090ae1cc90998b5a41d74b9ef2876\nDepends-On: https://review.opendev.org/739469\n(cherry picked from commit dac170c8b496fb13bd5b58b83ed061aef0626510)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/dab76ce12ad5743eb206807bcfd1e9bb1b979bc3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/dab76ce12ad5743eb206807bcfd1e9bb1b979bc3"}]},"branch":"refs/heads/stable/stein"},"816434fe32e57d80d5efc817a3bee0de1288fd99":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-08-03 14:03:57.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/85/743985/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/85/743985/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/85/743985/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/85/743985/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/85/743985/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/85/743985/2"}}},"commit":{"parents":[{"commit":"7b8895cb848e7fdd4f3c6411882237215b63b8f3","subject":"Update sphinx reqs to match global","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/7b8895cb848e7fdd4f3c6411882237215b63b8f3"}]}],"author":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2020-07-07 15:14:58.000000000","tz":120},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-08-03 14:03:57.000000000","tz":0},"subject":"Fix pep8 job","message":"Fix pep8 job\n\nNew versions of isort broke pylint. This patch fixes it at 4.3.21.\n\nChange-Id: Ie1b3b162e18090ae1cc90998b5a41d74b9ef2876\nDepends-On: https://review.opendev.org/739469\n(cherry picked from commit dac170c8b496fb13bd5b58b83ed061aef0626510)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/816434fe32e57d80d5efc817a3bee0de1288fd99"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/816434fe32e57d80d5efc817a3bee0de1288fd99"}]},"branch":"refs/heads/stable/stein"},"4c7866aee2946ca892a5fbe2dc46a5c05c18f363":{"kind":"REWORK","_number":3,"created":"2020-08-03 14:53:22.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/85/743985/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/85/743985/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/85/743985/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/85/743985/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/85/743985/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/85/743985/3"}}},"commit":{"parents":[{"commit":"1e538a2d714d0a1ae8944450f07a024c98ca73ec","subject":"Accept a Row as a result in AddCommand","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/1e538a2d714d0a1ae8944450f07a024c98ca73ec"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-08-03 14:03:27.000000000","tz":-300},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-08-03 14:52:57.000000000","tz":-300},"subject":"Update sphinx reqs to match global","message":"Update sphinx reqs to match global\n\nChange-Id: Ia3094ca6298ed198dd1baf5f95e859165a2fc210\n\nFix pep8 job\n\nNew versions of isort broke pylint. This patch fixes it at 4.3.21.\n\nChange-Id: Ie1b3b162e18090ae1cc90998b5a41d74b9ef2876\nDepends-On: https://review.opendev.org/739469\n(cherry picked from commit dac170c8b496fb13bd5b58b83ed061aef0626510)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/4c7866aee2946ca892a5fbe2dc46a5c05c18f363"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/4c7866aee2946ca892a5fbe2dc46a5c05c18f363"}]},"branch":"refs/heads/stable/stein"},"7c785866cf978dfaeb06afd0335acd8fbc21b459":{"kind":"REWORK","_number":4,"created":"2020-08-03 16:31:34.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/85/743985/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/85/743985/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/85/743985/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/85/743985/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/85/743985/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/85/743985/4"}}},"commit":{"parents":[{"commit":"1e538a2d714d0a1ae8944450f07a024c98ca73ec","subject":"Accept a Row as a result in AddCommand","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/1e538a2d714d0a1ae8944450f07a024c98ca73ec"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-08-03 14:03:27.000000000","tz":-300},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-08-03 16:30:18.000000000","tz":-300},"subject":"Update sphinx reqs to match global","message":"Update sphinx reqs to match global\n\nI had to add the lower bound to get this to pass--\neven running check-requirements against the\nstable/stein requirements repo fails in the same way\nwithout it for some reason.\n\nChange-Id: Ia3094ca6298ed198dd1baf5f95e859165a2fc210\n\nFix pep8 job\n\nNew versions of isort broke pylint. This patch fixes it at 4.3.21.\n\nChange-Id: Ie1b3b162e18090ae1cc90998b5a41d74b9ef2876\nDepends-On: https://review.opendev.org/739469\n(cherry picked from commit dac170c8b496fb13bd5b58b83ed061aef0626510)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/7c785866cf978dfaeb06afd0335acd8fbc21b459"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/7c785866cf978dfaeb06afd0335acd8fbc21b459"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
