)]}'
{"id":"openstack%2Fovsdbapp~744476","triplet_id":"openstack%2Fovsdbapp~stable%2Fstein~Ia3094ca6298ed198dd1baf5f95e859165a2fc210","project":"openstack/ovsdbapp","branch":"stable/stein","hashtags":[],"change_id":"Ia3094ca6298ed198dd1baf5f95e859165a2fc210","subject":"Update sphinx reqs to match global","status":"ABANDONED","created":"2020-08-03 14:03:32.000000000","updated":"2020-08-03 14:54:06.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"b309dcd031e291730ed3e4813fedd202f52a2d8e","_number":744476,"virtual_id_number":744476,"owner":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2020-08-03 14:48:48.000000000","_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-08-03 14:17:16.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2020-08-03 14:31:14.000000000","_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-03 14:17:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-08-03 14:31:14.000000000","updated_by":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"reviewer":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"state":"REVIEWER"}],"messages":[{"id":"825948219c643e3589f7b05939309da8056fed10","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-08-03 14:03:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f9ebb1d918b64d347842f4c31cabd411abdeea04","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-03 14:17:16.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6a39e683f93b409fb7e6564197b7a7df : FAILURE in 4m 06s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/077c9a32fd3e4d638625c87632ed18d2 : SUCCESS in 4m 35s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/6b07b406ea8e483783dcaa39661cfd5d : SUCCESS in 3m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/97214e0398bb42cea0bcf6ce15781906 : SUCCESS in 4m 24s\n- requirements-check https://zuul.opendev.org/t/openstack/build/b9b24a35267b48d3bb444d706a236ee3 : FAILURE in 2m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e6f5106dc5f74a6caecd0dc9c74994de : SUCCESS in 5m 22s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/1eda57cd7dce484798815da9e02e979a : SUCCESS in 4m 35s","accounts_in_message":[],"_revision_number":1},{"id":"4e9fd038ad0a0aef9e3e148565d4e2fdcc65263d","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2020-08-03 14:31:14.000000000","message":"Patch Set 1:\n\nwe have to squash the patches into one?","accounts_in_message":[],"_revision_number":1},{"id":"cd254a97f3c342063f459e7ab79692b527c3def4","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-08-03 14:48:48.000000000","message":"Patch Set 1:\n\nyeah. :-/","accounts_in_message":[],"_revision_number":1},{"id":"b309dcd031e291730ed3e4813fedd202f52a2d8e","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-08-03 14:54:06.000000000","message":"Abandoned\n\nsquahsed into https://review.opendev.org/#/c/743985/","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7b8895cb848e7fdd4f3c6411882237215b63b8f3","revisions":{"7b8895cb848e7fdd4f3c6411882237215b63b8f3":{"kind":"REWORK","_number":1,"created":"2020-08-03 14:03:32.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/76/744476/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/76/744476/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/76/744476/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/76/744476/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/76/744476/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/76/744476/1"}}},"commit":{"parents":[{"commit":"1e538a2d714d0a1ae8944450f07a024c98ca73ec","subject":"Accept a Row as a result in AddCommand","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/1e538a2d714d0a1ae8944450f07a024c98ca73ec"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-08-03 14:03:27.000000000","tz":-300},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-08-03 14:03:27.000000000","tz":-300},"subject":"Update sphinx reqs to match global","message":"Update sphinx reqs to match global\n\nChange-Id: Ia3094ca6298ed198dd1baf5f95e859165a2fc210\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/7b8895cb848e7fdd4f3c6411882237215b63b8f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/7b8895cb848e7fdd4f3c6411882237215b63b8f3"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
