)]}'
{"id":"openstack%2Fovsdbapp~752092","triplet_id":"openstack%2Fovsdbapp~master~Ia068650d2db3d5d8642771a6df5a260d692aea20","project":"openstack/ovsdbapp","branch":"master","hashtags":[],"change_id":"Ia068650d2db3d5d8642771a6df5a260d692aea20","subject":"Don\u0027t give up when an Exception happens in idl.run","status":"MERGED","created":"2020-09-15 18:49:27.000000000","updated":"2020-10-26 11:24:44.000000000","submitted":"2020-09-28 19:27:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"752092-1601321235735-276e733d","meta_rev_id":"955282239d528a051df8dced013737fe7df48da2","_number":752092,"virtual_id_number":752092,"owner":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-28 19:27:15.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"recommended":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"all":[{"value":1,"date":"2020-09-28 14:55:47.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"value":2,"date":"2020-09-28 14:55:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":2,"date":"2020-09-28 14:55:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"all":[{"value":0,"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-09-28 14:56:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-22 13:21:29.000000000","updated_by":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"reviewer":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"state":"REVIEWER"},{"updated":"2020-09-22 13:27:48.000000000","updated_by":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"reviewer":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"state":"REVIEWER"},{"updated":"2020-09-22 14:44:37.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-09-22 15:38:47.000000000","updated_by":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"reviewer":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"state":"REVIEWER"},{"updated":"2020-09-28 19:27:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"79070ce92ff68542796a8d231c969e4ade0405a1","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-09-15 18:49:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3916e8f867c1506fa462db6a0ac3693a19ccd019","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-15 19:03:36.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/421a9bc343284876a2c08b6097f7c4c8 : SUCCESS in 4m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/473f4c542e8149f0a50624ee9102f812 : SUCCESS in 3m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8cc3c1562a024b1d9560b6e889f152b8 : SUCCESS in 4m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1582ab1521b44c5e9b5bc43084e50d54 : SUCCESS in 4m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e8e30409976c4ba8bf87a7775c2aaf25 : SUCCESS in 3m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea3ff676a3da4d6b85461eccb2670f4c : SUCCESS in 12m 36s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/b140afe57e824278bb5e4a7693339ba6 : SUCCESS in 6m 00s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/259873fcaf8742b390f7520f5eac7391 : FAILURE in 11m 31s","accounts_in_message":[],"_revision_number":1},{"id":"80c318170f257fdad423ee1eef06525eaa601dbd","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-09-15 19:06:46.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)\n\nremove redundant min()","accounts_in_message":[],"_revision_number":1},{"id":"781dca396a581fe3ed5210ee4a78cae5fc1ebc87","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-09-15 19:19:11.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9501e98d3d7f1f72385d86a0965787395cf3ef9a","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-09-15 19:22:38.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c1aec0ba9e1c3d3b848e29fd64bbab3c8d7b05a8","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-09-15 19:26:17.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"1d9a1b7283307014d2374dc6ab32b182e75ed61f","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-09-15 19:27:57.000000000","message":"Patch Set 3: Code-Review+1\n\nLGTM!","accounts_in_message":[],"_revision_number":3},{"id":"d9cbadf2af6f39aa6e4481fd1989ab5596b22f01","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-15 19:53:28.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b62fb482caa54aa5a4a2def3869024dc : SUCCESS in 4m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dbfec0e9e6b6486dac5edaaefe521d5c : SUCCESS in 3m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e72d7c85fd59451ea8dd070c1ed4e553 : SUCCESS in 4m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/eaa7d388edaf4dae8f96a0037791bcde : SUCCESS in 5m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/14195ddfd9314bf2ac9f6f3b68bcf177 : SUCCESS in 4m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3837e58a763d4333a0917f1878178c4f : SUCCESS in 25m 49s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/4d42aae339204ef99b2e910f59d52086 : SUCCESS in 5m 43s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/57c2b26eb0164771a90f7ad1ebfe04d1 : FAILURE in 11m 20s","accounts_in_message":[],"_revision_number":3},{"id":"4275ab114a4c7d0aaca58f9486a786c11b55d974","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-09-16 00:22:31.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"132c68e7eb48f96b613cee8865911be2b3221ca1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-16 01:16:15.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6ad8d9a292e7461bb57482f5fc8cbc12 : SUCCESS in 4m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c485f9b062c748988fef33a5ef34029b : SUCCESS in 3m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/34bd9d3df5c94fbba9d96abc2f3d8c15 : SUCCESS in 3m 35s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5dd55a9cbbf547cb911c9160cd0c9dd8 : SUCCESS in 4m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2fe85710c9f04cb78dd79f836d9bf40a : SUCCESS in 4m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c282302877064e258a19cc841b6aaaa5 : SUCCESS in 13m 43s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/69a813c0ad604d36807d262ba1515f02 : SUCCESS in 5m 38s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/a9eed3ce34f149059763db3ec7731157 : SUCCESS in 52m 44s","accounts_in_message":[],"_revision_number":3},{"id":"db1665ca28b8f394f3a9a45375f9b66a5dcb87e2","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2020-09-16 12:43:42.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"b009d4ffce24de6cf543aa5184ae4fcf98d026d7","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-09-17 01:39:22.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"f6495e6e77a148de8afe0e028262cb32b2487cce","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-09-22 13:20:20.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"803dabe15942d2b6bb5e37694c96d4cf6844b4fc","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-09-22 13:20:31.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"92200e1eb5f935a3e3b9c9eb9cd904d59f5e158c","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-09-22 13:21:29.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"420fa2db66b6f7d7dec07658d489990f823cc04f","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2020-09-22 13:27:48.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"006f9db24ebd7cd2966b175065c01277da05b084","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-22 13:39:04.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4977f38ffcfa4bd5990a48acfc39f889 : SUCCESS in 4m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8102c954b60943aa89523f5d424f4959 : SUCCESS in 3m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/78748e65ca194ca0bb8ee5f018d94a33 : SUCCESS in 4m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0cd64c55c19b40b79914f02b06093733 : SUCCESS in 3m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ea968ebae28d4146a3f1ce3de335b8e3 : SUCCESS in 3m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b926c3dce03044239f0570cd8d31e83e : SUCCESS in 12m 40s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/04ce8bee8d224a9f8577a944ee68ff5b : SUCCESS in 5m 48s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/6e4b129994bf41168e776e2746e0558e : FAILURE in 15m 01s","accounts_in_message":[],"_revision_number":4},{"id":"f3dd2890d4fbde7907d0aa0d94eea7a96a4b897e","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-09-22 14:44:37.000000000","message":"Patch Set 4:\n\nThis might need https://review.opendev.org/#/c/752412/ ?  Just a guess.","accounts_in_message":[],"_revision_number":4},{"id":"ed59b4b95fa9439feb09d5e9b9c356c8eb83637b","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2020-09-22 15:38:47.000000000","message":"Patch Set 4: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"1f5864f3afe7eca7ef03c3aed6d291bae313b11a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-22 15:55:19.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6996552bcbec4bcb954ef16a59a55052 : SUCCESS in 3m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e38afd6c5b144f9bbff1a60bacb7e1be : SUCCESS in 4m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6292ee9a388d4057ad7d0db719cb9f9e : SUCCESS in 4m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f6730234aecd45f491014d84fbad3fa6 : SUCCESS in 4m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d89c189808e9411baa7cf59909cddbe4 : SUCCESS in 3m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b792beb4fdb34c61a4beedd8bae8824b : SUCCESS in 13m 43s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/4b28f3c375324d74a6da73081d3fa54f : SUCCESS in 7m 55s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/8a9736d5180e45d1a9f57c4949c088ce : FAILURE in 11m 08s","accounts_in_message":[],"_revision_number":4},{"id":"bedd7d23e5746ca365da8082a58230b9ad86fec7","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-09-22 18:11:05.000000000","message":"Patch Set 4:\n\n\u003e This might need https://review.opendev.org/#/c/752412/ ?  Just a guess.\n\nOh, well at least it is a devstack patch so we shouldn\u0027t have to wait long.","accounts_in_message":[],"_revision_number":4},{"id":"481afad698af5ed135b22f72987a4c3321c1e76e","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-09-28 12:59:01.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"006e9bfefe9ffcad0bfa30f58d98f913a43625e9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-28 14:22:32.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f0459893d1c24598958c9b070973ce1f : SUCCESS in 3m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/30d1eef357bb456c9a56aa93af4283d3 : SUCCESS in 4m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fcc10372ab0e4d6a8b2da6c98f4bd74b : SUCCESS in 4m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/20794b362df94194b3f01e979afdcc26 : SUCCESS in 5m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6773fb9233ef4af8a7452561d5f34aa3 : SUCCESS in 4m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3fc93c41542c4c7287be8876ada3057a : SUCCESS in 5m 27s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/972b6a65175545fb8acef5d7ec57e94e : SUCCESS in 5m 29s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/40724fbc2999404e826e4b9ed2e9ba70 : SUCCESS in 1h 12m 40s","accounts_in_message":[],"_revision_number":4},{"id":"eff550470421e94e3495b8e178dc56c9f93b208f","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-09-28 14:55:47.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"1c3cee0a013a67ee3efe2160041329767d9b33de","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-09-28 14:56:00.000000000","message":"Patch Set 5: Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"314af7272913348777659e82f3dba98f81a45bc2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-28 16:15:59.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/54451baa8a674a4db49752e9f71d34b8 : SUCCESS in 4m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ffa44088c8048ddb745788461632a9d : SUCCESS in 4m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a7f7b7d603674ee4bfb78c1fa4580dc4 : SUCCESS in 5m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c877d5a26be446769b2e8f78c3dd4aa4 : SUCCESS in 4m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5de98373e30140fba40a86393fdf516d : SUCCESS in 4m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/899c966f407449afb84d2f0491412d31 : SUCCESS in 5m 13s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/7cd81a09f37d4786bc19cce5b82ea85c : SUCCESS in 6m 23s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/471c773c692d42b099bb4995c92d622f : FAILURE in 1h 11m 19s","accounts_in_message":[],"_revision_number":5},{"id":"dad96b614a8e46c1406783a21262e0eb8bc12dd4","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-09-28 16:30:05.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"5633ca1de9db37767b5c183456ddb8955f491e18","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-28 17:58:32.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/77fd9ed6b4714748835935ef51acfc05 : SUCCESS in 3m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b3b830a07d4c4d53b52878d64545d02c : SUCCESS in 7m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2083649c9cbf469195eb6ea51aac80eb : SUCCESS in 4m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a222b5d4a0f9473eb2a7ac773a80b65d : SUCCESS in 3m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b2a9e06c70964f32b1671f34c84720db : SUCCESS in 4m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c30b6fb581dd4d0197bfa4fe6113bc27 : SUCCESS in 5m 09s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/f0f689d2168443c0a957d8cc57dbc13b : SUCCESS in 6m 12s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/463ac4d4c6e7432eb42a3e47134bd983 : SUCCESS in 1h 10m 44s","accounts_in_message":[],"_revision_number":5},{"id":"37acfa38e81e7701925cb78173f746558521b437","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-28 17:58:45.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"2e887ac225e4ce56d38b31b79333caf5034bd9dd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-28 19:27:15.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"bf9a3363ac02c23c0321d754364a0103b81ec8b9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-28 19:27:15.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2c64b67938de4d9991a44aa017a1f0fc : SUCCESS in 4m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e7c2189255e4be98ce0cb51b06e9078 : SUCCESS in 3m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/42cce7c2c6d2468d8c2ffde4aa783a39 : SUCCESS in 4m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a9c8b7934ec447b78b96d75cc8f469bb : SUCCESS in 4m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86e1b25d256f41998d5807649acc5adc : SUCCESS in 6m 32s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/f85cba4c71124ec9b5b55b9b69466972 : SUCCESS in 6m 57s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/afc900aa62c848388568332c24a96cb7 : SUCCESS in 1h 20m 19s","accounts_in_message":[],"_revision_number":5},{"id":"c54476b4791f65c8031c24cc4ec8f0815723c3f0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-28 19:28:47.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1030e0e33ab4faca2b6ea49084479c1 : SUCCESS in 1m 07s","accounts_in_message":[],"_revision_number":5},{"id":"955282239d528a051df8dced013737fe7df48da2","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2020-10-26 11:24:44.000000000","message":"Patch Set 5: Cherry Picked\n\nThis patchset was cherry picked to branch stable/victoria as commit 4807809ba7ec09a7e9cf533e334de282e0d373cd","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"83cf7aa6c81f1b2341b2bba1fe156047fa5d29f6","revisions":{"514648593c334fd68fa24a34106ac3db8ece1014":{"kind":"REWORK","_number":1,"created":"2020-09-15 18:49:27.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/92/752092/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/92/752092/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/92/752092/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/92/752092/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/92/752092/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/92/752092/1"}}},"commit":{"parents":[{"commit":"324e8dfd3fb4c57ab78bdfde28ce3cb2a622bc67","subject":"Merge \"Switch from unittest2 compat methods to Python 3.x methods\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/324e8dfd3fb4c57ab78bdfde28ce3cb2a622bc67"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-09-15 18:42:08.000000000","tz":-300},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-09-15 18:42:08.000000000","tz":-300},"subject":"Don\u0027t give up when an Exception happens in idl.run","message":"Don\u0027t give up when an Exception happens in idl.run\n\nIt\u0027s possible that idl.run() could have a bug where it raises an\nException for an extended period of time while ovsdb-server is\ndown, but recover once ovsdb-server comes back up. Specifically,\npython-ovs currently doesn\u0027t properly catch an exception when the\nsocket type is \u0027ssl\u0027 that it catches for other protocols.\n\nChange-Id: Ia068650d2db3d5d8642771a6df5a260d692aea20\nCloses-Bug: #1895727\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/514648593c334fd68fa24a34106ac3db8ece1014"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/514648593c334fd68fa24a34106ac3db8ece1014"}]},"branch":"refs/heads/master"},"1e63732868c3c164f96d76c509902321a813f646":{"kind":"REWORK","_number":2,"created":"2020-09-15 19:22:38.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/92/752092/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/92/752092/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/92/752092/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/92/752092/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/92/752092/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/92/752092/2"}}},"commit":{"parents":[{"commit":"324e8dfd3fb4c57ab78bdfde28ce3cb2a622bc67","subject":"Merge \"Switch from unittest2 compat methods to Python 3.x methods\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/324e8dfd3fb4c57ab78bdfde28ce3cb2a622bc67"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-09-15 18:42:08.000000000","tz":-300},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-09-15 19:22:30.000000000","tz":-300},"subject":"Don\u0027t give up when an Exception happens in idl.run","message":"Don\u0027t give up when an Exception happens in idl.run\n\nIt\u0027s possible that idl.run() could have a bug where it raises an\nException for an extended period of time while ovsdb-server is\ndown, but recover once ovsdb-server comes back up. Specifically,\npython-ovs currently doesn\u0027t properly catch an exception when the\nsocket type is \u0027ssl\u0027 that it catches for other protocols.\n\nChange-Id: Ia068650d2db3d5d8642771a6df5a260d692aea20\nCloses-Bug: #1895727\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/1e63732868c3c164f96d76c509902321a813f646"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/1e63732868c3c164f96d76c509902321a813f646"}]},"branch":"refs/heads/master"},"04263acc850b10c38a4cc5c352df95297d7b9ce0":{"kind":"REWORK","_number":3,"created":"2020-09-15 19:26:17.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/92/752092/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/92/752092/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/92/752092/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/92/752092/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/92/752092/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/92/752092/3"}}},"commit":{"parents":[{"commit":"324e8dfd3fb4c57ab78bdfde28ce3cb2a622bc67","subject":"Merge \"Switch from unittest2 compat methods to Python 3.x methods\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/324e8dfd3fb4c57ab78bdfde28ce3cb2a622bc67"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-09-15 18:42:08.000000000","tz":-300},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-09-15 19:26:12.000000000","tz":-300},"subject":"Don\u0027t give up when an Exception happens in idl.run","message":"Don\u0027t give up when an Exception happens in idl.run\n\nIt\u0027s possible that idl.run() could have a bug where it raises an\nException for an extended period of time while ovsdb-server is\ndown, but recover once ovsdb-server comes back up. Specifically,\npython-ovs currently doesn\u0027t properly catch an exception when the\nsocket type is \u0027ssl\u0027 that it catches for other protocols.\n\nChange-Id: Ia068650d2db3d5d8642771a6df5a260d692aea20\nCloses-Bug: #1895727\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/04263acc850b10c38a4cc5c352df95297d7b9ce0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/04263acc850b10c38a4cc5c352df95297d7b9ce0"}]},"branch":"refs/heads/master"},"879faa00f70e52db58a26373fc1a9f8d933e926e":{"kind":"REWORK","_number":4,"created":"2020-09-22 13:20:20.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/92/752092/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/92/752092/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/92/752092/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/92/752092/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/92/752092/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/92/752092/4"}}},"commit":{"parents":[{"commit":"324e8dfd3fb4c57ab78bdfde28ce3cb2a622bc67","subject":"Merge \"Switch from unittest2 compat methods to Python 3.x methods\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/324e8dfd3fb4c57ab78bdfde28ce3cb2a622bc67"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-09-15 18:42:08.000000000","tz":-300},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-09-22 13:20:14.000000000","tz":-300},"subject":"Don\u0027t give up when an Exception happens in idl.run","message":"Don\u0027t give up when an Exception happens in idl.run\n\nIt\u0027s possible that idl.run() could have a bug where it raises an\nException for an extended period of time while ovsdb-server is\ndown, but recover once ovsdb-server comes back up. Specifically,\npython-ovs currently doesn\u0027t properly catch an exception when the\nsocket type is \u0027ssl\u0027 that it catches for other protocols.\n\nChange-Id: Ia068650d2db3d5d8642771a6df5a260d692aea20\nCloses-Bug: #1895727\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/879faa00f70e52db58a26373fc1a9f8d933e926e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/879faa00f70e52db58a26373fc1a9f8d933e926e"}]},"branch":"refs/heads/master"},"83cf7aa6c81f1b2341b2bba1fe156047fa5d29f6":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2020-09-28 14:55:47.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/92/752092/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/92/752092/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/92/752092/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/92/752092/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/92/752092/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/92/752092/5"}}},"commit":{"parents":[{"commit":"e365273ad21b42662c9079e95b72fb6ef0b76d54","subject":"Merge \"Don\u0027t use name mangling when we don\u0027t need it\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/e365273ad21b42662c9079e95b72fb6ef0b76d54"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-09-15 18:42:08.000000000","tz":-300},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-09-28 14:55:47.000000000","tz":0},"subject":"Don\u0027t give up when an Exception happens in idl.run","message":"Don\u0027t give up when an Exception happens in idl.run\n\nIt\u0027s possible that idl.run() could have a bug where it raises an\nException for an extended period of time while ovsdb-server is\ndown, but recover once ovsdb-server comes back up. Specifically,\npython-ovs currently doesn\u0027t properly catch an exception when the\nsocket type is \u0027ssl\u0027 that it catches for other protocols.\n\nChange-Id: Ia068650d2db3d5d8642771a6df5a260d692aea20\nCloses-Bug: #1895727\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/83cf7aa6c81f1b2341b2bba1fe156047fa5d29f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/83cf7aa6c81f1b2341b2bba1fe156047fa5d29f6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
