)]}'
{"id":"openstack%2Fovsdbapp~756148","triplet_id":"openstack%2Fovsdbapp~master~Ie477f03f82550535d0324689babe549a7b323373","project":"openstack/ovsdbapp","branch":"master","hashtags":[],"change_id":"Ie477f03f82550535d0324689babe549a7b323373","subject":"Bump ovs minimum version for support of indexing","status":"MERGED","created":"2020-10-05 18:23:40.000000000","updated":"2020-10-07 18:04:27.000000000","submitted":"2020-10-07 18:02:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"756148-1602093773919-f61217dc","meta_rev_id":"86dd35e597ed8561126890dd66c0d6a2504ec3f7","_number":756148,"virtual_id_number":756148,"owner":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-07 18:02:53.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"recommended":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"all":[{"value":1,"date":"2020-10-07 15:23:01.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2020-10-07 15:22:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"all":[{"value":0,"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2020-10-07 15:22:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-06 17:51:12.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2020-10-06 17:51:17.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2020-10-06 17:53:20.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2020-10-07 14:40:08.000000000","updated_by":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"reviewer":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"state":"REVIEWER"},{"updated":"2020-10-07 14:41:40.000000000","updated_by":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"reviewer":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"state":"REVIEWER"},{"updated":"2020-10-07 15:23:01.000000000","updated_by":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"reviewer":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"state":"REVIEWER"},{"updated":"2020-10-07 18:02:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"21826dcd5b132f0d805dbfd8fb45ee920e0b68f8","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-10-05 18:23:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"28dbf4fdf35b782a62035602e98132b926490b2c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-05 18:38:01.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/a3edc0e8447a4a16b1ac8b6f28de0b92 : FAILURE in 3m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/737c81d3664e4558a1191527e9a3140a : SUCCESS in 4m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9dc6cc620fab4f018c947af70257fa54 : SUCCESS in 4m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6bebc72b2a4744ae89c3d3ce7969998a : SUCCESS in 3m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f59da4a5571f4f2fbbbad4d24bae9c27 : SUCCESS in 6m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/90bfc5e5add94e78ae1ec1c3b517870e : SUCCESS in 3m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5703655763b043c59d1650cf50c13f83 : SUCCESS in 5m 37s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/0ebce95f31a348fdb7a3d602e54335ff : SUCCESS in 6m 05s","accounts_in_message":[],"_revision_number":1},{"id":"a866763f953450ad199ee49c4fe2b590c86b2fc2","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-10-05 19:00:46.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"42c371886780b94243188fae93707da64aec72a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-05 19:32:53.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/d25af4484bee41daa54a0ac20b745760 : SUCCESS in 2m 49s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/68264a6d656c4fda9e3d9799fe4af081 : SUCCESS in 4m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d0aba2f3148d45faa02b9a7e0a84f7a2 : SUCCESS in 7m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/07e44771c3a14faeb3fb7a302742a60a : SUCCESS in 4m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/639d4ef56b804481a75f3b077226aa1f : SUCCESS in 4m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f447b747702b44fa8a47fbb2d0995d78 : SUCCESS in 5m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce4fa5261b394fadbbeb8d1f62ab2c17 : SUCCESS in 4m 51s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/e0768cf2ba00440693061fddf33877d5 : SUCCESS in 5m 51s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/274c95529ddf4e5a9b01c9679883dc9e : FAILURE in 28m 42s","accounts_in_message":[],"_revision_number":2},{"id":"b1d360b14624b63dff6d601d69f1cdd087e29cb0","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-10-05 19:44:30.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"e8ee72d424c5781ce5f358a6e9cc7ae9ad6200d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-05 21:32:47.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/7283a51da13c4778ac50a90ffd10fe1f : SUCCESS in 2m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5e670b9c730e4dbd8737003b4a5587a9 : SUCCESS in 4m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/71a907dfcc23426d90dc2ac8c40dc95c : SUCCESS in 6m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7a4bd896987640569c82d8cbf7595844 : SUCCESS in 4m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/71fa5c7bf9c941eca1317b3669351a16 : SUCCESS in 4m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/286a9adcae8e408b9f11c80532a24763 : SUCCESS in 3m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d829bab2aa7442eac232d3a292347ff : SUCCESS in 5m 21s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/7b8e47fa607d44f49a87c0a27721da90 : SUCCESS in 5m 27s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/1c1d4a3ddd2c4c3e9673c88d04c28ca9 : SUCCESS in 1h 17m 58s","accounts_in_message":[],"_revision_number":3},{"id":"76f611f6ec529dd678953605d0d8344b92c64d93","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-10-06 14:55:39.000000000","message":"Patch Set 3: Cherry Picked\n\nThis patchset was cherry picked to branch stable/victoria as commit 240e7d41f37a8fa32dc565d432284b878e775bfb","accounts_in_message":[],"_revision_number":3},{"id":"5efd0014d9a6d7bf1865a64f62a2ced0de2ffafe","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2020-10-07 14:40:08.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"53cebdcb7624661e09ffe5c855ec750c0c4ab9b1","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2020-10-07 14:41:40.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"afe8dbf3864471f604a8c728a66d494ea9a7a166","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-10-07 15:21:12.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"c0f7621e075b8ec8af3b1800ba9b9c60fb94fe4e","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-10-07 15:22:28.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\nFor some reason it wasn\u0027t kicking off the CI. Removed the Depends-On and re-marking +2+1 since nothing changed.","accounts_in_message":[],"_revision_number":4},{"id":"811bd195648b974dae7efde900732e8a78429e05","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-10-07 15:23:01.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"02169bc66ad1c449bb549c51423736ab537f9fc2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-07 16:46:29.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/ddaa4db54f97491aa26dd091091d64d3 : SUCCESS in 3m 03s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b2228275d0a54971b9810c08bbef5e31 : SUCCESS in 3m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/73f150ad53704d58849aaf792df58fda : SUCCESS in 3m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a1b903b813ca462baf22a73e906c730d : SUCCESS in 3m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1bef9ca6b6fb4be98a3fe90ea00e1b91 : SUCCESS in 4m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2b010128efa04f93a38e5e9a12a28bb6 : SUCCESS in 4m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e20d04dba6e94dc3bab4257cb1a58582 : SUCCESS in 6m 04s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/40084639677b4c21ab9a61a044359c7f : SUCCESS in 7m 10s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/08abc475d1944f70a0ea6b6d4ecafe65 : SUCCESS in 1h 14m 18s","accounts_in_message":[],"_revision_number":4},{"id":"ca24045f1c2c2cebc6f4bd33b3ce230b972b7d52","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-07 16:46:41.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"7802197ba71a201535729a7e983940f73f770d77","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-07 18:02:53.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/578c669232ba41fd98c1036c908367c1 : SUCCESS in 2m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d1996ad2763c44d59f38190c93049433 : SUCCESS in 3m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b21cd293303942488a34cd0666be2b3c : SUCCESS in 3m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5d7ca071c4b4492e90fa00280a551714 : SUCCESS in 3m 35s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5246bfeb087149cd933f9011867878ba : SUCCESS in 4m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ad4bb8878394cf5b5c19a74dc39662b : SUCCESS in 4m 58s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/ad0a4ca823a0433ca045e47c44a3ebf7 : SUCCESS in 5m 53s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/ee57ebe6d7034898a4b5d106ec87dd56 : SUCCESS in 1h 11m 07s","accounts_in_message":[],"_revision_number":4},{"id":"62c838047996b960a5aeb53686bb5931dee38207","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-07 18:02:53.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"86dd35e597ed8561126890dd66c0d6a2504ec3f7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-07 18:04:27.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09296cc3c5f54e78a90a66d790173a3e : SUCCESS in 1m 05s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"1a1219f7cd659afde6a5c2cb282c93cad879d472","revisions":{"25672817c1ff1e1260719ed973145aa8cae53311":{"kind":"REWORK","_number":1,"created":"2020-10-05 18:23:40.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/48/756148/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/48/756148/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/48/756148/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/48/756148/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/48/756148/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/48/756148/1"}}},"commit":{"parents":[{"commit":"5e5d9071a89a91d0aa2638447e287284f7952077","subject":"Merge \"Avoid race condition with RowEvent handling\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/5e5d9071a89a91d0aa2638447e287284f7952077"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-10-05 18:22:55.000000000","tz":-300},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-10-05 18:22:55.000000000","tz":-300},"subject":"Bump ovs minimum version for support of indexing","message":"Bump ovs minimum version for support of indexing\n\nThe custom indexing patch requires ovs \u003e\u003d 2.10\n\nChange-Id: Ie477f03f82550535d0324689babe549a7b323373\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/25672817c1ff1e1260719ed973145aa8cae53311"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/25672817c1ff1e1260719ed973145aa8cae53311"}]},"branch":"refs/heads/master"},"0e9eb4f6a9459e6491f8e357988e38a842aa14d3":{"kind":"REWORK","_number":2,"created":"2020-10-05 19:00:46.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/48/756148/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/48/756148/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/48/756148/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/48/756148/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/48/756148/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/48/756148/2"}}},"commit":{"parents":[{"commit":"5e5d9071a89a91d0aa2638447e287284f7952077","subject":"Merge \"Avoid race condition with RowEvent handling\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/5e5d9071a89a91d0aa2638447e287284f7952077"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-10-05 18:22:55.000000000","tz":-300},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-10-05 19:00:43.000000000","tz":-300},"subject":"Bump ovs minimum version for support of indexing","message":"Bump ovs minimum version for support of indexing\n\nThe custom indexing patch requires ovs \u003e\u003d 2.10\n\nChange-Id: Ie477f03f82550535d0324689babe549a7b323373\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/0e9eb4f6a9459e6491f8e357988e38a842aa14d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/0e9eb4f6a9459e6491f8e357988e38a842aa14d3"}]},"branch":"refs/heads/master"},"40dc8600fded978b0a8aef54feeb886f8a05f0ce":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2020-10-05 19:44:30.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/48/756148/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/48/756148/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/48/756148/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/48/756148/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/48/756148/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/48/756148/3"}}},"commit":{"parents":[{"commit":"5e5d9071a89a91d0aa2638447e287284f7952077","subject":"Merge \"Avoid race condition with RowEvent handling\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/5e5d9071a89a91d0aa2638447e287284f7952077"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-10-05 18:22:55.000000000","tz":-300},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-10-05 19:44:27.000000000","tz":0},"subject":"Bump ovs minimum version for support of indexing","message":"Bump ovs minimum version for support of indexing\n\nThe custom indexing patch requires ovs \u003e\u003d 2.10\n\nDepends-On: Icb4fd80133ee10ad6c40e160013b4ab91d2c5a6b\nChange-Id: Ie477f03f82550535d0324689babe549a7b323373\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/40dc8600fded978b0a8aef54feeb886f8a05f0ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/40dc8600fded978b0a8aef54feeb886f8a05f0ce"}]},"branch":"refs/heads/master"},"1a1219f7cd659afde6a5c2cb282c93cad879d472":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2020-10-07 15:21:12.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/48/756148/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/48/756148/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/48/756148/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/48/756148/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/48/756148/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/48/756148/4"}}},"commit":{"parents":[{"commit":"5e5d9071a89a91d0aa2638447e287284f7952077","subject":"Merge \"Avoid race condition with RowEvent handling\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/5e5d9071a89a91d0aa2638447e287284f7952077"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-10-05 18:22:55.000000000","tz":-300},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-10-07 15:21:08.000000000","tz":0},"subject":"Bump ovs minimum version for support of indexing","message":"Bump ovs minimum version for support of indexing\n\nThe custom indexing patch requires ovs \u003e\u003d 2.10\n\nChange-Id: Ie477f03f82550535d0324689babe549a7b323373\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/1a1219f7cd659afde6a5c2cb282c93cad879d472"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/1a1219f7cd659afde6a5c2cb282c93cad879d472"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
