)]}'
{"id":"openstack%2Fovsdbapp~758154","triplet_id":"openstack%2Fovsdbapp~master~Ibd4338840bd945f0add4f7c427fc00bf4e617a0a","project":"openstack/ovsdbapp","branch":"master","hashtags":[],"change_id":"Ibd4338840bd945f0add4f7c427fc00bf4e617a0a","subject":"Close alertpipe file descriptors when stoping Connection","status":"ABANDONED","created":"2020-10-14 15:09:00.000000000","updated":"2020-10-14 21:25:36.000000000","total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"546d85e7076ca737b33b8c2bfb6ef88bb46082b5","_number":758154,"virtual_id_number":758154,"owner":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2020-10-14 16:43:55.000000000","_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-10-14 18:08:08.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-14 15:19:27.000000000","updated_by":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"reviewer":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"state":"REVIEWER"},{"updated":"2020-10-14 16:43:55.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-10-14 18:08:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a23e781cb3c27f808924db87ceb655ff96c8959c","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-10-14 15:09:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2c5d8acd82eb5c56040ee6c398501d98df036071","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2020-10-14 15:13:13.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"74ce620dad46f5c4a1fe891eafc1a7e10fbebb6b","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-10-14 15:17:09.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0a3c01d8dd57e6ab0799e49a3917fad324a243c6","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2020-10-14 15:19:27.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"158157f7dc457eae7fa552894e3f840fcc7d9536","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-10-14 15:27:17.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"db361c86940e08c1e9414ef0a8aa3f41f58e8483","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-10-14 15:30:27.000000000","message":"Patch Set 2: Code-Review+2\n\nMy testing is happy, and no extraneous PIPEs showing in lsof :)","accounts_in_message":[],"_revision_number":2},{"id":"40fa4aa3b5b796f7272fad5c7e8d539ecc0d7f13","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-10-14 15:59:32.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"a512430e54ec1622710d8e9dc67603104a5e91dd","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-10-14 16:40:56.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"d9f3fca391be7e04296aaab89a9f49304358427a","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-10-14 16:43:55.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"a2c9c767a30464ac267d4ed1c75c38d396489186","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-14 18:08:08.000000000","message":"Patch Set 4: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3ede3bb1b47842ea964283582329a388 : SUCCESS in 3m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da6231d232bd432fa6befac448871d6e : FAILURE in 4m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/898e52a6fb374a2fac9e3a7d4eb01ffd : SUCCESS in 4m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/78a4644ad2244aa4bd8e9309bd4258c0 : SUCCESS in 4m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/42cd34c8100d4cb492df6e55ad078294 : SUCCESS in 5m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cecf483d34fc47bfba3d87436be12fb8 : SUCCESS in 5m 48s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/15624d0697ab48f5950cb8aab411ad8d : FAILURE in 7m 28s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/46428727ea6847dfb9cef62ffe316058 : SUCCESS in 1h 18m 13s","accounts_in_message":[],"_revision_number":4},{"id":"546d85e7076ca737b33b8c2bfb6ef88bb46082b5","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-10-14 21:25:36.000000000","message":"Abandoned\n\nAbandoning for now as it looks like we can just delete connection/api objects in ovn-octavia-provider. Gets a bit complicated here as we need to be able to queue transactions while connection is stopped.","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"c106cd5aeaf67740d4e88772543664b33b8cb5cb","revisions":{"256cf5ab90b87b1a3b2b4060367e0dc7c0b086b7":{"kind":"REWORK","_number":1,"created":"2020-10-14 15:09:00.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/54/758154/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/54/758154/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/54/758154/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/54/758154/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/54/758154/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/54/758154/1"}}},"commit":{"parents":[{"commit":"86e35523f5080dbe026a43686035096a930b8ccc","subject":"Merge \"Stop to use the __future__ module.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/86e35523f5080dbe026a43686035096a930b8ccc"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-10-14 15:05:47.000000000","tz":0},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-10-14 15:09:00.000000000","tz":0},"subject":"Close alertpipe file descriptors when stoping Connection","message":"Close alertpipe file descriptors when stoping Connection\n\nIf Connections are routinely recreated, not closing these file\ndescriptors will cause a leak. It is necessary to call stop()\non the Connection to close file handles.\n\nChange-Id: Ibd4338840bd945f0add4f7c427fc00bf4e617a0a\nCloses-Bug: #1894136\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/256cf5ab90b87b1a3b2b4060367e0dc7c0b086b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/256cf5ab90b87b1a3b2b4060367e0dc7c0b086b7"}]},"branch":"refs/heads/master"},"fbb794301be4c3672d7d057b10fe80f2d19004cf":{"kind":"REWORK","_number":2,"created":"2020-10-14 15:27:17.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/54/758154/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/54/758154/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/54/758154/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/54/758154/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/54/758154/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/54/758154/2"}}},"commit":{"parents":[{"commit":"86e35523f5080dbe026a43686035096a930b8ccc","subject":"Merge \"Stop to use the __future__ module.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/86e35523f5080dbe026a43686035096a930b8ccc"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-10-14 15:05:47.000000000","tz":0},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-10-14 15:24:41.000000000","tz":0},"subject":"Close alertpipe file descriptors when stoping Connection","message":"Close alertpipe file descriptors when stoping Connection\n\nIf Connections are routinely recreated, not closing these file\ndescriptors will cause a leak. It is necessary to call stop()\non the Connection to close file handles.\n\nChange-Id: Ibd4338840bd945f0add4f7c427fc00bf4e617a0a\nCloses-Bug: #1894136\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/fbb794301be4c3672d7d057b10fe80f2d19004cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/fbb794301be4c3672d7d057b10fe80f2d19004cf"}]},"branch":"refs/heads/master"},"4eb6280c014669e6cf4c77336ac57521559a6e1d":{"kind":"REWORK","_number":3,"created":"2020-10-14 15:59:32.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/54/758154/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/54/758154/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/54/758154/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/54/758154/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/54/758154/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/54/758154/3"}}},"commit":{"parents":[{"commit":"86e35523f5080dbe026a43686035096a930b8ccc","subject":"Merge \"Stop to use the __future__ module.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/86e35523f5080dbe026a43686035096a930b8ccc"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-10-14 15:05:47.000000000","tz":0},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-10-14 15:59:15.000000000","tz":0},"subject":"Close alertpipe file descriptors when stoping Connection","message":"Close alertpipe file descriptors when stoping Connection\n\nIf Connections are routinely recreated, not closing these file\ndescriptors will cause a leak. It is necessary to call stop()\non the Connection to close file handles.\n\nChange-Id: Ibd4338840bd945f0add4f7c427fc00bf4e617a0a\nCloses-Bug: #1894136\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/4eb6280c014669e6cf4c77336ac57521559a6e1d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/4eb6280c014669e6cf4c77336ac57521559a6e1d"}]},"branch":"refs/heads/master"},"c106cd5aeaf67740d4e88772543664b33b8cb5cb":{"kind":"REWORK","_number":4,"created":"2020-10-14 16:40:56.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/54/758154/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/54/758154/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/54/758154/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/54/758154/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/54/758154/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/54/758154/4"}}},"commit":{"parents":[{"commit":"86e35523f5080dbe026a43686035096a930b8ccc","subject":"Merge \"Stop to use the __future__ module.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/86e35523f5080dbe026a43686035096a930b8ccc"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-10-14 15:05:47.000000000","tz":0},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-10-14 16:40:51.000000000","tz":0},"subject":"Close alertpipe file descriptors when stoping Connection","message":"Close alertpipe file descriptors when stoping Connection\n\nIf Connections are routinely recreated, not closing these file\ndescriptors will cause a leak. It is necessary to call stop()\non the Connection to close file handles.\n\nChange-Id: Ibd4338840bd945f0add4f7c427fc00bf4e617a0a\nCloses-Bug: #1894136\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/c106cd5aeaf67740d4e88772543664b33b8cb5cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/c106cd5aeaf67740d4e88772543664b33b8cb5cb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
