)]}'
{"id":"openstack%2Fovsdbapp~760414","triplet_id":"openstack%2Fovsdbapp~stable%2Fussuri~Ia068650d2db3d5d8642771a6df5a260d692aea20","project":"openstack/ovsdbapp","branch":"stable/ussuri","topic":"add_a_connection_lock-stable/ussuri","hashtags":[],"change_id":"Ia068650d2db3d5d8642771a6df5a260d692aea20","subject":"Don\u0027t give up when an Exception happens in idl.run","status":"MERGED","created":"2020-10-29 16:57:51.000000000","updated":"2020-10-30 14:53:44.000000000","submitted":"2020-10-30 14:52:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"760414-1604069527409-941169f8","meta_rev_id":"6766190a2a67ccd97819a42d624d3ae77cb92841","_number":760414,"virtual_id_number":760414,"owner":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-30 14:52:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"recommended":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"all":[{"value":1,"date":"2020-10-29 20:55:35.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-10-30 09:17:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"all":[{"value":0,"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"value":0,"date":"2020-10-29 20:26:14.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-10-30 09:21:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-29 20:26:14.000000000","updated_by":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"reviewer":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"state":"REVIEWER"},{"updated":"2020-10-29 20:55:35.000000000","updated_by":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"reviewer":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"state":"REVIEWER"},{"updated":"2020-10-30 09:21:44.000000000","updated_by":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"reviewer":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"state":"REVIEWER"},{"updated":"2020-10-30 14:52:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f0595cabff03cdb94a3b9ccc2189a7384d0e0196","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-10-29 16:57:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f364aca190fa71f021fa3910718ccd4fd5d20e5e","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-10-29 16:58:58.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"2ab028773b7f33add06e335668168de8ffaa8b25","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-29 18:49:18.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6a8d9d3707514b81bd3590d22281b13d : SUCCESS in 5m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eba32536a6b94e259deae667132d7100 : SUCCESS in 5m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ad933f5ea8d54fe5bb7b945c80a5273f : SUCCESS in 6m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9002901545514feca40c295d4a48324a : SUCCESS in 4m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/178408a9965546fdbea850a2ecaa981c : SUCCESS in 7m 49s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/2608b3eb18b94eafa14f10a1a2963192 : SUCCESS in 6m 50s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/7d0c12b1fc5a40489348c4231f4d2dd7 : SUCCESS in 1h 32m 39s","accounts_in_message":[],"_revision_number":2},{"id":"707794c82fa811dad30fb01d70d522264d4c40c0","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-10-29 20:55:35.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"d98ad703e07b91c8cde18886636630acf83356ee","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2020-10-30 09:17:42.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"20f030e70381db5f0a39b2f0509e5c2ca4f04ba2","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2020-10-30 09:21:44.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"eb01b434a436d5b2666591ebd0e3cb2cc0205ac6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-30 09:21:56.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"71d97676bcbd20a18f103ba122e79c9e8aaea49c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-30 11:45:30.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/80422d5649664cc6a8ca126989a0a5dd : SUCCESS in 4m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/44c44cdf2ec8439886a249ac6f71cac7 : SUCCESS in 5m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3756148bd2b74a559d8fb9fcf6a05c4d : SUCCESS in 3m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8ca19c10ab404773b2e0faea8ea00971 : SUCCESS in 5m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3f687f721454fad856154bf3c3812d3 : SUCCESS in 5m 22s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/fa2336e966d74f26ba8f5ad7183119ac : SUCCESS in 6m 06s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/4129f1c0a87d40b8a211589dfc19b459 : RETRY_LIMIT in 18m 29s","accounts_in_message":[],"_revision_number":2},{"id":"2ab950fe55a35c1f5df75c4ba05af28fc79d16c5","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2020-10-30 11:46:26.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"875fe98898c58ec4fdf95011da023730f800fe09","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-30 13:23:59.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2e726941e557475198650c89009c4954 : SUCCESS in 4m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d872ccbf0b644a15abfc256298f04de3 : SUCCESS in 4m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4aa52ff6c4494006a99730ae284080ba : SUCCESS in 3m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e13bc1bd5f6e45d9a2d2f5445f27b081 : SUCCESS in 4m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e5aa0554deb4612be537d0ef8d29345 : SUCCESS in 6m 21s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/8b95cb9a40de471fb6924c2a6cbbd3d8 : SUCCESS in 5m 46s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/d3971f795ccb4e019b8099b4b8de0445 : SUCCESS in 1h 29m 59s","accounts_in_message":[],"_revision_number":2},{"id":"ad4eb14bbd8ac027248e791adc402f0cfa14249c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-30 13:24:11.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"c74018a603171d578733b5ae0a1d9e046d6484d1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-30 14:52:07.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"1a1a76632fee0442b99de94af542cba81d8c66c4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-30 14:52:07.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/820edc77d9dc4a9b891a27f52106a59c : SUCCESS in 5m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f068fbfea4a4957b10a1408597d9064 : SUCCESS in 3m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/341880917ef448469efb2a2a90cc4c35 : SUCCESS in 3m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/49eebf6afd984864ab39246c796f7a11 : SUCCESS in 4m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4eeef9721f1f4d1eb9597b433d43c4ec : SUCCESS in 6m 37s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/b6a9e9be659d4cbb98219dc9e7a9eec5 : SUCCESS in 5m 57s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/35d32032884241ce9c52703f2f6c62ab : SUCCESS in 1h 15m 32s","accounts_in_message":[],"_revision_number":2},{"id":"b4edf1bc5a710276c7a409092e9b755700ef7151","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-30 14:53:44.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4335213f4d0c40d6b9d8b3af0da9bb70 : SUCCESS in 1m 14s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b49239d02408065e7e16ae5085c27df77ec4ac57","revisions":{"f1f11b596e690c053ae1674cf450fe770d5b3465":{"kind":"REWORK","_number":1,"created":"2020-10-29 16:57:51.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/14/760414/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/14/760414/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/14/760414/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/14/760414/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/14/760414/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/14/760414/1"}}},"commit":{"parents":[{"commit":"9168ddcfa8a6cef68f89252dbda5173f58ec7c5c","subject":"Add lock when calling idl.run()","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/9168ddcfa8a6cef68f89252dbda5173f58ec7c5c"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-09-15 18:42:08.000000000","tz":-300},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-10-29 16:57:38.000000000","tz":-300},"subject":"Don\u0027t give up when an Exception happens in idl.run","message":"Don\u0027t give up when an Exception happens in idl.run\n\nIt\u0027s possible that idl.run() could have a bug where it raises an\nException for an extended period of time while ovsdb-server is\ndown, but recover once ovsdb-server comes back up. Specifically,\npython-ovs currently doesn\u0027t properly catch an exception when the\nsocket type is \u0027ssl\u0027 that it catches for other protocols.\n\nChange-Id: Ia068650d2db3d5d8642771a6df5a260d692aea20\nCloses-Bug: #1895727\n(cherry picked from commit 83cf7aa6c81f1b2341b2bba1fe156047fa5d29f6)\n(cherry picked from commit 4807809ba7ec09a7e9cf533e334de282e0d373cd)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/f1f11b596e690c053ae1674cf450fe770d5b3465"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/f1f11b596e690c053ae1674cf450fe770d5b3465"}]},"branch":"refs/heads/stable/ussuri"},"b49239d02408065e7e16ae5085c27df77ec4ac57":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-10-29 16:58:58.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/14/760414/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/14/760414/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/14/760414/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/14/760414/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/14/760414/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/14/760414/2"}}},"commit":{"parents":[{"commit":"9168ddcfa8a6cef68f89252dbda5173f58ec7c5c","subject":"Add lock when calling idl.run()","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/9168ddcfa8a6cef68f89252dbda5173f58ec7c5c"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-09-15 18:42:08.000000000","tz":-300},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-10-29 16:58:31.000000000","tz":-300},"subject":"Don\u0027t give up when an Exception happens in idl.run","message":"Don\u0027t give up when an Exception happens in idl.run\n\nIt\u0027s possible that idl.run() could have a bug where it raises an\nException for an extended period of time while ovsdb-server is\ndown, but recover once ovsdb-server comes back up. Specifically,\npython-ovs currently doesn\u0027t properly catch an exception when the\nsocket type is \u0027ssl\u0027 that it catches for other protocols.\n\nConflicts:\n  ovsdbapp/backend/ovs_idl/connection.py\n\nChange-Id: Ia068650d2db3d5d8642771a6df5a260d692aea20\nCloses-Bug: #1895727\n(cherry picked from commit 83cf7aa6c81f1b2341b2bba1fe156047fa5d29f6)\n(cherry picked from commit 4807809ba7ec09a7e9cf533e334de282e0d373cd)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/b49239d02408065e7e16ae5085c27df77ec4ac57"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/b49239d02408065e7e16ae5085c27df77ec4ac57"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
