)]}'
{"id":"openstack%2Fovsdbapp~762562","triplet_id":"openstack%2Fovsdbapp~master~Ib071dd0f4d9c6e08fafc282c7551086de4b59500","project":"openstack/ovsdbapp","branch":"master","topic":"bug/1901527","hashtags":[],"change_id":"Ib071dd0f4d9c6e08fafc282c7551086de4b59500","subject":"Add update_tables method to Idl","status":"MERGED","created":"2020-11-12 18:05:02.000000000","updated":"2020-11-13 15:14:32.000000000","submitted":"2020-11-13 15:13:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"762562-1605280403624-dbd96a02","meta_rev_id":"96e1ca41c3b2b0cfa2c59316009464e408650964","_number":762562,"virtual_id_number":762562,"owner":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-11-13 15:13:23.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2020-11-13 13:41:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-11-13 13:46:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-11-13 13:46:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-13 13:41:17.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2020-11-13 13:46:21.000000000","updated_by":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"reviewer":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"state":"REVIEWER"},{"updated":"2020-11-13 15:13:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5d2a8f40360fe0de2c7649da9b986a2074b53e2b","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2020-11-12 18:05:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2c657c725f72181e17eb27a1f9396d769f697d0f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-12 19:25:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3c3064b26796418c9c870e6f64a1b0db : SUCCESS in 4m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0296438de4104ccf8cd6fc0a38a7a286 : SUCCESS in 4m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/63d7f3440014482898824d058046a669 : SUCCESS in 3m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7c8092377f614532bc40d7cab9dab33e : SUCCESS in 3m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0cc210a645244309a6af88ec76de0dd4 : SUCCESS in 3m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c21e9606b064a78b7ba5c07984f59bc : SUCCESS in 4m 45s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/5fd1d7c69225478f801f2f80ac984430 : SUCCESS in 5m 08s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/6ceca56dd2d34c8b96f42e2ad3f2f4d0 : SUCCESS in 1h 13m 34s","accounts_in_message":[],"_revision_number":1},{"id":"13e2aa105605c820b3995b186d159dd41c9142f6","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-11-12 19:44:20.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e6125f8fa3c00a82baa384ba8afaa39e5f075374","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-11-12 19:44:43.000000000","message":"Patch Set 1:\n\nnice, thanks!","accounts_in_message":[],"_revision_number":1},{"id":"52f7a9ea82136153d83a20b8525012e0b8feb9a8","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-11-13 09:07:44.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"62086a4e0fbe7fef15dea6e8ca317c2ea965ac44","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2020-11-13 09:10:44.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"b51c401ed73804beb02b5c82d9a85c3b8f670411","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2020-11-13 09:15:24.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"4ecbc08eb72e3cc590c5c1ab8194afb9143786dc","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-11-13 09:29:50.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8fc87f914fc3c14521bcf01139762cffba00e32a","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2020-11-13 09:39:22.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"998514967aa10fa36053412e1e14814a1c3573b1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-13 10:55:59.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/54a345738e404cf09db03247fdda20c2 : SUCCESS in 3m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2636a4130e064e7dbea5505ab32b40e5 : SUCCESS in 4m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4d1096153fd24314ac4868733916021d : SUCCESS in 3m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/eecc87b2df164482ab11cb6388927ed8 : SUCCESS in 3m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2377b1badf6441f5b24ab71f1db8d050 : SUCCESS in 3m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b83902a49ff940858c85f255eea651aa : SUCCESS in 4m 51s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/d403b079f72b4d09b20c996860352d81 : SUCCESS in 6m 01s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/cb9f33d3f82b4bd8a2c9e10658837353 : SUCCESS in 1h 36m 49s","accounts_in_message":[],"_revision_number":2},{"id":"c84e808e9d22949fb8bb59903c808a613f898b36","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-11-13 13:41:17.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6faf1c71de26ae41857aafc9d0cdcb86578c4e4a","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-11-13 13:46:21.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"6d96f5f747bfc965124d516898d3e5c0f37b4299","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-13 13:46:33.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"aca23cf5f8424fb8f4682ceb2993d78848c2c9cf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-13 15:13:23.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/14042ec249194808bd66e3c969d20b80 : SUCCESS in 3m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/978019045e5142a9b46bef724700b8ee : SUCCESS in 3m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a7780a1410ff4d3a85d45053057931fa : SUCCESS in 3m 06s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/872051a798054c10af9722d842ff33b0 : SUCCESS in 3m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e96d3c612a204512897ba35ee0084b8f : SUCCESS in 5m 37s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/ecbe2d519c84443d96310ea60a56aa74 : SUCCESS in 6m 45s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/e30cc61c0387418389075c76d0313dd2 : SUCCESS in 1h 20m 34s","accounts_in_message":[],"_revision_number":2},{"id":"38bb24d5a34d3bd9fd6114702744ab23298e81a7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-13 15:13:23.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"1f7a244e23907140580c951b31806b78621cfc55","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-13 15:14:32.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0931143e3772438d86e1ba9d25c0f7fd : SUCCESS in 45s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e36f3270e67d899848341e205ec2f95addd48899","revisions":{"341e1e5402a4653a63b68c44f0550f6e58069573":{"kind":"REWORK","_number":1,"created":"2020-11-12 18:05:02.000000000","uploader":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"ref":"refs/changes/62/762562/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/62/762562/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/62/762562/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/62/762562/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/62/762562/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/62/762562/1"}}},"commit":{"parents":[{"commit":"3b7d243dd8b3c7bc199ace841a80f76a5872a5ed","subject":"Ensure lists of Commands can be stringfied","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/3b7d243dd8b3c7bc199ace841a80f76a5872a5ed"}]}],"author":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2020-11-12 17:52:06.000000000","tz":0},"committer":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2020-11-12 18:05:00.000000000","tz":0},"subject":"Add update_tables method to Idl","message":"Add update_tables method to Idl\n\nThe method adds given tables to the current connection schema. If the\ntable is not already there, it\u0027s added. If the table in current\nconnection doesn\u0027t exist in the provided schema, it\u0027s not added, which\nseems like it\u0027s been removed.\n\nThis behavior is handy for cases where ovsdb-server is changed to new\nschema.\n\nRelated-bug: #1901527\n\nChange-Id: Ib071dd0f4d9c6e08fafc282c7551086de4b59500\nCo-Authored-By: Terry Wilson \u003ctwilson@redhat.com\u003e\nSigned-off-by: Jakub Libosvar \u003clibosvar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/341e1e5402a4653a63b68c44f0550f6e58069573"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/341e1e5402a4653a63b68c44f0550f6e58069573"}]},"branch":"refs/heads/master"},"e36f3270e67d899848341e205ec2f95addd48899":{"kind":"REWORK","_number":2,"created":"2020-11-13 09:15:24.000000000","uploader":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"ref":"refs/changes/62/762562/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/62/762562/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/62/762562/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/62/762562/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/62/762562/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/62/762562/2"}}},"commit":{"parents":[{"commit":"3b7d243dd8b3c7bc199ace841a80f76a5872a5ed","subject":"Ensure lists of Commands can be stringfied","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/3b7d243dd8b3c7bc199ace841a80f76a5872a5ed"}]}],"author":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2020-11-12 17:52:06.000000000","tz":0},"committer":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2020-11-13 09:14:58.000000000","tz":0},"subject":"Add update_tables method to Idl","message":"Add update_tables method to Idl\n\nThe method adds given tables to the current connection schema. If the\ntable is not already there, it\u0027s added. If the table in current\nconnection doesn\u0027t exist in the provided schema, it\u0027s not added, which\nseems like it\u0027s been removed.\n\nThis behavior is handy for cases where ovsdb-server is changed to new\nschema.\n\nRelated-bug: #1901527\n\nChange-Id: Ib071dd0f4d9c6e08fafc282c7551086de4b59500\nCo-Authored-By: Terry Wilson \u003ctwilson@redhat.com\u003e\nSigned-off-by: Jakub Libosvar \u003clibosvar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/e36f3270e67d899848341e205ec2f95addd48899"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/e36f3270e67d899848341e205ec2f95addd48899"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
