)]}'
{"id":"openstack%2Fovsdbapp~762730","triplet_id":"openstack%2Fovsdbapp~master~I7f31ee368ad26acd80ebbbd5ffce2fc0fc5eb25a","project":"openstack/ovsdbapp","branch":"master","topic":"bug/1888878","hashtags":[],"change_id":"I7f31ee368ad26acd80ebbbd5ffce2fc0fc5eb25a","subject":"Add NB_Global get method","status":"ABANDONED","created":"2020-11-14 09:42:32.000000000","updated":"2021-01-08 09:30:06.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"cf2dc6ab9dfd57d3998c8ad18c498437fa5b4ff2","_number":762730,"virtual_id_number":762730,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-11-14 18:26:31.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"date":"2020-11-16 08:36:48.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-11-16 08:36:37.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-14 18:26:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-16 08:36:37.000000000","updated_by":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"reviewer":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"state":"REVIEWER"},{"updated":"2020-11-16 08:36:48.000000000","updated_by":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"reviewer":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"state":"REVIEWER"}],"messages":[{"id":"c2b7aa0927aba222df6bd32cdcd6596550ac5c53","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-11-14 09:42:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9df91e910b2c9756a97b99ddc14a113079441fea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-14 11:14:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4a4a01f25e644bc08f66537410c9c545 : SUCCESS in 5m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f666acf05b544167bbe5bda5d79028d0 : SUCCESS in 4m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d4a0fc33d4f14c0cb5ea59fa5883adf0 : SUCCESS in 3m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/59c8ed29f40a4f8b9464bc9525422b18 : SUCCESS in 4m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7cae17e21b004236bdbac81c7f6ea838 : SUCCESS in 3m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56f3badd5c41435badbbc90d8d61b53c : SUCCESS in 5m 11s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/a7c8f9a4278846039b663d0a9dcf5cd9 : SUCCESS in 4m 48s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/1d84fddd22c14fbc8b8c4019766f5336 : SUCCESS in 1h 29m 35s","accounts_in_message":[],"_revision_number":1},{"id":"0453c6fbd12e0bd5f17e2293047d8662e3ad7cc9","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-11-14 15:53:05.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"15f0c690463f3e5abb37bd90ac23c3e7e0eefa93","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-11-14 17:01:13.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"d2390a90e10ca15818cb72434856effa41536998","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-14 18:26:31.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/48cd91fd4c2e4d049050e705965991f8 : SUCCESS in 3m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/24b1df7962094c748be4bb279de817e8 : SUCCESS in 5m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0410e2d15e1d45e894d60ef0d7c20440 : SUCCESS in 4m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5d4985505d5b416395b31859e186d102 : SUCCESS in 3m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/eaf51211ce404903b34111da5aca2d2b : SUCCESS in 4m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de38ad4a1ddb4591a4896103895f2fd6 : SUCCESS in 4m 38s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/c3bbb56e7844448080acc9b104ae1593 : SUCCESS in 6m 16s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/25ec31d4b85043ffa58d9efe08e8271d : SUCCESS in 1h 22m 59s","accounts_in_message":[],"_revision_number":3},{"id":"cf2dc6ab9dfd57d3998c8ad18c498437fa5b4ff2","tag":"autogenerated:gerrit:abandon","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-01-08 09:30:06.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"d5278cf79b3d66cb76cab1142dd6d975a97b6df1","revisions":{"01c49f381a29c585e22d101f46e9eb35c385c462":{"kind":"REWORK","_number":1,"created":"2020-11-14 09:42:32.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/30/762730/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/30/762730/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/30/762730/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/30/762730/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/30/762730/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/30/762730/1"}}},"commit":{"parents":[{"commit":"e36f3270e67d899848341e205ec2f95addd48899","subject":"Add update_tables method to Idl","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/e36f3270e67d899848341e205ec2f95addd48899"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-11-14 09:40:48.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-11-14 09:40:48.000000000","tz":0},"subject":"[WIP] Add NB Global get parameter method","message":"[WIP] Add NB Global get parameter method\n\nSTILL UNDER DEVELOPMENT\n\nChange-Id: I7f31ee368ad26acd80ebbbd5ffce2fc0fc5eb25a\nRelated-Bug: #1888878\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/01c49f381a29c585e22d101f46e9eb35c385c462"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/01c49f381a29c585e22d101f46e9eb35c385c462"}]},"branch":"refs/heads/master"},"3e2d5bfc9368aa88b0b0ace2858b5b27418ce081":{"kind":"REWORK","_number":2,"created":"2020-11-14 15:53:05.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/30/762730/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/30/762730/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/30/762730/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/30/762730/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/30/762730/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/30/762730/2"}}},"commit":{"parents":[{"commit":"e36f3270e67d899848341e205ec2f95addd48899","subject":"Add update_tables method to Idl","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/e36f3270e67d899848341e205ec2f95addd48899"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-11-14 09:40:48.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-11-14 15:45:52.000000000","tz":0},"subject":"Add NB_Global get parameter method","message":"Add NB_Global get parameter method\n\nRetrieve a parameter from the NB_Global single row. This table must\nhave only one single register.\n\nIn case the parameter given is not present in the row,\n\"AttributeError\" exception will be raisen.\n\nThis method is based on \"BaseCommand\". The client using this\nimplementation will be constrain to the OVSDB connection lock\nenforced in the API, instead of accessing directly to the\nin-memory DB cache. That could lead to sync errors as reported\nin the related bug.\n\nChange-Id: I7f31ee368ad26acd80ebbbd5ffce2fc0fc5eb25a\nRelated-Bug: #1888878\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/3e2d5bfc9368aa88b0b0ace2858b5b27418ce081"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/3e2d5bfc9368aa88b0b0ace2858b5b27418ce081"}]},"branch":"refs/heads/master"},"d5278cf79b3d66cb76cab1142dd6d975a97b6df1":{"kind":"REWORK","_number":3,"created":"2020-11-14 17:01:13.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/30/762730/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/30/762730/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/30/762730/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/30/762730/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/30/762730/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/30/762730/3"}}},"commit":{"parents":[{"commit":"e36f3270e67d899848341e205ec2f95addd48899","subject":"Add update_tables method to Idl","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/e36f3270e67d899848341e205ec2f95addd48899"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-11-14 09:40:48.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-11-14 16:59:31.000000000","tz":0},"subject":"Add NB_Global get method","message":"Add NB_Global get method\n\nRetrieves, from the NB_Global table, the first and only row that\nshould be present. If \"parameter\" is passed, only the column\nmatching to this input variable will be returned.\n\nIn case the parameter given is not present in the row,\n\"AttributeError\" exception will be raisen.\n\nThis method is based on \"BaseCommand\". The client using this\nimplementation will be constrain to the OVSDB connection lock\nenforced in the API, instead of accessing directly to the\nin-memory DB cache. That could lead to sync errors as reported\nin the related bug.\n\nChange-Id: I7f31ee368ad26acd80ebbbd5ffce2fc0fc5eb25a\nRelated-Bug: #1888878\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/d5278cf79b3d66cb76cab1142dd6d975a97b6df1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/d5278cf79b3d66cb76cab1142dd6d975a97b6df1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
