)]}'
{"id":"openstack%2Fovsdbapp~767574","triplet_id":"openstack%2Fovsdbapp~master~I076269d3d666caa5ebe3d70111136c78393c7c5f","project":"openstack/ovsdbapp","branch":"master","topic":"ovn_sg_log","hashtags":[],"change_id":"I076269d3d666caa5ebe3d70111136c78393c7c5f","subject":"Add northbound API support for meters","status":"MERGED","created":"2020-12-17 15:53:40.000000000","updated":"2021-01-13 18:24:22.000000000","submitted":"2021-01-13 18:22:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"767574-ovn_sg_log","meta_rev_id":"dde377195d6258f1d2e3667be25de5c352cf015a","_number":767574,"virtual_id_number":767574,"owner":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-13 18:22:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":2,"date":"2021-01-13 09:36:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-01-08 17:24:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":1,"date":"2021-01-13 09:36:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"reviewers":{"REVIEWER":[{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-17 15:54:59.000000000","updated_by":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"reviewer":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"state":"REVIEWER"},{"updated":"2020-12-17 15:54:59.000000000","updated_by":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"reviewer":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"state":"REVIEWER"},{"updated":"2020-12-17 15:54:59.000000000","updated_by":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"reviewer":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"state":"REVIEWER"},{"updated":"2020-12-17 16:59:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-13 09:36:03.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"5a88070c0af3e6872aaf7b8a1b5905b228fac603","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-12-17 15:53:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c0a0b448de8adc8d14871734764bb463e67361a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-17 16:59:58.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6c30f2256dce4586b98d3d911f3a5cfa : FAILURE in 3m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c327413c43cf4f4b88e4c058ea71b1e0 : SUCCESS in 3m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4be106d97b9e4110b1c472db652e5365 : SUCCESS in 3m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/64740e98d7d746aea9f637e46a883936 : SUCCESS in 3m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/37ee12d0171d4cee85ae0aa28b5f3538 : SUCCESS in 4m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/688c251c9b40449e99d8d068b26c4eb4 : SUCCESS in 4m 53s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/d50ff0634f41447bad27b37a1afa2bbb : SUCCESS in 5m 52s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/0e702e07664d4d87b856184a6dfb47ce : SUCCESS in 1h 02m 38s","accounts_in_message":[],"_revision_number":1},{"id":"fd6eef8b269c3caad75e44d98a05af9e7c78d16d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-12-18 21:33:02.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"226b92ab3d629083ed940c076ccfaba396ef9f1c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-12-18 21:37:08.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"33fe314957b220ef18ad0821cacbf6311bb38d88","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-18 23:00:34.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/832fb16ce229475797be09b7b2dfa1c4 : SUCCESS in 3m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/96fb7ae6099440638767b88652b9bd2f : SUCCESS in 3m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/41757725faf5490aa7c393789cdcc2f5 : SUCCESS in 3m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3785817cf4224756a8a2f139f48233e2 : SUCCESS in 3m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7a0727062b66462eaae9b252893363b3 : SUCCESS in 3m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e5064554ddeb4778aa8274eefac24b2e : SUCCESS in 5m 11s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/7f5047dcff5b4804a746b54bd8b663d7 : SUCCESS in 6m 44s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/ba957e6a71d3416d832f4afdd03b1a71 : SUCCESS in 1h 06m 42s","accounts_in_message":[],"_revision_number":3},{"id":"824a716fbd1a216161c87dfff72d329b88dd38fc","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-12-21 20:57:59.000000000","message":"Topic set to ovn_sg_log","accounts_in_message":[],"_revision_number":3},{"id":"e74ea3f507757e2ec32c4c6e0e6edaeea859e1ba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2021-01-05 14:46:30.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"3c37a208bfde9272e568cdf23a135ff0c1d49b7d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-05 16:11:27.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4997a6487eb94356b3efb77501a9d163 : SUCCESS in 3m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4443d9af275140d7ace587062a2b4492 : SUCCESS in 3m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/62232413d9ad4a5da67789c3ddd4591a : SUCCESS in 4m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0de46867c8304453a0027557c27bb9cb : SUCCESS in 3m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fdfe56a449e748868e085c8c9f5c2db5 : SUCCESS in 3m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a1bba2bcb6234d77aaf7ef2fcc5500e9 : SUCCESS in 5m 17s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/b4f214a710014d19bbce23fbdd2e6bd8 : SUCCESS in 5m 46s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/ee31f52323a74fb4bc8a9a18268c9240 : SUCCESS in 1h 17m 37s","accounts_in_message":[],"_revision_number":4},{"id":"c5d5dfb2430821fd158076e05c3f299a5a28453c","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2021-01-07 17:32:44.000000000","message":"Patch Set 4:\n\n(3 comments)\n\nJust a couple of fairly minor questions. Patch looks very nice. Great work!","accounts_in_message":[],"_revision_number":4},{"id":"97484328a6b4cceaf9769497d40b1c7761618e7a","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2021-01-07 23:13:36.000000000","message":"Patch Set 4:\n\n(4 comments)\n\nTY Terry for the awesome feedback! Please see my follow up responses, as well as the upcoming revision.","accounts_in_message":[],"_revision_number":4},{"id":"b576a5fcef5524df7a8df13103062fed9b3f6901","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2021-01-07 23:15:57.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"8c35ee3592aa09b253dc6a5222fc44573effe996","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-08 00:27:29.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cb3ee76812cb486b951baf3829b27c85 : SUCCESS in 3m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b1f8e2993ae4585915adb438356963d : SUCCESS in 3m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/faac7cb0d4d7498f9fa0934f983f1385 : SUCCESS in 4m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e16cab9916ae485cb3781afc8d024236 : SUCCESS in 3m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/52231877351a421b89428622f5e6247b : SUCCESS in 3m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a1d4fa49ed354344b739632afaca3429 : SUCCESS in 4m 43s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/f7357b5014eb4479a530e2b7e32dbab7 : SUCCESS in 7m 25s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/90147be0d0e845d1a207727e3164c3b8 : SUCCESS in 1h 02m 45s","accounts_in_message":[],"_revision_number":5},{"id":"08b46d4597bbd3a5e361294ecbc187f13c752c8a","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2021-01-08 17:24:53.000000000","message":"Patch Set 5: Code-Review+2\n\nLooks good to me, thanks!","accounts_in_message":[],"_revision_number":5},{"id":"d4f6618b95b54a39fc094eee2b048a88e6314169","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-01-13 09:36:03.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"63bb49eb31a9aaca53c7b47d30c06b9338c70f28","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-13 09:36:17.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"b62281886f86e3b199a6ba42e555ec3ed9d2dcba","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-13 10:36:55.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/31f49a8720804f6eaca19d58f5259d91 : SUCCESS in 4m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/67bdbfd762f445429404584efe174dc4 : SUCCESS in 3m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5e52d5641b0a49cfa6f25bc59ff570b0 : SUCCESS in 5m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5892516860fd47f78b6da5bf7356f81e : SUCCESS in 5m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/49f50b8da0194f088116e2ccfc39bb6b : SUCCESS in 12m 47s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/d630d3765efc4247b584ef0d3262eb5c : SUCCESS in 7m 34s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/4390ba8884a14fdfb6ba15ef3c29d380 : FAILURE in 4m 25s","accounts_in_message":[],"_revision_number":5},{"id":"256a50285d5859616f6f1a2dc6eea8d723af7691","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2021-01-13 15:32:56.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"f78ad1d1df2b23831837d746defa59c9dcc08add","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-13 16:46:55.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b6f5f8b9d48d4d40b2a0e984793f2eee : SUCCESS in 3m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/18a4e9674c0f47e4bdc6ee6c630d1b85 : SUCCESS in 3m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7e5a7a6af6694d16bdc4e42cfe9bc5b0 : SUCCESS in 3m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f3ca9ed5cf4c49afb3d49a9a5bcde1f2 : SUCCESS in 3m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7b27ca3dc9f046108b832684594d3d9c : SUCCESS in 4m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d11b48d5769e42b3a3666e0ab47c6839 : SUCCESS in 6m 04s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/029ab17f6b61475fb8fccad593ead500 : SUCCESS in 5m 43s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/4514fb38998d4d8e9544dce36945e04b : SUCCESS in 1h 07m 43s","accounts_in_message":[],"_revision_number":5},{"id":"b78f93841c33d1f91e5ae43247ac6a14ef0f8cc3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-13 16:47:13.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"c57feb17d53f274d962ee989056b359c6b073eb6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-13 18:22:49.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1c7c135f608a45568b70f3bf2a48d2a5 : SUCCESS in 4m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5da0748f5b0442bb85b8555c8cc2cfa4 : SUCCESS in 3m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8e96608318d149708c28db3a7497ffb1 : SUCCESS in 4m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7a9ff8276aac4d54bcc9ad612cdd57a5 : SUCCESS in 4m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df3d815d9d574bbf9792d63112a7455c : SUCCESS in 5m 44s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/a020ea8eddf74fceb064d7a66f22a401 : SUCCESS in 6m 27s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/f5a1d49c0a574f48ad6589aeebf54ef7 : SUCCESS in 1h 31m 35s","accounts_in_message":[],"_revision_number":5},{"id":"c87ad57d2db52f59e1fb7b7dc1810054f8518691","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-13 18:22:50.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"dde377195d6258f1d2e3667be25de5c352cf015a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-13 18:24:22.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd987057643947f5b9e73924ec17b795 : SUCCESS in 54s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"e2f782cd727506de3dbc2a32cc49920b0c792c0f","revisions":{"717f27d4aeeaadc5efc97553cdce7b09b8a1ca74":{"kind":"REWORK","_number":1,"created":"2020-12-17 15:53:40.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/74/767574/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/74/767574/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/74/767574/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/74/767574/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/74/767574/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/74/767574/1"}}},"commit":{"parents":[{"commit":"e36f3270e67d899848341e205ec2f95addd48899","subject":"Add update_tables method to Idl","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/e36f3270e67d899848341e205ec2f95addd48899"}]}],"author":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-12-17 15:43:41.000000000","tz":-300},"committer":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-12-17 15:49:43.000000000","tz":-300},"subject":"Add northbound API support for meters","message":"Add northbound API support for meters\n\nAdd meter (name) attribute to the ACL row, via sg_acl_add().\n\nAdd helpers for adding, removing, getting and listing rows of the\nMeter table.\n\nChange-Id: I076269d3d666caa5ebe3d70111136c78393c7c5f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/717f27d4aeeaadc5efc97553cdce7b09b8a1ca74"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/717f27d4aeeaadc5efc97553cdce7b09b8a1ca74"}]},"branch":"refs/heads/master"},"5e99aefc8734607d9c673b3f60d5c5b9774cda57":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-12-18 21:33:02.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/74/767574/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/74/767574/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/74/767574/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/74/767574/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/74/767574/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/74/767574/2"}}},"commit":{"parents":[{"commit":"c2abfa6cf126a75cd33ed9f102f2c150a747fdbc","subject":"Fix gate failure","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/c2abfa6cf126a75cd33ed9f102f2c150a747fdbc"}]}],"author":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-12-17 15:43:41.000000000","tz":-300},"committer":{"name":"Flavio Fernandes","email":"flavio@flaviof.com","date":"2020-12-18 21:33:02.000000000","tz":0},"subject":"Add northbound API support for meters","message":"Add northbound API support for meters\n\nAdd meter (name) attribute to the ACL row, via sg_acl_add().\n\nAdd helpers for adding, removing, getting and listing rows of the\nMeter table.\n\nChange-Id: I076269d3d666caa5ebe3d70111136c78393c7c5f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/5e99aefc8734607d9c673b3f60d5c5b9774cda57"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/5e99aefc8734607d9c673b3f60d5c5b9774cda57"}]},"branch":"refs/heads/master","description":"Rebase"},"ff4c3765f3a62c33e9564817daa8f066f05c5fbd":{"kind":"REWORK","_number":3,"created":"2020-12-18 21:37:08.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/74/767574/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/74/767574/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/74/767574/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/74/767574/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/74/767574/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/74/767574/3"}}},"commit":{"parents":[{"commit":"c2abfa6cf126a75cd33ed9f102f2c150a747fdbc","subject":"Fix gate failure","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/c2abfa6cf126a75cd33ed9f102f2c150a747fdbc"}]}],"author":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-12-17 15:43:41.000000000","tz":-300},"committer":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-12-18 21:37:04.000000000","tz":-300},"subject":"Add northbound API support for meters","message":"Add northbound API support for meters\n\nAdd meter (name) attribute to the ACL row, via sg_acl_add().\n\nAdd helpers for adding, removing, getting and listing rows of the\nMeter table.\n\nChange-Id: I076269d3d666caa5ebe3d70111136c78393c7c5f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/ff4c3765f3a62c33e9564817daa8f066f05c5fbd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/ff4c3765f3a62c33e9564817daa8f066f05c5fbd"}]},"branch":"refs/heads/master"},"f52d261ba2a57fe7807ef2708cd26ceee35bc033":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2021-01-05 14:46:30.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/74/767574/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/74/767574/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/74/767574/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/74/767574/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/74/767574/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/74/767574/4"}}},"commit":{"parents":[{"commit":"0968b410ce13b6420131b28b4d09bdd16e306853","subject":"Fix gate failure","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/0968b410ce13b6420131b28b4d09bdd16e306853"}]}],"author":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-12-17 15:43:41.000000000","tz":-300},"committer":{"name":"Flavio Fernandes","email":"flavio@flaviof.com","date":"2021-01-05 14:46:30.000000000","tz":0},"subject":"Add northbound API support for meters","message":"Add northbound API support for meters\n\nAdd meter (name) attribute to the ACL row, via sg_acl_add().\n\nAdd helpers for adding, removing, getting and listing rows of the\nMeter table.\n\nChange-Id: I076269d3d666caa5ebe3d70111136c78393c7c5f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/f52d261ba2a57fe7807ef2708cd26ceee35bc033"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/f52d261ba2a57fe7807ef2708cd26ceee35bc033"}]},"branch":"refs/heads/master","description":"Rebase"},"e2f782cd727506de3dbc2a32cc49920b0c792c0f":{"kind":"REWORK","_number":5,"created":"2021-01-07 23:15:57.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/74/767574/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/74/767574/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/74/767574/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/74/767574/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/74/767574/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/74/767574/5"}}},"commit":{"parents":[{"commit":"0968b410ce13b6420131b28b4d09bdd16e306853","subject":"Fix gate failure","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/0968b410ce13b6420131b28b4d09bdd16e306853"}]}],"author":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-12-17 15:43:41.000000000","tz":-300},"committer":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2021-01-07 23:14:20.000000000","tz":-300},"subject":"Add northbound API support for meters","message":"Add northbound API support for meters\n\nAdd meter (name) attribute to the ACL row, via sg_acl_add().\n\nAdd helpers for adding, removing, getting and listing rows of the\nMeter table.\n\nChange-Id: I076269d3d666caa5ebe3d70111136c78393c7c5f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/e2f782cd727506de3dbc2a32cc49920b0c792c0f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/e2f782cd727506de3dbc2a32cc49920b0c792c0f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
