)]}'
{"id":"openstack%2Fovsdbapp~769448","triplet_id":"openstack%2Fovsdbapp~master~Iebdb417bf2147b13d871409f58d6d9ea8183895f","project":"openstack/ovsdbapp","branch":"master","hashtags":[],"change_id":"Iebdb417bf2147b13d871409f58d6d9ea8183895f","subject":"enhancing lb_del so that removal of lb can optionally take logical routers","status":"ABANDONED","created":"2021-01-06 03:29:22.000000000","updated":"2021-01-21 02:55:48.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"2c26ae6ef858681521e6b49fb2430295022c3ffb","_number":769448,"virtual_id_number":769448,"owner":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2021-01-06 04:41:43.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-06 04:41:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4912023a1572ac275e2570d733a434b3ca6964e2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"date":"2021-01-06 03:29:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"22afefe17b981a20e48c829f332ee34eb5c77dc1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"date":"2021-01-06 03:32:12.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"97c42f2448e0472148210d41171bbcd5a950f5d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-06 04:41:43.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/768abcc08daf4193bc48f4342eea7c35 : SUCCESS in 3m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae99fa625c964cd78efb180287cbd07c : SUCCESS in 3m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/27ff2cdb459647d6a63fa49e552131d1 : SUCCESS in 3m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/95f626d71e1a4cafbd01a8c750dec13b : SUCCESS in 3m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ecbc440b78da4e99bc792a6b6746eee0 : SUCCESS in 3m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e7ccd144eab44ae94660f06da8c2bca : SUCCESS in 4m 35s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/217a5dd62e0f444ab2ab03659dead786 : SUCCESS in 5m 33s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/c93ef54824fa465491bc6a6cda1fcce3 : SUCCESS in 1h 04m 15s","accounts_in_message":[],"_revision_number":2},{"id":"2c26ae6ef858681521e6b49fb2430295022c3ffb","tag":"autogenerated:gerrit:abandon","author":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"date":"2021-01-21 02:55:48.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"6b6e878580acd4e4afbeaf00200c75213b5a53b9","revisions":{"5325724a8d34adf9ab244ae58da265eef0c4e78a":{"kind":"REWORK","_number":1,"created":"2021-01-06 03:29:22.000000000","uploader":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"ref":"refs/changes/48/769448/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/48/769448/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/48/769448/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/48/769448/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/48/769448/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/48/769448/1"}}},"commit":{"parents":[{"commit":"0968b410ce13b6420131b28b4d09bdd16e306853","subject":"Fix gate failure","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/0968b410ce13b6420131b28b4d09bdd16e306853"}]}],"author":{"name":"zhouhenglc","email":"zhouhenglc@inspur.com","date":"2021-01-06 03:16:00.000000000","tz":480},"committer":{"name":"zhouhenglc","email":"zhouhenglc@inspur.com","date":"2021-01-06 03:29:15.000000000","tz":480},"subject":"enhancing lb_del so that removal of lb can optionally take logical routers","message":"enhancing lb_del so that removal of lb can optionally take logical routers\n\nno vips will delete lb, if somre routers ref to the lb, lb it will not be\ndeleted successfully\n\nChange-Id: Iebdb417bf2147b13d871409f58d6d9ea8183895f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/5325724a8d34adf9ab244ae58da265eef0c4e78a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/5325724a8d34adf9ab244ae58da265eef0c4e78a"}]},"branch":"refs/heads/master"},"6b6e878580acd4e4afbeaf00200c75213b5a53b9":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-01-06 03:32:12.000000000","uploader":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"ref":"refs/changes/48/769448/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/48/769448/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/48/769448/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/48/769448/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/48/769448/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/48/769448/2"}}},"commit":{"parents":[{"commit":"0968b410ce13b6420131b28b4d09bdd16e306853","subject":"Fix gate failure","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/0968b410ce13b6420131b28b4d09bdd16e306853"}]}],"author":{"name":"zhouhenglc","email":"zhouhenglc@inspur.com","date":"2021-01-06 03:16:00.000000000","tz":480},"committer":{"name":"zhouhenglc","email":"zhouhenglc@inspur.com","date":"2021-01-06 03:31:40.000000000","tz":480},"subject":"enhancing lb_del so that removal of lb can optionally take logical routers","message":"enhancing lb_del so that removal of lb can optionally take logical routers\n\nno vips will delete lb, if some routers ref to the lb, lb it will not be\ndeleted successfully\n\nChange-Id: Iebdb417bf2147b13d871409f58d6d9ea8183895f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/6b6e878580acd4e4afbeaf00200c75213b5a53b9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/6b6e878580acd4e4afbeaf00200c75213b5a53b9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
