)]}'
{"id":"openstack%2Fovsdbapp~770676","triplet_id":"openstack%2Fovsdbapp~master~I6e39d2059c1c8ba1960b49c7be5e6a4dc5516d69","project":"openstack/ovsdbapp","branch":"master","hashtags":[],"change_id":"I6e39d2059c1c8ba1960b49c7be5e6a4dc5516d69","subject":"[ovn nb] Acl add: Avoid setting columns that have no values","status":"MERGED","created":"2021-01-13 23:06:33.000000000","updated":"2021-01-14 19:04:36.000000000","submitted":"2021-01-14 19:02:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"770676","meta_rev_id":"34e9fe603ff87a652b2e96a75a2dde74422e8ed2","_number":770676,"virtual_id_number":770676,"owner":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-14 19:02:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"all":[{"value":0,"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"value":2,"date":"2021-01-14 15:13:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-01-14 15:12:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"all":[{"value":0,"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"value":1,"date":"2021-01-14 15:13:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-14 00:36:04.000000000","updated_by":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"reviewer":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"state":"REVIEWER"},{"updated":"2021-01-14 00:36:04.000000000","updated_by":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"reviewer":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"state":"REVIEWER"},{"updated":"2021-01-14 01:59:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"11a4701686feafa201fcab14de2b03d9b54ee02f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2021-01-13 23:06:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ea1ed65aa57efae076288cd50fb395e9af4dee6b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2021-01-13 23:17:50.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f8e76c51154e5a12ce36ccd35444ac0f5b95e25c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2021-01-14 00:35:35.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"3b95a65f66e8d75cdb2759097b1aa8f8926dbf2c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2021-01-14 00:45:28.000000000","message":"Uploaded patch set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"9c0336a9b3d46f981082879fb55670138d4ef174","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-14 01:59:29.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dc44cec2d3c94c11a5d4124eb4a905fc : SUCCESS in 6m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ba6fbea3e3442c78eb5aa678e2df630 : SUCCESS in 4m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cd98f813c88747f3bf6a7eddf9cac4d2 : SUCCESS in 4m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/279c655c9d064194bf1840fcd433e480 : SUCCESS in 4m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b0ab7cf3148448c0a4e577a04addabf9 : SUCCESS in 4m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/042c66353e5f41dc94e97cc1d573cfb2 : SUCCESS in 6m 54s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/b41d15143c284cb495b59d067145690d : SUCCESS in 7m 08s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/ede669ee7804413ba772566e16c288b6 : SUCCESS in 1h 10m 55s","accounts_in_message":[],"_revision_number":4},{"id":"cbfe3603100592d3c51696383c8a7634500ac996","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2021-01-14 10:03:55.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"a2a4dca47d48cad2482b9a6b990ee857f4778718","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2021-01-14 10:26:39.000000000","message":"Patch Set 4: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"a0a925676fcd00daa1acde036b94ca0fef68d2aa","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2021-01-14 10:29:09.000000000","message":"Patch Set 4:\n\n(1 comment)\n\nEven tho","accounts_in_message":[],"_revision_number":4},{"id":"666a4d53dbee6c76461a21790e110ba96360b0ff","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2021-01-14 10:33:34.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"6a0c93282acf92af03d84f91ace2cb5f9efdf342","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2021-01-14 15:09:24.000000000","message":"Patch Set 4: Code-Review+2\n\nAfter IRC conversation with Terry and Flavio, it was made clear we do cover this scenario and only thing that happens without this patch is emitting the error log message. I don\u0027t think we need to check we don\u0027t issue an error. The patch is good as is imho :) Thanks Flavio for your patience.","accounts_in_message":[],"_revision_number":4},{"id":"5a19181292a638e75e59691eea1f33ad214b3329","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2021-01-14 15:10:07.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"d9fbbd40416f563a09f7c1833140dff072c76e68","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2021-01-14 15:10:53.000000000","message":"Uploaded patch set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"98270d6897d38ab3dfb76f49528e93be6e4c550e","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2021-01-14 15:12:21.000000000","message":"Patch Set 5:\n\n\u003e Patch Set 4: Code-Review+2\n\u003e \n\u003e After IRC conversation with Terry and Flavio, it was made clear we do cover this scenario and only thing that happens without this patch is emitting the error log message. I don\u0027t think we need to check we don\u0027t issue an error. The patch is good as is imho :) Thanks Flavio for your patience.\n\nTY Kuba!","accounts_in_message":[],"_revision_number":5},{"id":"78c6b939319b1a80a890a51dcbb8abb46a400e1e","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2021-01-14 15:12:46.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"ef033fd0078ddf7e6c05e9ff3eff560b774c6435","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2021-01-14 15:13:25.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"d42f4e23d85b9abbc422c8cfd95e2dcbf6a4f263","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-14 17:03:53.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b8e871e4b2214ae6a09eb80d9f56fd13 : SUCCESS in 4m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b984e4895cf4b108e68f2f4d666e31f : SUCCESS in 7m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/86af00498044407380caef4029c2fef8 : SUCCESS in 7m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7fbf125a85954272a5df5edc3238fcb6 : SUCCESS in 4m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/30057603ca74487d91c55443aba4725b : SUCCESS in 4m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/147e87756a534be1a198b19dd2d35457 : SUCCESS in 5m 46s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/2c87e12e1344473eb5ada7b3e4e9b992 : SUCCESS in 7m 45s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/cfa0bd22f3404c03ac52c9785191a5c4 : SUCCESS in 1h 46m 25s","accounts_in_message":[],"_revision_number":5},{"id":"cc53a139a1b97d71ebcca18502d224cb47132ec5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-14 17:04:10.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"f4f89c744ca9c02320caf7dd7b563f5ec1d5dfe9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-14 19:02:24.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f6111f9e04364b6a8d4c3fc8817631ad : SUCCESS in 3m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/47fffe36d4ab49ca9e05da76957b2243 : SUCCESS in 3m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e948e74d28b947cf9bc3e5065c030124 : SUCCESS in 4m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d706751c0dc04a82b0c6f07207b8e370 : SUCCESS in 5m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6fb32dff3ec41fda947df26cc44e077 : SUCCESS in 7m 20s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/4791799dea1e44d78c164e494d2a5fd7 : SUCCESS in 6m 57s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/0432e5903f664cee9b124d04637e79b3 : SUCCESS in 1h 52m 19s","accounts_in_message":[],"_revision_number":5},{"id":"4b4163e951ebbef8231535dad6d64bce3d371637","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-14 19:02:25.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"34e9fe603ff87a652b2e96a75a2dde74422e8ed2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-14 19:04:36.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ce4c56cef6048b097182b53865a2271 : SUCCESS in 1m 24s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"aac2da8f0cb5adc006613e49d8d4fcf08531ef37","revisions":{"3db8125add6d783863e55e3d4f9dc5b3c62baf02":{"kind":"REWORK","_number":1,"created":"2021-01-13 23:06:33.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/76/770676/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/76/770676/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/76/770676/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/76/770676/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/76/770676/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/76/770676/1"}}},"commit":{"parents":[{"commit":"dadf4e6540bfce358df29de81794efbab46d80e1","subject":"Merge \"Add northbound API support for meters\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/dadf4e6540bfce358df29de81794efbab46d80e1"}]}],"author":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2021-01-13 23:02:51.000000000","tz":-300},"committer":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2021-01-13 23:03:32.000000000","tz":-300},"subject":"[ovn] AclAddHelper command: use [] instead None for string columns","message":"[ovn] AclAddHelper command: use [] instead None for string columns\n\nBe aware of None values before setting it to values in the Meter table.\n\nChange-Id: I6e39d2059c1c8ba1960b49c7be5e6a4dc5516d69\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/3db8125add6d783863e55e3d4f9dc5b3c62baf02"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/3db8125add6d783863e55e3d4f9dc5b3c62baf02"}]},"branch":"refs/heads/master"},"379465ab9294d0cf8d89e37333741817d0541149":{"kind":"REWORK","_number":2,"created":"2021-01-13 23:17:50.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/76/770676/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/76/770676/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/76/770676/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/76/770676/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/76/770676/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/76/770676/2"}}},"commit":{"parents":[{"commit":"dadf4e6540bfce358df29de81794efbab46d80e1","subject":"Merge \"Add northbound API support for meters\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/dadf4e6540bfce358df29de81794efbab46d80e1"}]}],"author":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2021-01-13 23:02:51.000000000","tz":-300},"committer":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2021-01-13 23:17:46.000000000","tz":-300},"subject":"[ovn] AclAddHelper command: use [] instead None for string columns","message":"[ovn] AclAddHelper command: use [] instead None for string columns\n\nBe aware of None values before setting it to values in the Meter table.\n\nChange-Id: I6e39d2059c1c8ba1960b49c7be5e6a4dc5516d69\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/379465ab9294d0cf8d89e37333741817d0541149"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/379465ab9294d0cf8d89e37333741817d0541149"}]},"branch":"refs/heads/master"},"14da3bdedfd43f2d9f4fd99a3b25aaa88d38bcff":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2021-01-14 00:35:35.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/76/770676/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/76/770676/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/76/770676/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/76/770676/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/76/770676/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/76/770676/3"}}},"commit":{"parents":[{"commit":"dadf4e6540bfce358df29de81794efbab46d80e1","subject":"Merge \"Add northbound API support for meters\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/dadf4e6540bfce358df29de81794efbab46d80e1"}]}],"author":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2021-01-13 23:02:51.000000000","tz":-300},"committer":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2021-01-14 00:34:20.000000000","tz":-300},"subject":"[ovn] AclAddHelper command: use [] instead None for string columns","message":"[ovn] AclAddHelper command: use [] instead None for string columns\n\nBe aware of None values before setting it to values in the Meter table.\nThis issure was introduced in https://review.opendev.org/c/openstack/ovsdbapp/+/767574\n\nChange-Id: I6e39d2059c1c8ba1960b49c7be5e6a4dc5516d69\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/14da3bdedfd43f2d9f4fd99a3b25aaa88d38bcff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/14da3bdedfd43f2d9f4fd99a3b25aaa88d38bcff"}]},"branch":"refs/heads/master"},"a728b90f0370fdd6b80042631f165074daea2c88":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2021-01-14 00:45:28.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/76/770676/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/76/770676/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/76/770676/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/76/770676/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/76/770676/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/76/770676/4"}}},"commit":{"parents":[{"commit":"dadf4e6540bfce358df29de81794efbab46d80e1","subject":"Merge \"Add northbound API support for meters\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/dadf4e6540bfce358df29de81794efbab46d80e1"}]}],"author":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2021-01-13 23:02:51.000000000","tz":-300},"committer":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2021-01-14 00:45:06.000000000","tz":-300},"subject":"[ovn nb] Acl add: Avoid setting columns that have no values","message":"[ovn nb] Acl add: Avoid setting columns that have no values\n\nWhile creating ACL rows, avoid setting the name, severity, or meter\ncolumns when no values were provided.\n\nThis issue was introduced in\nhttps://review.opendev.org/c/openstack/ovsdbapp/+/767574\n\nChange-Id: I6e39d2059c1c8ba1960b49c7be5e6a4dc5516d69\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/a728b90f0370fdd6b80042631f165074daea2c88"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/a728b90f0370fdd6b80042631f165074daea2c88"}]},"branch":"refs/heads/master"},"aac2da8f0cb5adc006613e49d8d4fcf08531ef37":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2021-01-14 15:10:53.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/76/770676/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/76/770676/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/76/770676/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/76/770676/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/76/770676/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/76/770676/5"}}},"commit":{"parents":[{"commit":"dadf4e6540bfce358df29de81794efbab46d80e1","subject":"Merge \"Add northbound API support for meters\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/dadf4e6540bfce358df29de81794efbab46d80e1"}]}],"author":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2021-01-13 23:02:51.000000000","tz":-300},"committer":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2021-01-14 15:09:10.000000000","tz":-300},"subject":"[ovn nb] Acl add: Avoid setting columns that have no values","message":"[ovn nb] Acl add: Avoid setting columns that have no values\n\nWhile creating ACL rows, avoid setting the name, severity, or meter\ncolumns when no values were provided.\n\nThis issue was introduced in\nhttps://review.opendev.org/c/openstack/ovsdbapp/+/767574\n\nCloses-Bug: #1911671\nChange-Id: I6e39d2059c1c8ba1960b49c7be5e6a4dc5516d69\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/aac2da8f0cb5adc006613e49d8d4fcf08531ef37"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/aac2da8f0cb5adc006613e49d8d4fcf08531ef37"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
