)]}'
{"id":"openstack%2Fovsdbapp~774741","triplet_id":"openstack%2Fovsdbapp~stable%2Ftrain~I076269d3d666caa5ebe3d70111136c78393c7c5f","project":"openstack/ovsdbapp","branch":"stable/train","topic":"sglog/ovn","hashtags":[],"change_id":"I076269d3d666caa5ebe3d70111136c78393c7c5f","subject":"[DNM] [WIP] [ovn nb] Add northbound API support for meters","status":"ABANDONED","created":"2021-02-09 22:43:12.000000000","updated":"2021-03-02 11:05:35.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"47997da26d53c1acb5f27697faccdd9f188264c3","_number":774741,"virtual_id_number":774741,"owner":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-02-10 21:15:09.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"all":[{"value":-1,"date":"2021-02-10 19:01:12.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"all":[{"value":-1,"date":"2021-02-10 19:01:12.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-09 22:48:13.000000000","updated_by":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"reviewer":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"state":"REVIEWER"},{"updated":"2021-02-10 01:18:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-10 11:09:00.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"CC"}],"messages":[{"id":"1e69ffe67d726b8c4a4c706f20c558e0b5f0c7d8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2021-02-09 22:43:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b5a5f1ec067ac34a28914ba49a925254952d5e7b","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2021-02-09 22:45:29.000000000","message":"Topic set to sglog/ovn","accounts_in_message":[],"_revision_number":1},{"id":"180815cc2d34c6e86da5c71c9ea1572c6e3b97e3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2021-02-09 23:08:49.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a9b420ca34c4bdbce87d026b28c294bc40fcba3a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2021-02-09 23:10:22.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"215bfbba9add6ae419611c094e4e6aa1d8df8876","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-10 01:18:05.000000000","message":"Patch Set 3: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2bc1691d3c88428faf7069d4e88e30a6 : FAILURE in 4m 43s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/028197981f454e019e347c3b80f70953 : SUCCESS in 4m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/70214fa1855b4ab4a2d1fc6a9fa7cfc1 : SUCCESS in 4m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7455e0f7632046d6bef49f68a15eb2b0 : SUCCESS in 3m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f86cc104aa50469a9fcf6188ce80d82b : FAILURE in 3m 13s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/0350f2ccf724492c8943efe8a813a3bf : FAILURE in 5m 56s\n- ovsdbapp-tempest-dsvm-networking-ovn-ovs-release https://zuul.opendev.org/t/openstack/build/e528044b33334f4aaf903b2f77977427 : SUCCESS in 2h 04m 23s","accounts_in_message":[],"_revision_number":3},{"id":"a4a4c39b208f7696bddefa12333c032dfc6a947f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2021-02-10 01:19:57.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"876875c536daee96e0eb02029233f388e323bb67","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2021-02-10 02:13:46.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"2692b752397c0dbd9b7a45c2f4eb2d223e0bed33","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2021-02-10 02:33:59.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"2359dbee9ff622d2cb41180c61f6fe97f385ef62","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-10 04:18:49.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/47e41878480f4601be82e2b34f88ecde : SUCCESS in 3m 54s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/93038e513150419eb67e7f34738511fd : SUCCESS in 3m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e76894eedd354005bd110bcdb812f236 : SUCCESS in 3m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6f239e97af5840b5b9b109fbe0d54cd2 : SUCCESS in 3m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d87ec822b4cf4aeb8e7ab554c3db1448 : FAILURE in 3m 19s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/54aa7a9c82ef4831bbca5109334412a1 : SUCCESS in 5m 41s\n- ovsdbapp-tempest-dsvm-networking-ovn-ovs-release https://zuul.opendev.org/t/openstack/build/c7a971e0f3894451853817b6def87432 : SUCCESS in 1h 43m 00s","accounts_in_message":[],"_revision_number":6},{"id":"307809c0a6eac557f42c2daa87ccb5a8b7db1553","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2021-02-10 10:51:21.000000000","message":"Patch Set 7: Patch Set 6 was rebased","accounts_in_message":[],"_revision_number":7},{"id":"7ed4946ae07d8fb261b3e771d9af649a3cafccbf","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-02-10 11:09:00.000000000","message":"Patch Set 7:\n\nPlease, read my comment in https://review.opendev.org/c/openstack/ovsdbapp/+/774742","accounts_in_message":[],"_revision_number":7},{"id":"20dd0678ee732147ecee331d5aa7e87612389190","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-10 13:00:19.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4913ee6a337147f9a3e2e73517acc9c0 : SUCCESS in 3m 59s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9a0101679dfd4a7797b5dcee184a64cd : SUCCESS in 4m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d30afa166f44487ead1dd61867caed39 : SUCCESS in 3m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/758bb7437b4045fa9ece705e31c7c3e4 : SUCCESS in 5m 40s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/b848fd791e674f12aa2c15eced43b702 : SUCCESS in 6m 56s\n- ovsdbapp-tempest-dsvm-networking-ovn-ovs-release https://zuul.opendev.org/t/openstack/build/bc679b0b91f3413e9083f2f76c1d7234 : SUCCESS in 2h 04m 16s","accounts_in_message":[],"_revision_number":7},{"id":"fdd17d2e7a8c5391056fa74579e9c5c5d8abca51","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2021-02-10 16:36:58.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"3a8ff02c8c58e791e33733e6dc4bb8fc485db789","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2021-02-10 18:59:23.000000000","message":"Uploaded patch set 9: Commit message was updated.","accounts_in_message":[],"_revision_number":9},{"id":"31d2fa9456560018588e5ae967edbf1730d9340a","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2021-02-10 19:01:12.000000000","message":"Patch Set 9: Code-Review-1 Workflow-1","accounts_in_message":[],"_revision_number":9},{"id":"6e5f576c7d7d2f599cb3faa99f5980779f993c15","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-10 21:15:09.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9c659a397bdb4ac8924ba23a338d198d : SUCCESS in 3m 44s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1d91bc71259f45feb850d979a55a1cf0 : SUCCESS in 4m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dd798d4fa1664fd88a280e5b7dfb19e9 : SUCCESS in 5m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1cb409344816438b90c27eac7b6d00ca : SUCCESS in 4m 03s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/d0a752fa427c4325a894ace9feee4748 : SUCCESS in 8m 15s\n- ovsdbapp-tempest-dsvm-networking-ovn-ovs-release https://zuul.opendev.org/t/openstack/build/61421bfb6d124c83918aaa2c3fbfb0e8 : SUCCESS in 2h 02m 28s","accounts_in_message":[],"_revision_number":9},{"id":"47997da26d53c1acb5f27697faccdd9f188264c3","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2021-03-02 11:05:35.000000000","message":"Abandoned\n\nThis is actually a new feature, so it will not be backported to stable branches.","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"9a48d3017c920dc3ab2166b1179795e4aa6f6066","revisions":{"63ed47fec13090fc6b5769ef5f93b4baba642dc5":{"kind":"REWORK","_number":1,"created":"2021-02-09 22:43:12.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/41/774741/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/41/774741/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/41/774741/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/41/774741/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/41/774741/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/41/774741/1"}}},"commit":{"parents":[{"commit":"11ca35873fdb4d6e07b8b01d132ba85c69d62c74","subject":"Merge \"Don\u0027t give up when an Exception happens in idl.run\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/11ca35873fdb4d6e07b8b01d132ba85c69d62c74"}]}],"author":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-12-17 15:43:41.000000000","tz":-300},"committer":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2021-02-09 22:42:35.000000000","tz":-300},"subject":"Add northbound API support for meters","message":"Add northbound API support for meters\n\nAdd meter (name) attribute to the ACL row, via sg_acl_add().\n\nAdd helpers for adding, removing, getting and listing rows of the\nMeter table.\n\nConflicts:\novsdbapp/tests/functional/schema/ovn_northbound/test_impl_idl.py\n\nChange-Id: I076269d3d666caa5ebe3d70111136c78393c7c5f\n(cherry picked from commit e2f782cd727506de3dbc2a32cc49920b0c792c0f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/63ed47fec13090fc6b5769ef5f93b4baba642dc5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/63ed47fec13090fc6b5769ef5f93b4baba642dc5"}]},"branch":"refs/heads/stable/train"},"507a3715bfd76f7829829cace7d5dc90e02de57c":{"kind":"REWORK","_number":2,"created":"2021-02-09 23:08:49.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/41/774741/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/41/774741/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/41/774741/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/41/774741/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/41/774741/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/41/774741/2"}}},"commit":{"parents":[{"commit":"11ca35873fdb4d6e07b8b01d132ba85c69d62c74","subject":"Merge \"Don\u0027t give up when an Exception happens in idl.run\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/11ca35873fdb4d6e07b8b01d132ba85c69d62c74"}]}],"author":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-12-17 15:43:41.000000000","tz":-300},"committer":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2021-02-09 23:08:35.000000000","tz":-300},"subject":"Add northbound API support for meters","message":"Add northbound API support for meters\n\nAdd meter (name) attribute to the ACL row, via sg_acl_add().\n\nAdd helpers for adding, removing, getting and listing rows of the\nMeter table.\n\nConflicts:\novsdbapp/tests/functional/schema/ovn_northbound/test_impl_idl.py\n\nChange-Id: I076269d3d666caa5ebe3d70111136c78393c7c5f\n(cherry picked from commit e2f782cd727506de3dbc2a32cc49920b0c792c0f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/507a3715bfd76f7829829cace7d5dc90e02de57c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/507a3715bfd76f7829829cace7d5dc90e02de57c"}]},"branch":"refs/heads/stable/train"},"3e3df6279127f82b19c9900c5def07feafc32a08":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2021-02-09 23:10:22.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/41/774741/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/41/774741/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/41/774741/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/41/774741/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/41/774741/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/41/774741/3"}}},"commit":{"parents":[{"commit":"11ca35873fdb4d6e07b8b01d132ba85c69d62c74","subject":"Merge \"Don\u0027t give up when an Exception happens in idl.run\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/11ca35873fdb4d6e07b8b01d132ba85c69d62c74"}]}],"author":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-12-17 15:43:41.000000000","tz":-300},"committer":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2021-02-09 23:09:41.000000000","tz":-300},"subject":"Add northbound API support for meters","message":"Add northbound API support for meters\n\nAdd meter (name) attribute to the ACL row, via sg_acl_add().\n\nAdd helpers for adding, removing, getting and listing rows of the\nMeter table.\n\nConflicts:\novsdbapp/tests/functional/schema/ovn_northbound/test_impl_idl.py\novsdbapp/schema/ovn_northbound/commands.py : \"raise from\" not compatible\n\nChange-Id: I076269d3d666caa5ebe3d70111136c78393c7c5f\n(cherry picked from commit e2f782cd727506de3dbc2a32cc49920b0c792c0f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/3e3df6279127f82b19c9900c5def07feafc32a08"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/3e3df6279127f82b19c9900c5def07feafc32a08"}]},"branch":"refs/heads/stable/train"},"81841e3f62774ef3357f396d02c1ecc972ae9d5a":{"kind":"REWORK","_number":4,"created":"2021-02-10 01:19:57.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/41/774741/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/41/774741/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/41/774741/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/41/774741/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/41/774741/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/41/774741/4"}}},"commit":{"parents":[{"commit":"11ca35873fdb4d6e07b8b01d132ba85c69d62c74","subject":"Merge \"Don\u0027t give up when an Exception happens in idl.run\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/11ca35873fdb4d6e07b8b01d132ba85c69d62c74"}]}],"author":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-12-17 15:43:41.000000000","tz":-300},"committer":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2021-02-10 01:13:12.000000000","tz":-300},"subject":"Add northbound API support for meters","message":"Add northbound API support for meters\n\nAdd meter (name) attribute to the ACL row, via sg_acl_add().\n\nAdd helpers for adding, removing, getting and listing rows of the\nMeter table.\n\nConflicts:\novsdbapp/tests/functional/schema/ovn_northbound/test_impl_idl.py\novsdbapp/schema/ovn_northbound/commands.py : \"raise from\" not compatible\n\nChange-Id: I076269d3d666caa5ebe3d70111136c78393c7c5f\n(cherry picked from commit e2f782cd727506de3dbc2a32cc49920b0c792c0f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/81841e3f62774ef3357f396d02c1ecc972ae9d5a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/81841e3f62774ef3357f396d02c1ecc972ae9d5a"}]},"branch":"refs/heads/stable/train"},"40091eb6bad93b22832f1c2dd1642cd4d7329ac2":{"kind":"REWORK","_number":5,"created":"2021-02-10 02:13:46.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/41/774741/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/41/774741/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/41/774741/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/41/774741/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/41/774741/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/41/774741/5"}}},"commit":{"parents":[{"commit":"11ca35873fdb4d6e07b8b01d132ba85c69d62c74","subject":"Merge \"Don\u0027t give up when an Exception happens in idl.run\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/11ca35873fdb4d6e07b8b01d132ba85c69d62c74"}]}],"author":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-12-17 15:43:41.000000000","tz":-300},"committer":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2021-02-10 02:13:23.000000000","tz":-300},"subject":"Add northbound API support for meters","message":"Add northbound API support for meters\n\nAdd meter (name) attribute to the ACL row, via sg_acl_add().\n\nAdd helpers for adding, removing, getting and listing rows of the\nMeter table.\n\nConflicts:\novsdbapp/tests/functional/schema/ovn_northbound/test_impl_idl.py\novsdbapp/schema/ovn_northbound/commands.py : \"raise from\" not compatible\novsdbapp/tests/functional/schema/ovn_northbound/fixtures.py : init api\n\nChange-Id: I076269d3d666caa5ebe3d70111136c78393c7c5f\n(cherry picked from commit e2f782cd727506de3dbc2a32cc49920b0c792c0f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/40091eb6bad93b22832f1c2dd1642cd4d7329ac2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/40091eb6bad93b22832f1c2dd1642cd4d7329ac2"}]},"branch":"refs/heads/stable/train"},"132a022ec44c8d2f85fbda932a9c569f669a3d19":{"kind":"REWORK","_number":6,"created":"2021-02-10 02:33:59.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/41/774741/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/41/774741/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/41/774741/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/41/774741/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/41/774741/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/41/774741/6"}}},"commit":{"parents":[{"commit":"11ca35873fdb4d6e07b8b01d132ba85c69d62c74","subject":"Merge \"Don\u0027t give up when an Exception happens in idl.run\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/11ca35873fdb4d6e07b8b01d132ba85c69d62c74"}]}],"author":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-12-17 15:43:41.000000000","tz":-300},"committer":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2021-02-10 02:33:52.000000000","tz":-300},"subject":"Add northbound API support for meters","message":"Add northbound API support for meters\n\nAdd meter (name) attribute to the ACL row, via sg_acl_add().\n\nAdd helpers for adding, removing, getting and listing rows of the\nMeter table.\n\nConflicts:\novsdbapp/tests/functional/schema/ovn_northbound/test_impl_idl.py\novsdbapp/schema/ovn_northbound/commands.py : \"raise from\" not compatible\novsdbapp/tests/functional/schema/ovn_northbound/fixtures.py : init api\n\nChange-Id: I076269d3d666caa5ebe3d70111136c78393c7c5f\n(cherry picked from commit e2f782cd727506de3dbc2a32cc49920b0c792c0f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/132a022ec44c8d2f85fbda932a9c569f669a3d19"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/132a022ec44c8d2f85fbda932a9c569f669a3d19"}]},"branch":"refs/heads/stable/train"},"a10c6ca7cf16dad91cb2fc2f62d8a6b227e55083":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2021-02-10 10:51:21.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/41/774741/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/41/774741/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/41/774741/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/41/774741/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/41/774741/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/41/774741/7"}}},"commit":{"parents":[{"commit":"3cb2a8e86b3c94df427e7cf21ad12b4a18e9517e","subject":"Dropping lower constraints testing (stable Train)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/3cb2a8e86b3c94df427e7cf21ad12b4a18e9517e"}]}],"author":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-12-17 15:43:41.000000000","tz":-300},"committer":{"name":"Flavio Fernandes","email":"flavio@flaviof.com","date":"2021-02-10 10:51:21.000000000","tz":0},"subject":"Add northbound API support for meters","message":"Add northbound API support for meters\n\nAdd meter (name) attribute to the ACL row, via sg_acl_add().\n\nAdd helpers for adding, removing, getting and listing rows of the\nMeter table.\n\nConflicts:\novsdbapp/tests/functional/schema/ovn_northbound/test_impl_idl.py\novsdbapp/schema/ovn_northbound/commands.py : \"raise from\" not compatible\novsdbapp/tests/functional/schema/ovn_northbound/fixtures.py : init api\n\nChange-Id: I076269d3d666caa5ebe3d70111136c78393c7c5f\n(cherry picked from commit e2f782cd727506de3dbc2a32cc49920b0c792c0f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/a10c6ca7cf16dad91cb2fc2f62d8a6b227e55083"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/a10c6ca7cf16dad91cb2fc2f62d8a6b227e55083"}]},"branch":"refs/heads/stable/train","description":"Rebase"},"d9a090fcb8eaf70b26f95e581a5a0f6e8e139939":{"kind":"REWORK","_number":8,"created":"2021-02-10 16:36:58.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/41/774741/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/41/774741/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/41/774741/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/41/774741/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/41/774741/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/41/774741/8"}}},"commit":{"parents":[{"commit":"3cb2a8e86b3c94df427e7cf21ad12b4a18e9517e","subject":"Dropping lower constraints testing (stable Train)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/3cb2a8e86b3c94df427e7cf21ad12b4a18e9517e"}]}],"author":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-12-17 15:43:41.000000000","tz":-300},"committer":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2021-02-10 16:36:47.000000000","tz":-300},"subject":"Add northbound API support for meters","message":"Add northbound API support for meters\n\nAdd meter (name) attribute to the ACL row, via sg_acl_add().\n\nAdd helpers for adding, removing, getting and listing rows of the\nMeter table.\n\nConflicts:\novsdbapp/tests/functional/schema/ovn_northbound/test_impl_idl.py\novsdbapp/schema/ovn_northbound/commands.py : \"raise from\" not compatible\novsdbapp/tests/functional/schema/ovn_northbound/fixtures.py : init api\n\nChange-Id: I076269d3d666caa5ebe3d70111136c78393c7c5f\n(cherry picked from commit e2f782cd727506de3dbc2a32cc49920b0c792c0f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/d9a090fcb8eaf70b26f95e581a5a0f6e8e139939"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/d9a090fcb8eaf70b26f95e581a5a0f6e8e139939"}]},"branch":"refs/heads/stable/train"},"9a48d3017c920dc3ab2166b1179795e4aa6f6066":{"kind":"NO_CODE_CHANGE","_number":9,"created":"2021-02-10 18:59:23.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/41/774741/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/41/774741/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/41/774741/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/41/774741/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/41/774741/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/41/774741/9"}}},"commit":{"parents":[{"commit":"3cb2a8e86b3c94df427e7cf21ad12b4a18e9517e","subject":"Dropping lower constraints testing (stable Train)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/3cb2a8e86b3c94df427e7cf21ad12b4a18e9517e"}]}],"author":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-12-17 15:43:41.000000000","tz":-300},"committer":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2021-02-10 18:58:41.000000000","tz":-300},"subject":"[DNM] [WIP] [ovn nb] Add northbound API support for meters","message":"[DNM] [WIP] [ovn nb] Add northbound API support for meters\n\nDo not merge: This is actually a new feature, so it will not be backported\nto stable branches. This gerrit is mostly a placeholder, to be soon abandoned.\n\nAdd meter (name) attribute to the ACL row, via sg_acl_add().\n\nAdd helpers for adding, removing, getting and listing rows of the\nMeter table.\n\nConflicts:\novsdbapp/tests/functional/schema/ovn_northbound/test_impl_idl.py\novsdbapp/schema/ovn_northbound/commands.py : \"raise from\" not compatible\novsdbapp/tests/functional/schema/ovn_northbound/fixtures.py : init api\n\nChange-Id: I076269d3d666caa5ebe3d70111136c78393c7c5f\n(cherry picked from commit e2f782cd727506de3dbc2a32cc49920b0c792c0f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/9a48d3017c920dc3ab2166b1179795e4aa6f6066"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/9a48d3017c920dc3ab2166b1179795e4aa6f6066"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
