)]}'
{"id":"openstack%2Fovsdbapp~775370","triplet_id":"openstack%2Fovsdbapp~stable%2Ftrain~Ib83859fb038c441064bcf3d5d96c31cface7d5c9","project":"openstack/ovsdbapp","branch":"stable/train","hashtags":[],"change_id":"Ib83859fb038c441064bcf3d5d96c31cface7d5c9","subject":"More robustly handle venv ovs/ovn paths","status":"MERGED","created":"2021-02-12 12:22:11.000000000","updated":"2021-02-23 17:19:29.000000000","submitted":"2021-02-23 17:19:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"775370","meta_rev_id":"1e89c446502e0d2a511cc9f53e88b4c144864e6c","_number":775370,"virtual_id_number":775370,"owner":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-23 17:19:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"all":[{"value":2,"date":"2021-02-15 13:21:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-02-23 15:08:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"all":[{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-02-23 15:08:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"}],"reviewers":{"REVIEWER":[{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-12 12:22:11.000000000","updated_by":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"reviewer":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"state":"REVIEWER"},{"updated":"2021-02-12 12:36:34.000000000","updated_by":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"reviewer":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"state":"REVIEWER"},{"updated":"2021-02-12 12:36:34.000000000","updated_by":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"reviewer":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"state":"REVIEWER"},{"updated":"2021-02-12 14:43:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ad0326d6e399aa12d6a97f9951b140f61c5a2872","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2021-02-12 12:22:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d8f943b0254c3e2d17e797b96778d8f6397239e4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2021-02-12 12:28:05.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"22686d0922b81491c10bb2dc80334731f6ab9ef5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2021-02-12 12:31:33.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"34a8b306040d251c4ce9ac315e2bac15e63058c4","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2021-02-12 12:36:34.000000000","message":"Patch Set 3:\n\nAfter hitting this issue on stable/train again [0], I figure it would be\nbetter for me to have this here instead of completely forgetting about it, again. 😊 \n\n[0]: https://gist.github.com/flavio-fernandes/1bea8030c04edb0723f7630a23533a6d","accounts_in_message":[],"_revision_number":3},{"id":"209284551bfd735a90c898f684f841659a201270","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-12 14:43:50.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/77bf9e9cf1154119af9d88d116bee7af : SUCCESS in 4m 29s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ce9ec35da29a40d7b60aed7ce3467efd : SUCCESS in 4m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/70652864e14547409e45c8e01f4ddee3 : SUCCESS in 3m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cc392daece144579b8d9772f4b971c38 : SUCCESS in 3m 55s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/ac0f26f2745147ce973323ac8543ece4 : SUCCESS in 5m 01s\n- ovsdbapp-tempest-dsvm-networking-ovn-ovs-release https://zuul.opendev.org/t/openstack/build/e9f9cf0d00bb470a8397dc256105e4c8 : SUCCESS in 2h 10m 21s","accounts_in_message":[],"_revision_number":3},{"id":"f67824af0dc1bf09c7f5bd541c0f05148b76c7d0","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2021-02-15 13:21:05.000000000","message":"Patch Set 3: Code-Review+2\n\nThanks, flavio!","accounts_in_message":[],"_revision_number":3},{"id":"140930a598d52fadb0f10d2ca3e47fbe95e856fc","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2021-02-23 15:08:37.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"226d2cca0c5d737de2aa48934729ad2a5ca0fc00","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-23 15:09:15.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"0cf790c393b95183e90c023992654c3478225d45","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-23 17:19:28.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ea1d1708d1dc4e7981d15d2327bf5016 : SUCCESS in 4m 48s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b3cc96debbed4958b91771938cdf70ad : SUCCESS in 5m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5f71c0fd2ae84f42b888367ffea5fb5f : SUCCESS in 5m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0f725c1a0c6d497084c288364a51c451 : SUCCESS in 6m 09s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/270c6774a2e545698bb925088f73795c : SUCCESS in 7m 43s\n- ovsdbapp-tempest-dsvm-networking-ovn-ovs-release https://zuul.opendev.org/t/openstack/build/a4f5c8921ed8447db8dc25d71c63af7c : SUCCESS in 2h 02m 41s","accounts_in_message":[],"_revision_number":3},{"id":"1e89c446502e0d2a511cc9f53e88b4c144864e6c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-23 17:19:29.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"2fa0fb17960ca64eb132ddbe23fdd56191f874a5","revisions":{"afee269fa5df9d82c4d943a532836ad7be0ff9e8":{"kind":"REWORK","_number":1,"created":"2021-02-12 12:22:11.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/70/775370/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/70/775370/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/70/775370/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/70/775370/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/70/775370/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/70/775370/1"}}},"commit":{"parents":[{"commit":"11ca35873fdb4d6e07b8b01d132ba85c69d62c74","subject":"Merge \"Don\u0027t give up when an Exception happens in idl.run\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/11ca35873fdb4d6e07b8b01d132ba85c69d62c74"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-01-22 21:12:17.000000000","tz":-360},"committer":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2021-02-12 12:21:56.000000000","tz":-300},"subject":"More robustly handle venv ovs/ovn paths","message":"More robustly handle venv ovs/ovn paths\n\nThis should handle both src directories and system-installed ovs\nand ovn, whether by package in /usr or by source in /usr/local.\n\nThis patchset also cherry picks from\ncommit 5d70d45f2d014cf265b68844122eadd541f5c3cf which included\npre-requisite changes for OvsOvnVenvFixture that were improved.\nWith that, this changes tools/setup-ovs.sh to take an OVS_BRANCH\nenv variable to use a specific branch.\n\nCloses-Bug: 1856676\n\nConflicts:\n  ovsdbapp/constants.py\n  ovsdbapp/venv.py\n\nChange-Id: Ib83859fb038c441064bcf3d5d96c31cface7d5c9\n(cherry picked from commit 5d70d45f2d014cf265b68844122eadd541f5c3cf)\n(cherry picked from commit 781ecb5bf648774f5d365585a0c2c13442523754)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/afee269fa5df9d82c4d943a532836ad7be0ff9e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/afee269fa5df9d82c4d943a532836ad7be0ff9e8"}]},"branch":"refs/heads/stable/train"},"3e7c6770214db49473db3a2c46fdf71f11d48c4e":{"kind":"REWORK","_number":2,"created":"2021-02-12 12:28:05.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/70/775370/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/70/775370/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/70/775370/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/70/775370/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/70/775370/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/70/775370/2"}}},"commit":{"parents":[{"commit":"11ca35873fdb4d6e07b8b01d132ba85c69d62c74","subject":"Merge \"Don\u0027t give up when an Exception happens in idl.run\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/11ca35873fdb4d6e07b8b01d132ba85c69d62c74"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-01-22 21:12:17.000000000","tz":-360},"committer":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2021-02-12 12:27:56.000000000","tz":-300},"subject":"More robustly handle venv ovs/ovn paths","message":"More robustly handle venv ovs/ovn paths\n\nThis should handle both src directories and system-installed ovs\nand ovn, whether by package in /usr or by source in /usr/local.\n\nThis patchset also cherry picks from\ncommit 5d70d45f2d014cf265b68844122eadd541f5c3cf which included\npre-requisite changes for OvsOvnVenvFixture that were improved.\nWith that, this changes tools/setup-ovs.sh to take an OVS_BRANCH\nenv variable to use a specific branch.\n\nCloses-Bug: 1856676\n\nConflicts:\n  ovsdbapp/constants.py\n  ovsdbapp/venv.py\n\nChange-Id: Ib83859fb038c441064bcf3d5d96c31cface7d5c9\n(cherry picked from commit 5d70d45f2d014cf265b68844122eadd541f5c3cf)\n(cherry picked from commit 781ecb5bf648774f5d365585a0c2c13442523754)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/3e7c6770214db49473db3a2c46fdf71f11d48c4e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/3e7c6770214db49473db3a2c46fdf71f11d48c4e"}]},"branch":"refs/heads/stable/train"},"2fa0fb17960ca64eb132ddbe23fdd56191f874a5":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-02-12 12:31:33.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/70/775370/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/70/775370/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/70/775370/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/70/775370/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/70/775370/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/70/775370/3"}}},"commit":{"parents":[{"commit":"3cb2a8e86b3c94df427e7cf21ad12b4a18e9517e","subject":"Dropping lower constraints testing (stable Train)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/3cb2a8e86b3c94df427e7cf21ad12b4a18e9517e"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-01-22 21:12:17.000000000","tz":-360},"committer":{"name":"Flavio Fernandes","email":"flavio@flaviof.com","date":"2021-02-12 12:31:33.000000000","tz":0},"subject":"More robustly handle venv ovs/ovn paths","message":"More robustly handle venv ovs/ovn paths\n\nThis should handle both src directories and system-installed ovs\nand ovn, whether by package in /usr or by source in /usr/local.\n\nThis patchset also cherry picks from\ncommit 5d70d45f2d014cf265b68844122eadd541f5c3cf which included\npre-requisite changes for OvsOvnVenvFixture that were improved.\nWith that, this changes tools/setup-ovs.sh to take an OVS_BRANCH\nenv variable to use a specific branch.\n\nCloses-Bug: 1856676\n\nConflicts:\n  ovsdbapp/constants.py\n  ovsdbapp/venv.py\n\nChange-Id: Ib83859fb038c441064bcf3d5d96c31cface7d5c9\n(cherry picked from commit 5d70d45f2d014cf265b68844122eadd541f5c3cf)\n(cherry picked from commit 781ecb5bf648774f5d365585a0c2c13442523754)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/2fa0fb17960ca64eb132ddbe23fdd56191f874a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/2fa0fb17960ca64eb132ddbe23fdd56191f874a5"}]},"branch":"refs/heads/stable/train","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
