)]}'
{"id":"openstack%2Fovsdbapp~776198","triplet_id":"openstack%2Fovsdbapp~master~Ie90c6e26f04c4daaf2792dad81fabb7343ff8f75","project":"openstack/ovsdbapp","branch":"master","topic":"log-command-class-name","hashtags":[],"change_id":"Ie90c6e26f04c4daaf2792dad81fabb7343ff8f75","subject":"Log the name of the command in execute()","status":"MERGED","created":"2021-02-17 13:48:43.000000000","updated":"2021-02-23 11:44:32.000000000","submitted":"2021-02-23 11:42:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"776198-log-command-class-name","meta_rev_id":"dfc10be4da6fa11df0d869a7ab505396be570262","_number":776198,"virtual_id_number":776198,"owner":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-23 11:42:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":2,"date":"2021-02-23 10:07:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2021-02-18 13:57:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":1,"date":"2021-02-23 10:07:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-17 15:22:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-18 13:57:04.000000000","updated_by":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"reviewer":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"state":"REVIEWER"},{"updated":"2021-02-23 10:07:34.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"7c1eadc54e9eaf951327de0d0258d91d739c1e4d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2021-02-17 13:48:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fbb9c25816d3b1a4f6bc782b5b24c59384f2e797","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-17 15:22:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/aae51a95c6f041ab9733da19bede4778 : SUCCESS in 5m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/16cf4bb4ca96436ea2b61153f11d0949 : SUCCESS in 4m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7bd987e3dc3d40c88135667f2a7cd16b : SUCCESS in 4m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/06dee1c933134e228a9b6646e1afc14b : SUCCESS in 4m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/35c248f29fde41d780a0e2b8f366b2ed : SUCCESS in 4m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf375131683644ada9c10f26a617a9fe : SUCCESS in 6m 18s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/59f79476feec4f039463094a3e6340ae : SUCCESS in 6m 05s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/880b601ed9424db981798d214289c249 : SUCCESS in 1h 18m 16s","accounts_in_message":[],"_revision_number":1},{"id":"570551419a3787a8f9895b8ee05b24933e6bdaaa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2021-02-18 09:21:20.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"29e512e6f70e927d8894025f1b2de79ecdb99efb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2021-02-18 09:22:10.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"9385d5263ce2c18e3ab32ba34f6620cf4db3edbc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-18 10:29:45.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5034a115d1a34418a0dff1788fa92c5d : SUCCESS in 4m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cc3de37533ce41db95b5d73924728c9b : SUCCESS in 3m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7970650e9dc046268ab568d165e6fb3e : SUCCESS in 4m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/40ab6c1e714c4434a02d9f5371d46d73 : SUCCESS in 4m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/da998437d05f4d0d811904ff7acddba8 : SUCCESS in 6m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ed86dc2b8af452dab95ec2f3c669cf3 : SUCCESS in 6m 40s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/c5870cd399824c2aac1254eec9907a49 : SUCCESS in 7m 27s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/d577fb1019c64fb08984215e04a5ca8c : SUCCESS in 1h 04m 21s","accounts_in_message":[],"_revision_number":3},{"id":"1f3faa3ecd789ed3c7b172f264d78e0c30479e20","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2021-02-18 13:57:04.000000000","message":"Patch Set 3: Code-Review+2\n\nSeems like a good idea. I guess not everybody will be running in DEBUG mode, and finding the Command  through the line number in run_idl() has always been annoying. Thanks!","accounts_in_message":[],"_revision_number":3},{"id":"45081f76545bc4bb291e97c378bd2ce573e4a721","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-02-23 10:07:34.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"b7f52c850cf16c3cd7f90e43c7a00afdb053f5ca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-23 10:07:48.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"eac5990dd92baf73a603cb2ad350516f68670d53","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-23 11:42:37.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6a46589225884ff5bb0d07421de63b40 : SUCCESS in 4m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d4529788e8c44fd94d259a97e79f0e6 : SUCCESS in 4m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/86eaee5941d741cab652b0600ad1ffd1 : SUCCESS in 4m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/52b96f8fa6694a1a8e394676d61162ae : SUCCESS in 4m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5fa3da06d31b493fa03a28810e5ec9ba : SUCCESS in 5m 31s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/e41919fa54c24c27a460d9f13e2db528 : SUCCESS in 6m 06s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/62bdab7b15ce4e85a13367565fc7467c : SUCCESS in 1h 32m 15s","accounts_in_message":[],"_revision_number":3},{"id":"aa3652e649645c7fa8b4df4eed8031841e2e5a59","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-23 11:42:40.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"dfc10be4da6fa11df0d869a7ab505396be570262","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-23 11:44:32.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6969a158f0744857b0251d95210723ba : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"13b4dd15abbd27a755147b187393eaec5093beb8","revisions":{"d168cff1afc20aa992f51a6dd9fe9c91a996b04f":{"kind":"REWORK","_number":1,"created":"2021-02-17 13:48:43.000000000","uploader":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"ref":"refs/changes/98/776198/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/98/776198/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/98/776198/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/98/776198/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/98/776198/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/98/776198/1"}}},"commit":{"parents":[{"commit":"aac2da8f0cb5adc006613e49d8d4fcf08531ef37","subject":"[ovn nb] Acl add: Avoid setting columns that have no values","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/aac2da8f0cb5adc006613e49d8d4fcf08531ef37"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2021-02-17 13:48:14.000000000","tz":0},"committer":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2021-02-17 13:48:14.000000000","tz":0},"subject":"DO NOT REVIEW: Log command class name","message":"DO NOT REVIEW: Log command class name\n\nChange-Id: Ie90c6e26f04c4daaf2792dad81fabb7343ff8f75\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/d168cff1afc20aa992f51a6dd9fe9c91a996b04f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/d168cff1afc20aa992f51a6dd9fe9c91a996b04f"}]},"branch":"refs/heads/master"},"5cc96b2b77568c59181829e5a0d7fa103b464f39":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-02-18 09:21:20.000000000","uploader":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"ref":"refs/changes/98/776198/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/98/776198/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/98/776198/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/98/776198/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/98/776198/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/98/776198/2"}}},"commit":{"parents":[{"commit":"aac2da8f0cb5adc006613e49d8d4fcf08531ef37","subject":"[ovn nb] Acl add: Avoid setting columns that have no values","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/aac2da8f0cb5adc006613e49d8d4fcf08531ef37"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2021-02-17 13:48:14.000000000","tz":0},"committer":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2021-02-18 09:20:57.000000000","tz":0},"subject":"Log the name of the command in run_idl()","message":"Log the name of the command in run_idl()\n\nTo help with troubleshooting, let\u0027s log the name of the command\nbeing executed as part of the error log.\n\nChange-Id: Ie90c6e26f04c4daaf2792dad81fabb7343ff8f75\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/5cc96b2b77568c59181829e5a0d7fa103b464f39"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/5cc96b2b77568c59181829e5a0d7fa103b464f39"}]},"branch":"refs/heads/master","description":"Edit commit message"},"13b4dd15abbd27a755147b187393eaec5093beb8":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2021-02-18 09:22:10.000000000","uploader":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"ref":"refs/changes/98/776198/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/98/776198/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/98/776198/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/98/776198/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/98/776198/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/98/776198/3"}}},"commit":{"parents":[{"commit":"aac2da8f0cb5adc006613e49d8d4fcf08531ef37","subject":"[ovn nb] Acl add: Avoid setting columns that have no values","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/aac2da8f0cb5adc006613e49d8d4fcf08531ef37"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2021-02-17 13:48:14.000000000","tz":0},"committer":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2021-02-18 09:22:03.000000000","tz":0},"subject":"Log the name of the command in execute()","message":"Log the name of the command in execute()\n\nTo help with troubleshooting, let\u0027s log the name of the command\nbeing executed as part of the error log.\n\nChange-Id: Ie90c6e26f04c4daaf2792dad81fabb7343ff8f75\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/13b4dd15abbd27a755147b187393eaec5093beb8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/13b4dd15abbd27a755147b187393eaec5093beb8"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
