)]}'
{"id":"openstack%2Fovsdbapp~785433","triplet_id":"openstack%2Fovsdbapp~master~I580bfb640d8157b4e749d7b1fe5de7568e99943d","project":"openstack/ovsdbapp","branch":"master","topic":"bug/1922934","hashtags":[],"change_id":"I580bfb640d8157b4e749d7b1fe5de7568e99943d","subject":"Add an active wait in the \"Backend.lookup\"","status":"MERGED","created":"2021-04-08 15:10:29.000000000","updated":"2021-04-28 12:47:36.000000000","submitted":"2021-04-28 12:45:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":37,"unresolved_comment_count":15,"has_review_started":true,"submission_id":"785433-bug/1922934","meta_rev_id":"46fbc8b12e01691ef651b5e83b04bf7ac7b99adf","_number":785433,"virtual_id_number":785433,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-28 12:45:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"all":[{"value":0,"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"value":2,"date":"2021-04-28 08:56:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"all":[{"value":0,"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"value":1,"date":"2021-04-28 08:56:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-08 15:11:02.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"state":"REVIEWER"},{"updated":"2021-04-08 16:18:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-11 15:37:43.000000000","updated_by":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"reviewer":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"state":"CC"},{"updated":"2021-04-16 20:08:14.000000000","updated_by":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"reviewer":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"state":"REVIEWER"},{"updated":"2021-04-22 14:14:11.000000000","updated_by":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"reviewer":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"state":"REVIEWER"},{"updated":"2021-04-27 12:00:20.000000000","updated_by":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"reviewer":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"state":"REVIEWER"}],"messages":[{"id":"7a55c1f33209d27b2040b45dd9d6e0ee39272b97","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-04-08 15:10:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"914165cef79e63423f4754a543584ea555589432","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-08 16:18:57.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7895b6eada454bb281fe555394b32a5d : SUCCESS in 4m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1bfef53c13314971835761be1f0aea6b : FAILURE in 3m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ee752da640284801abe8bdfa9b918cf9 : SUCCESS in 4m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1c79e611712142b9af2554d325d42c1d : SUCCESS in 4m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9d5b3edf14f64770a48ea67ec33b5b54 : SUCCESS in 4m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a89b3ea1d4a420a8c830539b8f0d12f : SUCCESS in 5m 51s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/30f802a89db3439e9f6939a1f103021d : FAILURE in 5m 57s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/b2659bff7c244235b478663fa76e3ebd : SUCCESS in 1h 05m 29s","accounts_in_message":[],"_revision_number":1},{"id":"f91fc7a18fdd710b9e6350f5a5bbabccb6c8eba3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-04-09 06:56:01.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5b499f06c274f3bb27bee947adaf9e1407c0ca63","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-09 08:17:53.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/96d6bd044f6748fb8972ff3362d2814e : FAILURE in 3m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/28e9da36ddb148d1a436f59c00461625 : SUCCESS in 4m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b0004e43b2f645eea047dd1b776f4fca : SUCCESS in 3m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d2de45dd98324cd39510c2e451a8a9b1 : SUCCESS in 4m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dc28d1a8f03641e495d3e40cea7b34bb : SUCCESS in 4m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ef80e843bc614c95bf8f60e758c1a13a : SUCCESS in 4m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb2515fe1c564268aca87d205598c37f : SUCCESS in 6m 30s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/8dd7c5410c724d208c7067fc96b19765 : SUCCESS in 6m 02s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/e8d4bba448144b269a98d5378ad8dd31 : SUCCESS in 1h 15m 52s","accounts_in_message":[],"_revision_number":2},{"id":"b14f074fdb73655ea7e016d111aed9b1f60d5814","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2021-04-11 15:38:16.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c63f6330326c0b0d1cdd4db3f38c3d6526ef94fe","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2021-04-13 13:32:49.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"dbad5f0c3f8deb3c89acd75497c97ff7c17da22d","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-04-15 11:09:51.000000000","message":"Patch Set 2:\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"7316b5d9325b5c130a900c371a296a2fd120a2f3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-04-15 11:13:22.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"5be835363ff63e7182d4a8716453d9f5fbc70f40","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-15 12:28:24.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/2a6c3d9f172e47af9ec0e4ce571f1f01 : SUCCESS in 3m 03s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d6cc70ddfb804bb0a2dbd23d7d55cfbe : SUCCESS in 4m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eb399ff313d84b0aac29585c6e45a3d5 : SUCCESS in 3m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ad787642b518446d927401f507c920dd : SUCCESS in 4m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d8aac79080774a7b9e36dda8290b0877 : SUCCESS in 4m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2e01c95f92e042c087ef7972258a52b9 : SUCCESS in 4m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3bfbd256d3894682bfa4ffb67f4cc1bb : SUCCESS in 5m 18s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/9777bd4def6c447ba795f422d1180fcb : FAILURE in 5m 59s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/aece507e935e4948b09bfe968e69a0a1 : SUCCESS in 1h 09m 26s","accounts_in_message":[],"_revision_number":3},{"id":"167a1e05c4fba59922147233f2da0b98de410eaa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-04-15 13:37:15.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"8b7b9feb9ace000871b772c2fe2bb21a55b1b985","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2021-04-15 14:04:48.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"ea51fe88d368cd815bef2af4fb2751fe8899afa9","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2021-04-15 14:12:31.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"e9e379c103c28e217a474d8b24073b5e3a43df33","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-15 14:48:32.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/08c2905f4eef4e6b89da1ef42fbc5dc6 : SUCCESS in 4m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dfc9844e899f4fd1a04593cbeea50792 : SUCCESS in 4m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2458157c857f4362ac2eb857ae2d7beb : SUCCESS in 5m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dd2068bb7ef34fa7a6c6c924cb6d1258 : SUCCESS in 4m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/541973c8c0ec45a8bef9b4dd8b309fd5 : SUCCESS in 6m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c544c4b349b4e1fa20b32ba68333552 : SUCCESS in 5m 38s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/21e5cdf36f324f1494566046d1a167f7 : FAILURE in 10m 09s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/1d0d10cc4d57464daa78e6b331e1a1ed : SUCCESS in 1h 04m 51s","accounts_in_message":[],"_revision_number":4},{"id":"8a4cde8f8b4fc1a51d2d9b370131900033fc466e","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2021-04-15 19:49:33.000000000","message":"Patch Set 4: Code-Review-1\n\n(3 comments)\n\nGood work, latest comments in-line.","accounts_in_message":[],"_revision_number":4},{"id":"c7eb75e5273b63e3df2fb888fc94ca41eeb5aecf","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-04-16 08:25:10.000000000","message":"Patch Set 4:\n\n(3 comments)","accounts_in_message":[],"_revision_number":4},{"id":"f6f12200babe729f2329cb8e6ae2bb9d3fccfa04","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-04-16 09:16:42.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"f3a1664304b25e1dc5c679739c2679bea3807859","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-04-16 09:19:29.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"75cddbacedc34531654d1bc28829b8d3868d8a65","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-16 10:19:34.000000000","message":"Patch Set 5: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1ff13b3a0f8b404db2b68b725c0daf42 : SUCCESS in 3m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/540cc5b33b954d588360275ff8733d6f : FAILURE in 3m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a9a4e5bfd347421d95b3eb8a87cb90b9 : SUCCESS in 3m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4d2968019abf4fd194e0fe306742928e : SUCCESS in 3m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e6373e5c1f7b46169a7b9f45e2601a93 : SUCCESS in 4m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60889041a633489cad440e9755ba5681 : SUCCESS in 5m 01s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/1dbc744ce2af400d9dbd09630f14b200 : SUCCESS in 6m 16s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/1f1c92532d164c7d9129e4260ab28f68 : SUCCESS in 1h 01m 42s","accounts_in_message":[],"_revision_number":5},{"id":"6d75b0797e2e02b709c92748e8662b81fb002f97","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-04-16 10:37:56.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"2e94ab46631d293bd8b37f4f633af04bba045ccb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-16 11:47:28.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d2d526a4c53d4e8f8f804589d51f9071 : SUCCESS in 3m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bcf689f4183248dcb2af657164d0ab4b : SUCCESS in 3m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0004e628dbdf4e639d1600026aba6276 : SUCCESS in 3m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bbe1fba98e234476a3a43e4dfb9686bf : SUCCESS in 3m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/13f8d01b6bfe4b44b965437dd469c4e9 : SUCCESS in 4m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/719e6e0f58284e50951e61ff2238805f : SUCCESS in 6m 07s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/2757b3f2b86e4109b79a2f2f3cf67c2a : SUCCESS in 6m 07s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/a93fb3ee3ae44b509f63cb6a2617ad24 : SUCCESS in 1h 03m 22s","accounts_in_message":[],"_revision_number":6},{"id":"9022ca108c45daf71567bbb4bac33bafc6766728","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2021-04-16 20:08:14.000000000","message":"Patch Set 6: Code-Review-1\n\n(2 comments)\n\nsmall nits, but I agree with Terry\u0027s comments on saving from the extra lookup.","accounts_in_message":[],"_revision_number":6},{"id":"6d0d41b88b3dbad3f2a4fae313d120ea9ebfd456","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2021-04-16 20:42:56.000000000","message":"Patch Set 6:\n\n(1 comment)\n\nAnother thought I had: this just works on the granularity of a lookup() call. I\u0027ve seen issues where we have the multi-server sync issue in a call to ls_get() which calls lookup(). Are we going to run into issues where on one call a function we\u0027d want a timeout, and not others? (I honestly don\u0027t know)\n\nAfter talking to dceara, I had an idea for removing some lookups with LS/LSP that may help as well as this. I\u0027ll try to have something posted by end of Monday.","accounts_in_message":[],"_revision_number":6},{"id":"1f0cf02d542b2859430055ab1f180214b3c2a5af","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-04-19 08:22:55.000000000","message":"Patch Set 6:\n\n(4 comments)","accounts_in_message":[],"_revision_number":6},{"id":"3999503ace5c195c7ac50ed55320c7b309a34044","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-04-19 10:12:31.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"e0309c6371538efccdaaedcddd1aec374095b624","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-19 11:40:52.000000000","message":"Patch Set 7: Verified-1\n\n(4 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7bdc45407b39477c9083db31aa93cbb7 : SUCCESS in 5m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6abc8e20e29442b19a93ae7836722646 : FAILURE in 6m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7bf524bc05154804819c5fea3e9e29af : SUCCESS in 5m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4d7b258f2a6d4e84b5c5356f32f9d914 : SUCCESS in 5m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/558e775e4b3a457a97df02362d62a2d1 : SUCCESS in 4m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2063ea5c03434316a70547c483eefc6d : SUCCESS in 7m 06s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/1a1f80de306145d389fddbf64452c175 : FAILURE in 8m 37s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/13b1fa685ebd4a11a392117b66d77c5d : SUCCESS in 1h 17m 04s","accounts_in_message":[],"_revision_number":7},{"id":"e0f7ceab875b3497f3f17b251ddfa6e098c106bb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-04-19 11:51:37.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"1a656288b58d40458d909c189aa9ad1cd61d1374","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-04-19 11:54:22.000000000","message":"Patch Set 8:\n\n(1 comment)\n\nThis is not working...","accounts_in_message":[],"_revision_number":8},{"id":"8b109eb734adc8d45fde1873d44a22434b054c28","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-19 13:16:11.000000000","message":"Patch Set 8: Verified-1\n\n(6 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/80946210601343abb70462cd2f7cabea : SUCCESS in 4m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/236b9eaa50f8485782979e1e6ca59041 : FAILURE in 4m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d2772546d9b44b18b89735a49dbc1906 : SUCCESS in 3m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/489fe7f4d6c24a5896d591d42bf3b986 : SUCCESS in 4m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3e9994a3d77c45e78386e53116d5384e : SUCCESS in 4m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/40065eb9842e48bfac8919a85ecf4251 : SUCCESS in 5m 19s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/c77e7547832348e3ab2bd11c670bfdcf : FAILURE in 6m 04s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/ba060be7f4704161b629d6c086194d4d : SUCCESS in 1h 19m 46s","accounts_in_message":[],"_revision_number":8},{"id":"b48ed3ab9ce0e91ab2359dd9d6e0c3221fb2be33","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-04-19 17:14:14.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"598c3337ae01c086623a03cc7fc37a6b5ff778c5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-04-19 17:15:27.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"9a8ae8cb30a127184140c1ac68c9c69e168cf070","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-19 18:27:51.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ca823612d5d4418d814efa6acf63a3cc : SUCCESS in 4m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c76db8887f5b4ec3aca11c47276debc7 : SUCCESS in 5m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9cbc043d0dd3497f91571796e827a94c : SUCCESS in 4m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/671d4d2892ce4c329d594433234acfde : SUCCESS in 4m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e9fc760e133a4d1fa0f801fb5b076dbc : SUCCESS in 4m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa8dae87db1a4ecc9dbc541b6fd5833c : SUCCESS in 7m 07s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/13758e10123a43a9bcf66aaac1d4311d : SUCCESS in 7m 55s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/369f76ad0f3840b7bc9d1ff22b9968eb : SUCCESS in 1h 04m 30s","accounts_in_message":[],"_revision_number":10},{"id":"89e341b4e1558e1bf2c0aa60170b41d8f7867bbc","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2021-04-20 17:00:14.000000000","message":"Patch Set 10:\n\n(3 comments)\n\nNice work! Just a couple more questions and I think we\u0027re good to go.","accounts_in_message":[],"_revision_number":10},{"id":"80f2c428fdd8027a53419c5b7ae7b6082bc81d99","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2021-04-20 20:18:16.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"192301a3802c8b027040e788e0e65966575b2500","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-04-21 15:18:06.000000000","message":"Patch Set 10:\n\n(3 comments)","accounts_in_message":[],"_revision_number":10},{"id":"fb4fa024be8afae04c6e021eb35dcaaf88ee5fdd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-04-21 15:19:14.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"6b5cf7be749fccc8f96963e9ee1d4ac7d4e412f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-21 16:21:27.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6bb5fbba4b514989957817f9f7b2a66a : SUCCESS in 4m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/82dacb47e2134213bb3f96193628caf1 : SUCCESS in 4m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0e5d94a3f7134dcbbe2d6b3ddb43a14d : SUCCESS in 5m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/546a101ae2e44214815c8a012865699b : SUCCESS in 4m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ef1612fb24cb47ff87e8066963b3189f : SUCCESS in 4m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/27f845bc93254a50972eb065cb6763e1 : SUCCESS in 6m 41s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/528c2d7ebf9b4264b50db9faaf1ed2d7 : SUCCESS in 6m 56s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/7816d62cf3d04d519e6ae7288c9c49a2 : SUCCESS in 59m 40s","accounts_in_message":[],"_revision_number":11},{"id":"37892f50ddd6687ed977e9877dd7bf653d223862","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2021-04-21 18:14:14.000000000","message":"Patch Set 11: Code-Review+2\n\n(1 comment)\n\nI think this is a great addition. Thanks, Rodolfo!","accounts_in_message":[],"_revision_number":11},{"id":"508c3c1afc6353d8a3fc57b9a53dbe58dabd6ed6","author":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"date":"2021-04-22 14:14:11.000000000","message":"Patch Set 11: Code-Review+2","accounts_in_message":[],"_revision_number":11},{"id":"e87de4c521e368d0a1eb1a0207ed0685cdb209a9","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-04-22 15:22:20.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"3c105f1f0c6e23baa03278e9bb980aadc129660b","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2021-04-27 12:00:20.000000000","message":"Patch Set 11: Code-Review+2\n\n(2 comments)\n\nQuestion inline wrt logging","accounts_in_message":[],"_revision_number":11},{"id":"af1c91b393035e9ba36880f8ffbd104be1ee2b15","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-04-27 16:42:33.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"7b3a5ff436bf8d023a3ed1789f050859d5217675","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-04-27 16:50:21.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"8ef6a98e34ab1cc392f7335839cf4d31b8a94ce0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-27 18:09:52.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4babd96c410d47cb8b4944912f97bc27 : SUCCESS in 5m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e78c76c0f104f6597e668f7f98e2b19 : SUCCESS in 5m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dd8ac23eaff940d6b990a16cbce1a88b : SUCCESS in 5m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ffd02fc7ebee4c0197fa5eef3b63d2ed : SUCCESS in 5m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6efaf89d013f417c9be5bb41f8b606de : SUCCESS in 5m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3849c77e0bcc425387ccc770235390eb : SUCCESS in 8m 11s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/63066ff206c64f958eefb49393dad91c : SUCCESS in 6m 21s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/022cba5297f04952a738db1da599cd0b : SUCCESS in 1h 12m 10s","accounts_in_message":[],"_revision_number":12},{"id":"53dcf361b58f9cb81c71babf90c5477d342f762b","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2021-04-27 20:52:57.000000000","message":"Patch Set 12: Code-Review+2\n\nNice!","accounts_in_message":[],"_revision_number":12},{"id":"fc9e5b3deef6f8c0ceb5dab0624cb92baee29fdf","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2021-04-28 07:05:05.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"222393aa3e68842ade4946f3ae3d5b534fdfba0e","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-04-28 08:51:28.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"cadfd729252682595dca472aa8a22c3dc73dd181","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-04-28 08:53:15.000000000","message":"Patch Set 13: Published edit on patch set 12.","accounts_in_message":[],"_revision_number":13},{"id":"187162f792dbb04a953700174583344e12577323","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2021-04-28 08:56:16.000000000","message":"Patch Set 13: Code-Review+2 Workflow+1\n\nThe previous PS had already +2 and the change was just in log formatting. Hence approving without second +2","accounts_in_message":[],"_revision_number":13},{"id":"ce42050ec33ed12a6fa9d7cf9fe530afae2c7f7a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-28 10:06:41.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8a3354fb5fc24bbaacb3ec30b5ef328e : SUCCESS in 4m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/22e4a17660aa470dab67809bf178efe5 : SUCCESS in 4m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/236f36e79ae642ce9dce55c77b72b263 : SUCCESS in 4m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bb7beda2bcbd456aac2570ed225f7fa4 : POST_FAILURE in 4m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8d4b45cd0bfd493889a53565cda41a00 : SUCCESS in 4m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8546927e3343495abfa630caea46b5c5 : SUCCESS in 5m 17s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/30c7383738bd49368125547483243cbd : SUCCESS in 5m 38s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/bb3572f61b2e410ab01594b0b7741c1e : SUCCESS in 1h 12m 30s","accounts_in_message":[],"_revision_number":13},{"id":"ddc21f9bacb135085b46b14478a497d97e5815d2","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-04-28 10:16:11.000000000","message":"Patch Set 13:\n\nrecheck","accounts_in_message":[],"_revision_number":13},{"id":"b9f9ca5606759e1bf1b004d0b6e618719d4bd69e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-28 11:25:07.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0f92382a0f5c49f79f963efc42c1472c : SUCCESS in 4m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4685148736684ab383f25f32e8f6c461 : SUCCESS in 3m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e259cb2e46054b4fb3bd23402c5f87da : SUCCESS in 3m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4423b635f53d4289aa93acffc93ba4a2 : SUCCESS in 3m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3290d87c09ab42008f19cbf97a02790e : SUCCESS in 4m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9856df61286487ebb1e3f98be9d6644 : SUCCESS in 6m 27s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/080278ce4bee481b88c1aa927913cebf : SUCCESS in 6m 15s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/956ac2999aa044498e8ea9e491d91c77 : SUCCESS in 1h 07m 09s","accounts_in_message":[],"_revision_number":13},{"id":"d8b921c419b59e0fd09fa2cbb07a9cd4bb27fc28","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-28 11:25:25.000000000","message":"Patch Set 13: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":13},{"id":"57c5485fdca3f707cc3fffbfbe012e9836e25b30","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-28 12:45:17.000000000","message":"Patch Set 13: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b66098a506a84fbab411afca071c64c0 : SUCCESS in 4m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/698a8bb2e261442c807c34540989f66a : SUCCESS in 3m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/59baaeca471e438d88829d475503a213 : SUCCESS in 3m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1e119304855c4faa974b1fa9f52bf8df : SUCCESS in 3m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a857b6b501f840eca7a4b99fd275a55f : SUCCESS in 6m 12s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/2c35cf1ab2f344f19d3777b496be5ae7 : SUCCESS in 6m 43s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/301b23ae99ee413481cd91de34efe935 : SUCCESS in 1h 13m 44s","accounts_in_message":[],"_revision_number":13},{"id":"948165948669260e27f1289c8b2c429670d22220","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-28 12:45:18.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":13},{"id":"46fbc8b12e01691ef651b5e83b04bf7ac7b99adf","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-28 12:47:36.000000000","message":"Patch Set 13:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/475ca625ad8e464b84f3b581c9791e08 : SUCCESS in 1m 53s","accounts_in_message":[],"_revision_number":13}],"current_revision_number":13,"current_revision":"bc004045cd30a708eadd64ded1af1b30ef1d1918","revisions":{"f1a18b381a799a4d4f8315711145963ae35a5fe0":{"kind":"REWORK","_number":1,"created":"2021-04-08 15:10:29.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/33/785433/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/33/785433/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/33/785433/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/33/785433/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/33/785433/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/33/785433/1"}}},"commit":{"parents":[{"commit":"627b14e2a784c7eddd7b82d35b4ff00968f994e1","subject":"Add Python3 xena unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/627b14e2a784c7eddd7b82d35b4ff00968f994e1"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-04-08 13:41:21.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-04-08 15:02:58.000000000","tz":0},"subject":"Add an active wait in the \"Backend.lookup\"","message":"Add an active wait in the \"Backend.lookup\"\n\nAdded an active wait in the method \"Backend.lookup\". If a timeout\nand a RowEventHandler instance, listening on the same IDL connection,\nare passed, if the \"lookup\" method does not find the requested\nregister, the method creates a WaitEvent instance related to the\ntable and register requested. Once the wait event returns, the method\ntries to retrieve again the register. The second time, if the register\nif not found, an \"RowNotFound\" exception is raised or the default\nvalue is returned.\n\nCloses-Bug: #1922934\n\nChange-Id: I580bfb640d8157b4e749d7b1fe5de7568e99943d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/f1a18b381a799a4d4f8315711145963ae35a5fe0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/f1a18b381a799a4d4f8315711145963ae35a5fe0"}]},"branch":"refs/heads/master"},"0fc9ef29dee99358d8d7030273493dabd5233114":{"kind":"REWORK","_number":2,"created":"2021-04-09 06:56:01.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/33/785433/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/33/785433/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/33/785433/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/33/785433/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/33/785433/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/33/785433/2"}}},"commit":{"parents":[{"commit":"627b14e2a784c7eddd7b82d35b4ff00968f994e1","subject":"Add Python3 xena unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/627b14e2a784c7eddd7b82d35b4ff00968f994e1"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-04-08 13:41:21.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-04-09 06:55:44.000000000","tz":0},"subject":"Add an active wait in the \"Backend.lookup\"","message":"Add an active wait in the \"Backend.lookup\"\n\nAdded an active wait in the method \"Backend.lookup\". If a timeout\nand a RowEventHandler instance, listening on the same IDL connection,\nare passed, if the \"lookup\" method does not find the requested\nregister, the method creates a WaitEvent instance related to the\ntable and register requested. Once the wait event returns, the method\ntries to retrieve again the register. The second time, if the register\nif not found, an \"RowNotFound\" exception is raised or the default\nvalue is returned.\n\nCloses-Bug: #1922934\n\nChange-Id: I580bfb640d8157b4e749d7b1fe5de7568e99943d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/0fc9ef29dee99358d8d7030273493dabd5233114"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/0fc9ef29dee99358d8d7030273493dabd5233114"}]},"branch":"refs/heads/master"},"5d39272ae1167b17cbffe048b490e40d12b17a8b":{"kind":"REWORK","_number":3,"created":"2021-04-15 11:13:22.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/33/785433/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/33/785433/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/33/785433/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/33/785433/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/33/785433/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/33/785433/3"}}},"commit":{"parents":[{"commit":"627b14e2a784c7eddd7b82d35b4ff00968f994e1","subject":"Add Python3 xena unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/627b14e2a784c7eddd7b82d35b4ff00968f994e1"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-04-08 13:41:21.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-04-15 11:13:09.000000000","tz":0},"subject":"Add an active wait in the \"Backend.lookup\"","message":"Add an active wait in the \"Backend.lookup\"\n\nAdded an active wait in the method \"Backend.lookup\". If a timeout\nand a RowEventHandler instance, listening on the same IDL connection,\nare passed, if the \"lookup\" method does not find the requested\nregister, the method creates a WaitEvent instance related to the\ntable and register requested. Once the wait event returns, the method\ntries to retrieve again the register. The second time, if the register\nif not found, an \"RowNotFound\" exception is raised or the default\nvalue is returned.\n\nCloses-Bug: #1922934\n\nChange-Id: I580bfb640d8157b4e749d7b1fe5de7568e99943d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/5d39272ae1167b17cbffe048b490e40d12b17a8b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/5d39272ae1167b17cbffe048b490e40d12b17a8b"}]},"branch":"refs/heads/master"},"2ef0408e72a6991509070809dcf8016589ad9be3":{"kind":"REWORK","_number":4,"created":"2021-04-15 13:37:15.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/33/785433/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/33/785433/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/33/785433/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/33/785433/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/33/785433/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/33/785433/4"}}},"commit":{"parents":[{"commit":"627b14e2a784c7eddd7b82d35b4ff00968f994e1","subject":"Add Python3 xena unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/627b14e2a784c7eddd7b82d35b4ff00968f994e1"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-04-08 13:41:21.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-04-15 13:36:57.000000000","tz":0},"subject":"Add an active wait in the \"Backend.lookup\"","message":"Add an active wait in the \"Backend.lookup\"\n\nAdded an active wait in the method \"Backend.lookup\". If a timeout\nand a RowEventHandler instance, listening on the same IDL connection,\nare passed, if the \"lookup\" method does not find the requested\nregister, the method creates a WaitEvent instance related to the\ntable and register requested. Once the wait event returns, the method\ntries to retrieve again the register. The second time, if the register\nif not found, an \"RowNotFound\" exception is raised or the default\nvalue is returned.\n\nCloses-Bug: #1922934\n\nChange-Id: I580bfb640d8157b4e749d7b1fe5de7568e99943d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/2ef0408e72a6991509070809dcf8016589ad9be3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/2ef0408e72a6991509070809dcf8016589ad9be3"}]},"branch":"refs/heads/master"},"6065ac223b5c2943a1252694bcbe6321098d2302":{"kind":"REWORK","_number":5,"created":"2021-04-16 09:16:42.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/33/785433/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/33/785433/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/33/785433/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/33/785433/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/33/785433/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/33/785433/5"}}},"commit":{"parents":[{"commit":"627b14e2a784c7eddd7b82d35b4ff00968f994e1","subject":"Add Python3 xena unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/627b14e2a784c7eddd7b82d35b4ff00968f994e1"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-04-08 13:41:21.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-04-16 09:16:28.000000000","tz":0},"subject":"Add an active wait in the \"Backend.lookup\"","message":"Add an active wait in the \"Backend.lookup\"\n\nAdded an active wait in the method \"Backend.lookup\". If a timeout\nand a RowEventHandler instance, listening on the same IDL connection,\nare passed, if the \"lookup\" method does not find the requested\nregister, the method creates a WaitEvent instance related to the\ntable and register requested. Once the wait event returns, the method\ntries to retrieve again the register. The second time, if the register\nif not found, an \"RowNotFound\" exception is raised or the default\nvalue is returned.\n\nCloses-Bug: #1922934\n\nChange-Id: I580bfb640d8157b4e749d7b1fe5de7568e99943d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/6065ac223b5c2943a1252694bcbe6321098d2302"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/6065ac223b5c2943a1252694bcbe6321098d2302"}]},"branch":"refs/heads/master"},"9c5b98a673109ae799c0f79fa1c79fe21b5ffe75":{"kind":"REWORK","_number":6,"created":"2021-04-16 10:37:56.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/33/785433/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/33/785433/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/33/785433/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/33/785433/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/33/785433/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/33/785433/6"}}},"commit":{"parents":[{"commit":"627b14e2a784c7eddd7b82d35b4ff00968f994e1","subject":"Add Python3 xena unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/627b14e2a784c7eddd7b82d35b4ff00968f994e1"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-04-08 13:41:21.000000000","tz":0},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2021-04-16 10:37:22.000000000","tz":0},"subject":"Add an active wait in the \"Backend.lookup\"","message":"Add an active wait in the \"Backend.lookup\"\n\nAdded an active wait in the method \"Backend.lookup\". If a timeout\nand a RowEventHandler instance, listening on the same IDL connection,\nare passed, if the \"lookup\" method does not find the requested\nregister, the method creates a WaitEvent instance related to the\ntable and register requested. Once the wait event returns, the method\ntries to retrieve again the register. The second time, if the register\nif not found, an \"RowNotFound\" exception is raised or the default\nvalue is returned.\n\nCloses-Bug: #1922934\n\nChange-Id: I580bfb640d8157b4e749d7b1fe5de7568e99943d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/9c5b98a673109ae799c0f79fa1c79fe21b5ffe75"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/9c5b98a673109ae799c0f79fa1c79fe21b5ffe75"}]},"branch":"refs/heads/master"},"021abd3a97da80cb5372e12d34e435b3c4aff06c":{"kind":"REWORK","_number":7,"created":"2021-04-19 10:12:31.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/33/785433/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/33/785433/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/33/785433/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/33/785433/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/33/785433/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/33/785433/7"}}},"commit":{"parents":[{"commit":"627b14e2a784c7eddd7b82d35b4ff00968f994e1","subject":"Add Python3 xena unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/627b14e2a784c7eddd7b82d35b4ff00968f994e1"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-04-08 13:41:21.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-04-19 10:12:16.000000000","tz":0},"subject":"Add an active wait in the \"Backend.lookup\"","message":"Add an active wait in the \"Backend.lookup\"\n\nAdded an active wait in the method \"Backend.lookup\". If a timeout\nand a RowEventHandler instance, listening on the same IDL connection,\nare passed, if the \"lookup\" method does not find the requested\nregister, the method creates a WaitEvent instance related to the\ntable and register requested. Once the wait event returns, the method\ntries to retrieve again the register. The second time, if the register\nif not found, an \"RowNotFound\" exception is raised or the default\nvalue is returned.\n\nCloses-Bug: #1922934\n\nChange-Id: I580bfb640d8157b4e749d7b1fe5de7568e99943d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/021abd3a97da80cb5372e12d34e435b3c4aff06c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/021abd3a97da80cb5372e12d34e435b3c4aff06c"}]},"branch":"refs/heads/master"},"1a01563aa5ec462c9f66c4e61d4762ce5e21ea6d":{"kind":"REWORK","_number":8,"created":"2021-04-19 11:51:37.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/33/785433/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/33/785433/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/33/785433/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/33/785433/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/33/785433/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/33/785433/8"}}},"commit":{"parents":[{"commit":"627b14e2a784c7eddd7b82d35b4ff00968f994e1","subject":"Add Python3 xena unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/627b14e2a784c7eddd7b82d35b4ff00968f994e1"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-04-08 13:41:21.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-04-19 11:51:23.000000000","tz":0},"subject":"Add an active wait in the \"Backend.lookup\"","message":"Add an active wait in the \"Backend.lookup\"\n\nAdded an active wait in the method \"Backend.lookup\". If a timeout\nand a RowEventHandler instance, listening on the same IDL connection,\nare passed, if the \"lookup\" method does not find the requested\nregister, the method creates a WaitEvent instance related to the\ntable and register requested. Once the wait event returns, the method\ntries to retrieve again the register. The second time, if the register\nif not found, an \"RowNotFound\" exception is raised or the default\nvalue is returned.\n\nCloses-Bug: #1922934\n\nChange-Id: I580bfb640d8157b4e749d7b1fe5de7568e99943d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/1a01563aa5ec462c9f66c4e61d4762ce5e21ea6d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/1a01563aa5ec462c9f66c4e61d4762ce5e21ea6d"}]},"branch":"refs/heads/master"},"5f8cafc8ae1252e9428618d934b9dc1877624abc":{"kind":"REWORK","_number":9,"created":"2021-04-19 17:14:14.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/33/785433/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/33/785433/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/33/785433/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/33/785433/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/33/785433/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/33/785433/9"}}},"commit":{"parents":[{"commit":"627b14e2a784c7eddd7b82d35b4ff00968f994e1","subject":"Add Python3 xena unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/627b14e2a784c7eddd7b82d35b4ff00968f994e1"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-04-08 13:41:21.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-04-19 17:13:38.000000000","tz":0},"subject":"Add an active wait in the \"Backend.lookup\"","message":"Add an active wait in the \"Backend.lookup\"\n\nAdded an active wait in the method \"Backend.lookup\". If a timeout\nand a RowEventHandler instance, listening on the same IDL connection,\nare passed, if the \"lookup\" method does not find the requested\nregister, the method creates a WaitEvent instance related to the\ntable and register requested. Once the wait event returns, the method\ntries to retrieve again the register. The second time, if the register\nif not found, an \"RowNotFound\" exception is raised or the default\nvalue is returned.\n\nCloses-Bug: #1922934\n\nChange-Id: I580bfb640d8157b4e749d7b1fe5de7568e99943d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/5f8cafc8ae1252e9428618d934b9dc1877624abc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/5f8cafc8ae1252e9428618d934b9dc1877624abc"}]},"branch":"refs/heads/master"},"86aaa93c965ae0da4c8af927b1e406b237eda5de":{"kind":"REWORK","_number":10,"created":"2021-04-19 17:15:27.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/33/785433/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/33/785433/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/33/785433/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/33/785433/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/33/785433/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/33/785433/10"}}},"commit":{"parents":[{"commit":"627b14e2a784c7eddd7b82d35b4ff00968f994e1","subject":"Add Python3 xena unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/627b14e2a784c7eddd7b82d35b4ff00968f994e1"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-04-08 13:41:21.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-04-19 17:15:15.000000000","tz":0},"subject":"Add an active wait in the \"Backend.lookup\"","message":"Add an active wait in the \"Backend.lookup\"\n\nAdded an active wait in the method \"Backend.lookup\". If a timeout\nand a RowEventHandler instance, listening on the same IDL connection,\nare passed, if the \"lookup\" method does not find the requested\nregister, the method creates a WaitEvent instance related to the\ntable and register requested. Once the wait event returns, the method\ntries to retrieve again the register. The second time, if the register\nif not found, an \"RowNotFound\" exception is raised or the default\nvalue is returned.\n\nCloses-Bug: #1922934\n\nChange-Id: I580bfb640d8157b4e749d7b1fe5de7568e99943d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/86aaa93c965ae0da4c8af927b1e406b237eda5de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/86aaa93c965ae0da4c8af927b1e406b237eda5de"}]},"branch":"refs/heads/master"},"63509d68741669a2ed701765b88d7cbf82002ac1":{"kind":"REWORK","_number":11,"created":"2021-04-21 15:19:14.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/33/785433/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/33/785433/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/33/785433/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/33/785433/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/33/785433/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/33/785433/11"}}},"commit":{"parents":[{"commit":"627b14e2a784c7eddd7b82d35b4ff00968f994e1","subject":"Add Python3 xena unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/627b14e2a784c7eddd7b82d35b4ff00968f994e1"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-04-08 13:41:21.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-04-21 15:18:58.000000000","tz":0},"subject":"Add an active wait in the \"Backend.lookup\"","message":"Add an active wait in the \"Backend.lookup\"\n\nAdded an active wait in the method \"Backend.lookup\". If a timeout\nand a RowEventHandler instance, listening on the same IDL connection,\nare passed, if the \"lookup\" method does not find the requested\nregister, the method creates a WaitEvent instance related to the\ntable and register requested. Once the wait event returns, the method\ntries to retrieve again the register. The second time, if the register\nif not found, an \"RowNotFound\" exception is raised or the default\nvalue is returned.\n\nCloses-Bug: #1922934\n\nChange-Id: I580bfb640d8157b4e749d7b1fe5de7568e99943d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/63509d68741669a2ed701765b88d7cbf82002ac1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/63509d68741669a2ed701765b88d7cbf82002ac1"}]},"branch":"refs/heads/master"},"713fc8174f618329061246b6a8ac62aa9eb1eac0":{"kind":"REWORK","_number":12,"created":"2021-04-27 16:50:21.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/33/785433/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/33/785433/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/33/785433/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/33/785433/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/33/785433/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/33/785433/12"}}},"commit":{"parents":[{"commit":"627b14e2a784c7eddd7b82d35b4ff00968f994e1","subject":"Add Python3 xena unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/627b14e2a784c7eddd7b82d35b4ff00968f994e1"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-04-08 13:41:21.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-04-27 16:46:42.000000000","tz":0},"subject":"Add an active wait in the \"Backend.lookup\"","message":"Add an active wait in the \"Backend.lookup\"\n\nAdded an active wait in the method \"Backend.lookup\". If a timeout\nand a RowEventHandler instance, listening on the same IDL connection,\nare passed, if the \"lookup\" method does not find the requested\nregister, the method creates a WaitEvent instance related to the\ntable and register requested.\n\nIf the wait event returns, the expected record will be stored in the\nevent \"result\" member and will be the value returned by this method.\n\nIf the wait event timeout expires, the method will throw an\n``idlutils.RowNotFound`` exception.\n\nCloses-Bug: #1922934\n\nChange-Id: I580bfb640d8157b4e749d7b1fe5de7568e99943d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/713fc8174f618329061246b6a8ac62aa9eb1eac0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/713fc8174f618329061246b6a8ac62aa9eb1eac0"}]},"branch":"refs/heads/master"},"bc004045cd30a708eadd64ded1af1b30ef1d1918":{"kind":"REWORK","_number":13,"created":"2021-04-28 08:53:15.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/33/785433/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/33/785433/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/33/785433/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/33/785433/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/33/785433/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/33/785433/13"}}},"commit":{"parents":[{"commit":"627b14e2a784c7eddd7b82d35b4ff00968f994e1","subject":"Add Python3 xena unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/627b14e2a784c7eddd7b82d35b4ff00968f994e1"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-04-08 13:41:21.000000000","tz":0},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2021-04-28 08:53:09.000000000","tz":0},"subject":"Add an active wait in the \"Backend.lookup\"","message":"Add an active wait in the \"Backend.lookup\"\n\nAdded an active wait in the method \"Backend.lookup\". If a timeout\nand a RowEventHandler instance, listening on the same IDL connection,\nare passed, if the \"lookup\" method does not find the requested\nregister, the method creates a WaitEvent instance related to the\ntable and register requested.\n\nIf the wait event returns, the expected record will be stored in the\nevent \"result\" member and will be the value returned by this method.\n\nIf the wait event timeout expires, the method will throw an\n``idlutils.RowNotFound`` exception.\n\nCloses-Bug: #1922934\n\nChange-Id: I580bfb640d8157b4e749d7b1fe5de7568e99943d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/bc004045cd30a708eadd64ded1af1b30ef1d1918"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/bc004045cd30a708eadd64ded1af1b30ef1d1918"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
