)]}'
{"id":"openstack%2Fovsdbapp~786609","triplet_id":"openstack%2Fovsdbapp~master~I89c7bea2c95028c515d7cf573e99431676e1dcdd","project":"openstack/ovsdbapp","branch":"master","topic":"bug/1924716","hashtags":[],"change_id":"I89c7bea2c95028c515d7cf573e99431676e1dcdd","subject":"Improve \"OvsdbIdl.from_server\" method","status":"MERGED","created":"2021-04-16 09:34:53.000000000","updated":"2021-08-02 21:45:15.000000000","submitted":"2021-08-02 21:43:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"786609-bug/1924716","meta_rev_id":"5993c009f98450ac46940c6da1de52815f0fad9a","_number":786609,"virtual_id_number":786609,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-02 21:43:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"all":[{"value":2,"date":"2021-08-02 20:50:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-08-02 21:45:15.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"all":[{"value":1,"date":"2021-08-02 20:50:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"reviewers":{"REVIEWER":[{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-16 11:21:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-06 10:58:33.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"state":"REVIEWER"},{"updated":"2021-05-06 10:58:33.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"state":"REVIEWER"},{"updated":"2021-08-02 20:50:02.000000000","updated_by":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"reviewer":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"state":"REVIEWER"}],"messages":[{"id":"6117acff78b0c625f0cbab952487f78939c86cbd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-04-16 09:34:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6c27957b2c227feba9b0c54623b6b8871737be3a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-04-16 10:12:31.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"24861aad737a6b117efc95e7122948f7f759b8bd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-16 11:21:36.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/17c0378f543d4c728e5c7109b0f39fe7 : SUCCESS in 5m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e9961b1707849acabc70affa8eef855 : SUCCESS in 3m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1fe6b9ef642d4b989332427c25ff977e : SUCCESS in 4m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/13acb6aaf2ba4cb1bddac1c5304a2b8a : SUCCESS in 4m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/deb5f1e1fe29454f9972dcd5ffb57930 : SUCCESS in 3m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/17b2e124ccd84526b6259499d94067dc : SUCCESS in 6m 02s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/8e783bfdacd44a38a890599daccc10ff : SUCCESS in 7m 27s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/fa069570ebd14421aecd0d53c6b2dc18 : SUCCESS in 1h 07m 50s","accounts_in_message":[],"_revision_number":2},{"id":"1d2b46df5eb2f78d739b6e8af7a5f0136d8564d8","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2021-05-10 07:50:47.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d20a35b8cd08fcd614cebaeb24aee9c1637c3228","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-05-10 16:04:31.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4b4972a34f6c9427ac646441f93e7639458361de","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2021-05-11 07:52:01.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e57d3cecf9bc868998da55be1f17838ec4be9a60","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-05-11 16:18:19.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5954751fc24f5bb4f959eac4bad9846007c25649","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-05-11 16:24:46.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4649d248ffc923e985a05ffe1736a56e896f8a1b","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2021-05-11 19:04:52.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0d4e1c7f79732b8fceb9602794f1f77a50fd9d36","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2021-05-11 19:20:00.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)\n\nI\u0027m ok with this if this is where everyone thinks is a good place to draw the line on what schema restrictions we do w/ arguments and what ones we do buy calling the underlying SchemaHelper.register_* methods directly in subclasses. I originally considered these methods more of a reference implementation as opposed to an API. Generally anything in ovsdbapp.backend was supposed to be hidden from the users, that just never really survived first contact with the real world.","accounts_in_message":[],"_revision_number":2},{"id":"d7e5bbc3c9529fe8e44e66e9b05b48e200f07dfd","tag":"autogenerated:gerrit:abandon","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-06-11 15:27:28.000000000","message":"Abandoned\n\nI don\u0027t see too much traction on this, let\u0027s keep the current implementation","accounts_in_message":[],"_revision_number":2},{"id":"a9e72e87db59dbed63fab368c23af8b51daf5a25","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2021-06-11 19:11:36.000000000","message":"Patch Set 2:\n\nsorry, this just fell off my radar. I\u0027m not actually opposed to this change since it\u0027s backwards compatible.","accounts_in_message":[],"_revision_number":2},{"id":"c0de869ee515aaf4c62f66ad3e9bbac992344e09","tag":"autogenerated:gerrit:restore","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2021-06-11 19:35:16.000000000","message":"Restored","accounts_in_message":[],"_revision_number":2},{"id":"dad8752fc03c2471eca9a5a888b6e087d2589237","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2021-06-11 19:35:26.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"469edf416f1c0ea92c95d591616d46a60e70ee32","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-11 20:09:19.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7b019132f4f14412940628d7a7702a08 : SUCCESS in 3m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6fa34b9e24684fe1865f5b2a0bef0c98 : SUCCESS in 4m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7882333f1908437697eac9bc3e1ba36f : SUCCESS in 9m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d97919a0ba7a456781071970437de993 : SUCCESS in 3m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5d53a42c15b54262851b27fcfc88e319 : SUCCESS in 4m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/079e0be2dfb24403ab509a168603b2c3 : SUCCESS in 5m 11s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/05b2aca57b544a9f8d8edc69018ce679 : SUCCESS in 7m 07s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/e5ffabaa630544f0b40d9088ac8c74bb : FAILURE in 33m 07s","accounts_in_message":[],"_revision_number":2},{"id":"9dedef5a79f9bd525a8d77ecb2e645569244f969","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2021-07-27 20:01:11.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"c1888f48af0a4a5648a35ddbb41974ac7c0de9a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-27 21:10:14.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/12c9b1f6bf2f4b09a3307e407234027f : SUCCESS in 4m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f61cb0b9786e4f3786f857b5d26495cd : SUCCESS in 5m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bb863afa755c4a83b1048c86b1d180f6 : SUCCESS in 4m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1d5b4369348b46029913f59f5a70e983 : SUCCESS in 4m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/968d867391df4c7ea8eca8c891c501cb : SUCCESS in 5m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/556f60fcf59d4060a824eb685f971000 : SUCCESS in 6m 26s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/f862bb5d657b4b059bff46cd31c7f68e : SUCCESS in 6m 39s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/be3bb31bbe784ecd8c7a280fe50b0e60 : SUCCESS in 1h 02m 56s","accounts_in_message":[],"_revision_number":3},{"id":"cca77c1f7cb9941d9b5ed690effa8ac807cf1ecf","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2021-08-02 20:06:30.000000000","message":"Patch Set 3:\n\nI would like to reiterate that I think this is a great improvement. ;)","accounts_in_message":[],"_revision_number":3},{"id":"7e4d0d4cb0d9fdcd332bcb463c928158afcca34b","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2021-08-02 20:50:02.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\nlgtm","accounts_in_message":[],"_revision_number":3},{"id":"344562d5252f09bc2172a96b5d8f6b30fa0e7ac5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-02 20:50:30.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"986d2f7a12bb61f57eeb3c058f475d4e2ce34652","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-02 21:43:58.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5a35f6fd3582407e83cd6ca5749fb96a : SUCCESS in 5m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7fa6bc0f66ab458ea040c9415161a391 : SUCCESS in 4m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d017fe542a924fafae0d743d89cb5818 : SUCCESS in 6m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0322c34cb4254dac8497012b74581fda : SUCCESS in 4m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/69a9ef9ad0554513a48ad31f72451562 : SUCCESS in 6m 36s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/0b8744be065a4e819f26d8066937395e : SUCCESS in 7m 19s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/09288c46b7f749509a93114ef3e66e33 : SUCCESS in 52m 34s","accounts_in_message":[],"_revision_number":3},{"id":"e4824f4e8bc590c6319fcd75871941f299ba2a94","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-02 21:43:58.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"81ee785a8447644e8b5574203d2dc2df5813c199","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-02 21:45:15.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5564719858bb42fa9db26fd601aa4660 : SUCCESS in 59s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"c8cb43eaa2a87bed1ac66571c3c9baaa0db3d578","revisions":{"a4094ade8f6be2164c90081abeeacef982ad59d9":{"kind":"REWORK","_number":1,"created":"2021-04-16 09:34:53.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/09/786609/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/09/786609/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/09/786609/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/09/786609/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/09/786609/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/09/786609/1"}}},"commit":{"parents":[{"commit":"627b14e2a784c7eddd7b82d35b4ff00968f994e1","subject":"Add Python3 xena unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/627b14e2a784c7eddd7b82d35b4ff00968f994e1"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-04-16 09:29:54.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-04-16 09:29:54.000000000","tz":0},"subject":"Improve \"OvsdbIdl.from_server\" method","message":"Improve \"OvsdbIdl.from_server\" method\n\nAllow other positional and key-worded arguments in the method\nsignature. That will allow to, for example, implement other child\nclasses \"from_server\" method without having an incompatible signature.\n\nThis patch also adds key-worded argument \"helper_tables\". If any,\nthe helper will only register the provided tables instead of registering\nall.\n\nChange-Id: I89c7bea2c95028c515d7cf573e99431676e1dcdd\nCloses-Bug: #1924716\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/a4094ade8f6be2164c90081abeeacef982ad59d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/a4094ade8f6be2164c90081abeeacef982ad59d9"}]},"branch":"refs/heads/master"},"35d0772a0028bb8baba4c5955b33e5274725c740":{"kind":"REWORK","_number":2,"created":"2021-04-16 10:12:31.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/09/786609/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/09/786609/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/09/786609/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/09/786609/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/09/786609/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/09/786609/2"}}},"commit":{"parents":[{"commit":"627b14e2a784c7eddd7b82d35b4ff00968f994e1","subject":"Add Python3 xena unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/627b14e2a784c7eddd7b82d35b4ff00968f994e1"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-04-16 09:29:54.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-04-16 10:00:34.000000000","tz":0},"subject":"Improve \"OvsdbIdl.from_server\" method","message":"Improve \"OvsdbIdl.from_server\" method\n\nAllow other positional and key-worded arguments in the method\nsignature. That will allow to, for example, implement other child\nclasses \"from_server\" method without having an incompatible signature.\nThe key-worded variables are passed to the class \"idl.Idl\"\ninstantiation. Now \"probe_interval\" or \"leader_only\" can be passed to\n\"idl.Idl.__init__\" method.\n\nThis patch also adds two new key-worded variables:\n- \"helper_tables\": the helper will only register the provided tables\n  instead of registering all.\n- \"helper\": a \"idl.SchemaHelper\" instance; if passed, the method won\u0027t\n  create the helper using the schema_name.\n\nChange-Id: I89c7bea2c95028c515d7cf573e99431676e1dcdd\nCloses-Bug: #1924716\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/35d0772a0028bb8baba4c5955b33e5274725c740"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/35d0772a0028bb8baba4c5955b33e5274725c740"}]},"branch":"refs/heads/master"},"c8cb43eaa2a87bed1ac66571c3c9baaa0db3d578":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-07-27 20:01:11.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/09/786609/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/09/786609/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/09/786609/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/09/786609/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/09/786609/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/09/786609/3"}}},"commit":{"parents":[{"commit":"f02b1c20c062cca190d9e39bbf8612f1663bf604","subject":"Make exceptions in the event logs easier to read","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/f02b1c20c062cca190d9e39bbf8612f1663bf604"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-04-16 09:29:54.000000000","tz":0},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2021-07-27 20:01:11.000000000","tz":0},"subject":"Improve \"OvsdbIdl.from_server\" method","message":"Improve \"OvsdbIdl.from_server\" method\n\nAllow other positional and key-worded arguments in the method\nsignature. That will allow to, for example, implement other child\nclasses \"from_server\" method without having an incompatible signature.\nThe key-worded variables are passed to the class \"idl.Idl\"\ninstantiation. Now \"probe_interval\" or \"leader_only\" can be passed to\n\"idl.Idl.__init__\" method.\n\nThis patch also adds two new key-worded variables:\n- \"helper_tables\": the helper will only register the provided tables\n  instead of registering all.\n- \"helper\": a \"idl.SchemaHelper\" instance; if passed, the method won\u0027t\n  create the helper using the schema_name.\n\nChange-Id: I89c7bea2c95028c515d7cf573e99431676e1dcdd\nCloses-Bug: #1924716\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/c8cb43eaa2a87bed1ac66571c3c9baaa0db3d578"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/c8cb43eaa2a87bed1ac66571c3c9baaa0db3d578"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
