)]}'
{"id":"openstack%2Fovsdbapp~788916","triplet_id":"openstack%2Fovsdbapp~stable%2Fwallaby~I580bfb640d8157b4e749d7b1fe5de7568e99943d","project":"openstack/ovsdbapp","branch":"stable/wallaby","topic":"bug/1922934-stable/wallaby","hashtags":[],"change_id":"I580bfb640d8157b4e749d7b1fe5de7568e99943d","subject":"Add an active wait in the \"Backend.lookup\"","status":"MERGED","created":"2021-04-30 08:45:03.000000000","updated":"2021-05-12 15:35:28.000000000","submitted":"2021-05-12 15:34:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"788916-bug/1922934-stable/wallaby","cherry_pick_of_change":785433,"cherry_pick_of_patch_set":13,"meta_rev_id":"bba403fbb78da8f7c049e2d2e573f3371c07f1d3","_number":788916,"virtual_id_number":788916,"owner":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-12 15:34:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"all":[{"value":2,"date":"2021-05-12 15:35:28.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"value":2,"date":"2021-05-12 15:35:28.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-05-12 15:35:28.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"all":[{"value":1,"date":"2021-05-12 13:35:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-30 08:45:03.000000000","updated_by":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2021-04-30 10:02:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-06 09:49:00.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"state":"REVIEWER"},{"updated":"2021-05-11 21:45:37.000000000","updated_by":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"reviewer":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"state":"REVIEWER"}],"messages":[{"id":"2ed24737bdd5ceb05005341a87523e2a84872a8b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2021-04-30 08:45:03.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"d4c58e3b47b4bba389542adf2065331a2499a803","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-30 10:02:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ea58f4b9fd054e8eb0043ae12c34fcc8 : SUCCESS in 4m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8c31df6d2a894a8b89cce788321f0fe7 : SUCCESS in 3m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9b1056100ef445168def968683bb54ed : SUCCESS in 4m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a22394ee3d59410fb1f45d2be8f866d3 : SUCCESS in 3m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cd518bf912304782afcb18e9bc2f5be1 : SUCCESS in 3m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62a9a1090c1f4e1da871b2741d1e5a39 : SUCCESS in 5m 21s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/34deb42c054e435d89120190256fb0e6 : SUCCESS in 7m 30s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/77283a0d2f844663a7375cf0c4e248c9 : SUCCESS in 1h 04m 26s","accounts_in_message":[],"_revision_number":1},{"id":"f9232a469fd262226e9d46b8e81d17f7d6a3c4fa","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-04-30 10:02:56.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"73b9fd13c776163a5d03a3188c82dec3972d6e7e","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2021-05-11 20:51:29.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c678e86a98899bf49f166bcc5594812d972c7910","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2021-05-11 21:44:47.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6e4c037c983aa04e44caa69ec479e344adb93136","tag":"autogenerated:gerrit:deleteVote","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2021-05-11 21:44:59.000000000","message":"Removed Workflow+1 by Terry Wilson \u003ctwilson@redhat.com\u003e\n","accounts_in_message":[],"_revision_number":1},{"id":"1217ba17731f8671b35ffe8297c805c425179a32","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-11 21:45:16.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"742a6cc60107f64a16373b73dd680f280e198220","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2021-05-11 21:45:37.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0ef27e38f25017516c65fb44833e70cb5f5e7a8e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-11 23:08:58.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/030e339ef2664d87b6116edfa399c978 : SUCCESS in 5m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/020252231b074cd0bc31978ba2531f0e : SUCCESS in 4m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2ccd0fa7634a46929a5cda5ce73c2b37 : SUCCESS in 4m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/df792a39ee1e4760a28013a80daf1f26 : SUCCESS in 3m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72d4a1b4bbfd492baa2f28121a9fce25 : FAILURE in 5m 58s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/087f06b2c8944ce2b0b7cd178e1b1014 : SUCCESS in 6m 14s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/54505fbdf4514853b40c85c92bcc95e6 : SUCCESS in 1h 04m 17s","accounts_in_message":[],"_revision_number":1},{"id":"3b690c4b6f9ab1982698b4a60743ba1910975f02","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2021-05-12 08:39:10.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"207de8afbe762daa43d869edc312e47e16bd3e62","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-12 09:47:07.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/69ee855d801942c1b64c8e9e49724dbe : SUCCESS in 4m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/23dc09aec8a542e897ca09ac2f75e9f5 : SUCCESS in 4m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d13398b2192a46bc9274738229aaebba : SUCCESS in 4m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a28df7586b274cd5923e066d3fdd16fe : SUCCESS in 3m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a663772bf4ce44c3ba3df40e64f1a962 : SUCCESS in 4m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/80bee08f57834c499589497929ea5fbc : FAILURE in 5m 58s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/bf1035adff0d45ba9ff463f500ee6305 : SUCCESS in 8m 30s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/0ebc72ce281a49a88878ba37be34a11c : SUCCESS in 1h 06m 53s","accounts_in_message":[],"_revision_number":1},{"id":"13935bf6a395f3b04dfaf76bc9b93e3715e6c428","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2021-05-12 13:34:57.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"2a1588ae72102cdd5c5c07eb397a24f3e11b5843","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2021-05-12 13:35:47.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"49d798b6bb0ff044f3eb9eac831e159aba7f1518","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-12 14:33:31.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bdf8f80ce47e4f7d80acc3863934eff0 : SUCCESS in 4m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f1520a09d51942a1b5554aa6a290553b : SUCCESS in 4m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f3e4be5873a44bb4b34ff2028911784b : SUCCESS in 4m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/259108179a3d4e2f9c460efdfb412d85 : SUCCESS in 4m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1067dc8bff084aa39474f15300526791 : SUCCESS in 4m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b06aa90cb36e4465b69b3a5fa73ef2de : SUCCESS in 5m 50s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/37e587f4c44d4a19ae61d703c77bf87c : SUCCESS in 6m 19s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/17769ed41f874fac870018c0f3aea6c8 : SUCCESS in 55m 43s","accounts_in_message":[],"_revision_number":2},{"id":"022b0a5433b9f9536c0b50a2878646c94fb35169","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-12 14:34:01.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"7e9e4b05da373f2219f8567971959a7e609e1aae","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-12 15:34:15.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0ff352885c2a44cebd3b2515d3bd1e37 : SUCCESS in 4m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7788a6b0a0744a77a36add67cd4b297e : SUCCESS in 4m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/da43c304af8e4790a9508a479edefda1 : SUCCESS in 4m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cafe166f97ec407482d1a5b9e22c6a30 : SUCCESS in 4m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f364598312ea452fa588dad134225298 : SUCCESS in 5m 01s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/2aaf1ce0559941e696b487645f2e9fb8 : SUCCESS in 6m 25s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/098b719f87374c0c8a8ca1ee140884ae : SUCCESS in 59m 09s","accounts_in_message":[],"_revision_number":2},{"id":"0350bda279e9c353f15e1ce31d88459554d1adc8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-12 15:34:16.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"ad9e37eb1b7984b942c8f7387a65e65eb99fb6bd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-12 15:35:28.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/88797798520b4b25ba1ce227763f7bcc : SUCCESS in 52s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d0cc49248da613c5cb4a485f077ea3ab9d6cbbee","revisions":{"46c7766a6d542b250783ac232888e4241db69cc9":{"kind":"REWORK","_number":1,"created":"2021-04-30 08:45:03.000000000","uploader":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"ref":"refs/changes/16/788916/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/16/788916/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/16/788916/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/16/788916/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/16/788916/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/16/788916/1"}}},"commit":{"parents":[{"commit":"f6a19a35e0fd7df680ccf5fde77f001cd4b1ad2b","subject":"Update TOX_CONSTRAINTS_FILE for stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/f6a19a35e0fd7df680ccf5fde77f001cd4b1ad2b"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-04-08 13:41:21.000000000","tz":0},"committer":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2021-04-30 08:45:03.000000000","tz":0},"subject":"Add an active wait in the \"Backend.lookup\"","message":"Add an active wait in the \"Backend.lookup\"\n\nAdded an active wait in the method \"Backend.lookup\". If a timeout\nand a RowEventHandler instance, listening on the same IDL connection,\nare passed, if the \"lookup\" method does not find the requested\nregister, the method creates a WaitEvent instance related to the\ntable and register requested.\n\nIf the wait event returns, the expected record will be stored in the\nevent \"result\" member and will be the value returned by this method.\n\nIf the wait event timeout expires, the method will throw an\n``idlutils.RowNotFound`` exception.\n\nCloses-Bug: #1922934\n\nChange-Id: I580bfb640d8157b4e749d7b1fe5de7568e99943d\n(cherry picked from commit bc004045cd30a708eadd64ded1af1b30ef1d1918)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/46c7766a6d542b250783ac232888e4241db69cc9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/46c7766a6d542b250783ac232888e4241db69cc9"}]},"branch":"refs/heads/stable/wallaby"},"d0cc49248da613c5cb4a485f077ea3ab9d6cbbee":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-05-12 13:34:57.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/16/788916/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/16/788916/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/16/788916/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/16/788916/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/16/788916/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/16/788916/2"}}},"commit":{"parents":[{"commit":"007c0f646b7535616adc19adbbb4e3168f5efa5d","subject":"Fix docs job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/007c0f646b7535616adc19adbbb4e3168f5efa5d"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-04-08 13:41:21.000000000","tz":0},"committer":{"name":"Flavio Fernandes","email":"flavio@flaviof.com","date":"2021-05-12 13:34:57.000000000","tz":0},"subject":"Add an active wait in the \"Backend.lookup\"","message":"Add an active wait in the \"Backend.lookup\"\n\nAdded an active wait in the method \"Backend.lookup\". If a timeout\nand a RowEventHandler instance, listening on the same IDL connection,\nare passed, if the \"lookup\" method does not find the requested\nregister, the method creates a WaitEvent instance related to the\ntable and register requested.\n\nIf the wait event returns, the expected record will be stored in the\nevent \"result\" member and will be the value returned by this method.\n\nIf the wait event timeout expires, the method will throw an\n``idlutils.RowNotFound`` exception.\n\nCloses-Bug: #1922934\n\nChange-Id: I580bfb640d8157b4e749d7b1fe5de7568e99943d\n(cherry picked from commit bc004045cd30a708eadd64ded1af1b30ef1d1918)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/d0cc49248da613c5cb4a485f077ea3ab9d6cbbee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/d0cc49248da613c5cb4a485f077ea3ab9d6cbbee"}]},"branch":"refs/heads/stable/wallaby","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
