)]}'
{"id":"openstack%2Fovsdbapp~790028","triplet_id":"openstack%2Fovsdbapp~stable%2Fvictoria~I5d5139bb7f1ff1224e421d528e5a1bf2bbb20f33","project":"openstack/ovsdbapp","branch":"stable/victoria","topic":"bug/1922934","hashtags":[],"change_id":"I5d5139bb7f1ff1224e421d528e5a1bf2bbb20f33","subject":"Set the default \"lookup\" timeout to 2 seconds","status":"ABANDONED","created":"2021-05-06 09:51:59.000000000","updated":"2021-05-12 08:03:41.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"eb1c9879cc14d207e5f3799406a2b2e6a5abaff5","_number":790028,"virtual_id_number":790028,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2021-05-06 11:09:55.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-06 11:09:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"326f5ae73f6e8dd8dde51c866b8844774716fba5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-05-06 09:51:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4660b80a1aa92a34e686da7896fc59f3bc39069e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-06 11:09:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e7acf37a19bc4b6aafd81a9aab1e8560 : SUCCESS in 4m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aca8d862bb6244429452b989f3b832bc : SUCCESS in 4m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/434199033a284b6597567091780b53d7 : SUCCESS in 4m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1315fc632fa2415c86f118a53a14cc08 : SUCCESS in 6m 53s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/901a75a6e05540fd9af862241b83dc19 : SUCCESS in 6m 53s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/ad9d861b3a7b497ab91229e3c0be8479 : SUCCESS in 1h 16m 11s","accounts_in_message":[],"_revision_number":1},{"id":"eb1c9879cc14d207e5f3799406a2b2e6a5abaff5","tag":"autogenerated:gerrit:abandon","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-05-12 08:03:41.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2ac190fe9a8b65c3a9e1066e587317c2a9583b20","revisions":{"2ac190fe9a8b65c3a9e1066e587317c2a9583b20":{"kind":"REWORK","_number":1,"created":"2021-05-06 09:51:59.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/28/790028/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/28/790028/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/28/790028/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/28/790028/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/28/790028/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/28/790028/1"}}},"commit":{"parents":[{"commit":"ba4dd24e2fe3c517b2ec3005b7644bd6e980dafd","subject":"Add an active wait in the \"Backend.lookup\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/ba4dd24e2fe3c517b2ec3005b7644bd6e980dafd"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-05-05 07:54:38.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-05-06 09:51:51.000000000","tz":0},"subject":"Set the default \"lookup\" timeout to 2 seconds","message":"Set the default \"lookup\" timeout to 2 seconds\n\nIn order to broadly cover the issue described in the referred bug,\nthis patch sets a default timeout of 2 seconds in the\n``ovs_idl.Backend.lookup`` method.\n\nThis timeout should cover most of the situations where the IDL local\ncache update is delayed. This patch does not change the default\nbehavior, except it will not fail if the DB cache is synchronized\nwithin 2 seconds.\n\nIf we don\u0027t pass the notify handler or the backend does not implement\none, in case of not finding the requested register, the method will\nraise like before.\n\nRelated-Bug: #1922934\n\nChange-Id: I5d5139bb7f1ff1224e421d528e5a1bf2bbb20f33\n(cherry picked from commit 52026e0fe06e6d1d7ff8869a3f4682fedf0661a2)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/2ac190fe9a8b65c3a9e1066e587317c2a9583b20"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/2ac190fe9a8b65c3a9e1066e587317c2a9583b20"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
