)]}'
{"id":"openstack%2Fovsdbapp~790857","triplet_id":"openstack%2Fovsdbapp~stable%2Ftrain~I501659d48b1cf2d64cea8dc80ade7a068e37e904","project":"openstack/ovsdbapp","branch":"stable/train","hashtags":[],"change_id":"I501659d48b1cf2d64cea8dc80ade7a068e37e904","subject":"Fix docs job","status":"ABANDONED","created":"2021-05-11 22:14:17.000000000","updated":"2021-05-12 13:25:41.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"ef113156ab1bf3990914348cdf2132e3e6f26815","_number":790857,"virtual_id_number":790857,"owner":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-05-11 22:22:40.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"all":[{"value":-1,"date":"2021-05-11 22:21:56.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"all":[{"value":-1,"date":"2021-05-11 22:20:38.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-11 22:22:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"81988645b4b6d17a78f8563699bf55cf7b9fe8ea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2021-05-11 22:14:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2e07e044daa6ca48427a8967f7397a9272f0b014","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2021-05-11 22:20:38.000000000","message":"Patch Set 1: Workflow-1\n\nthis may not be needed after all: no docs job in stable/train for this repo!","accounts_in_message":[],"_revision_number":1},{"id":"497e2fce39ca8c824e09311e1c07367b3ed4919a","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2021-05-11 22:21:56.000000000","message":"Patch Set 1: Code-Review-1","accounts_in_message":[],"_revision_number":1},{"id":"4ada97b13fb6782613f203d8e0c3893ad4700275","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-11 22:22:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e594a39b7fb746819acf6b4e73b60bba : SUCCESS in 4m 10s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2c775ee3aad54624ba854e1ed14ecd49 : SUCCESS in 4m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b6fff0c611154388b333390512f2152b : SUCCESS in 4m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/63ef2f08be82443d84ef858be87f21a4 : SUCCESS in 5m 34s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/01d13b2f08b64361a10a2fcc177745ce : SUCCESS in 5m 34s","accounts_in_message":[],"_revision_number":1},{"id":"ef113156ab1bf3990914348cdf2132e3e6f26815","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2021-05-12 13:25:41.000000000","message":"Abandoned\n\nNot needed, as we do not have docs job in stable/train.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f986ac4b2ccfeafd386451116eecaf5209d1eaf9","revisions":{"f986ac4b2ccfeafd386451116eecaf5209d1eaf9":{"kind":"REWORK","_number":1,"created":"2021-05-11 22:14:17.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/57/790857/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/57/790857/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/57/790857/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/57/790857/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/57/790857/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/57/790857/1"}}},"commit":{"parents":[{"commit":"2fa0fb17960ca64eb132ddbe23fdd56191f874a5","subject":"More robustly handle venv ovs/ovn paths","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/2fa0fb17960ca64eb132ddbe23fdd56191f874a5"}]}],"author":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2021-05-11 21:55:24.000000000","tz":-240},"committer":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2021-05-11 22:13:28.000000000","tz":-240},"subject":"Fix docs job","message":"Fix docs job\n\nSphinx 4.0.0 added a new dependency [0] which is causing the job to fail\nat the moment.\n\nThis patch fix the problem by adding UC to the docs jobs.\n\n[0] https://www.sphinx-doc.org/en/master/changes.html (LaTeX: add\ntex-gyre font dependency)\n\nConflicts:\ntox.ini\n\nChange-Id: I501659d48b1cf2d64cea8dc80ade7a068e37e904\nSigned-off-by: Flavio Fernandes \u003cflaviof@redhat.com\u003e\n(cherry picked from commit 007c0f646b7535616adc19adbbb4e3168f5efa5d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/f986ac4b2ccfeafd386451116eecaf5209d1eaf9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/f986ac4b2ccfeafd386451116eecaf5209d1eaf9"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
