)]}'
{"id":"openstack%2Fovsdbapp~791589","triplet_id":"openstack%2Fovsdbapp~stable%2Fwallaby~Ib757f182a4c7fdf25ed9374963ac5210e4d35523","project":"openstack/ovsdbapp","branch":"stable/wallaby","topic":"bug/1907836-stable/wallaby","hashtags":[],"change_id":"Ib757f182a4c7fdf25ed9374963ac5210e4d35523","subject":"Don\u0027t spam retries 100s of times a second","status":"MERGED","created":"2021-05-17 13:40:55.000000000","updated":"2021-05-18 16:46:54.000000000","submitted":"2021-05-18 16:45:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"791589-bug/1907836-stable/wallaby","cherry_pick_of_change":791309,"cherry_pick_of_patch_set":4,"meta_rev_id":"80c414ab4d078b9d243791a8f3a92b84390863a3","_number":791589,"virtual_id_number":791589,"owner":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-18 16:45:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2021-05-18 15:38:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2021-05-17 17:02:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2021-05-18 15:38:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-17 13:40:55.000000000","updated_by":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"reviewer":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"state":"REVIEWER"},{"updated":"2021-05-17 14:51:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-17 15:09:25.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"}],"messages":[{"id":"88cd9f1fbd0600d2be76c8b2796d9ef7a33a129d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2021-05-17 13:40:55.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"da2ea79dea455862ec3e88d720e1ccbb83114fb7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-17 14:51:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/51103fb919a5450585d3d5864ca8d41e : SUCCESS in 4m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bcebc14f1d5f4785b06ec9d793b2d8d4 : SUCCESS in 4m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5c61ff87380446fe9251c89ada5151d4 : SUCCESS in 4m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8947c81a554f49e78843dfb84aac7004 : SUCCESS in 4m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8cd8d1570e8b4ab1998fa6c17eb459fd : SUCCESS in 3m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7744cc6cac5343cf8e4e23121bbde6e7 : SUCCESS in 6m 06s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/c66eae8ff94149fd9c139ecaaf75ab3d : SUCCESS in 6m 00s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/6f6ed1c350834e16ba73f033a3330990 : SUCCESS in 1h 05m 40s","accounts_in_message":[],"_revision_number":1},{"id":"a03740ff907d75e9523024b0db77330f1be29e01","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-05-17 15:09:25.000000000","message":"Patch Set 1: Code-Review-1\n\nThis is a backport and you should provide the hash reference from the master branch","accounts_in_message":[],"_revision_number":1},{"id":"efdaae445f9529754d53700dadc25afd65b082df","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2021-05-17 15:46:12.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1: Code-Review-1\n\u003e \n\u003e This is a backport and you should provide the hash reference from the master branch\n\nalways the problem with clicking cherry-pick before merge :)","accounts_in_message":[],"_revision_number":1},{"id":"295065abc1b8448e5eda9a41533a6520ff45d3c3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2021-05-17 17:01:43.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"fc16d9c78f713e712ae489e0f14539bfe1ccd677","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2021-05-17 17:02:15.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"0afb8baa825288f0fb1f9f33c7b992c450a78744","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-17 18:06:02.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b8ba9f67127c499ea6f26280b7f57e19 : SUCCESS in 4m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e84e93c0cc14b4088645d550507b8c1 : SUCCESS in 3m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dcfba6c5c41f4c50a39d186550b3049b : SUCCESS in 5m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c71507a363d34f9488ca5511e1cc83a7 : SUCCESS in 4m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b1f4b0b494ee4134a8119ddffaa0bb69 : SUCCESS in 4m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b52c7bfc440340169d2385490257b882 : SUCCESS in 6m 19s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/e1153164594d4bb3af543341b429c68e : SUCCESS in 5m 35s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/08baadf7d3f044f48301cfc69327f6e1 : SUCCESS in 1h 03m 15s","accounts_in_message":[],"_revision_number":2},{"id":"fa768acd325cb255bf71b9e3245bb2813d4f1548","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-05-18 15:38:39.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"353ef2dd0d23788b1813442d395e428626440624","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-18 15:39:04.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"4bc8d00c81f44416450f1004da930623db3dc14c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-18 16:45:15.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fafd80db11ed4aadaaebbd018b95626c : SUCCESS in 4m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5666270bcb10462da942292b7966b04a : SUCCESS in 3m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7da2f4b7e5c048f5b1e827c92a775327 : SUCCESS in 4m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b243ae435a8246a4897dcb464852c2dc : SUCCESS in 4m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec105053bb1a42399254a99f99e30ab6 : SUCCESS in 5m 13s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/fd8d7f867fe14327b4ca6b18b4229fb8 : SUCCESS in 6m 24s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/66e964438ce143678a1afde21640d5af : SUCCESS in 1h 00m 19s","accounts_in_message":[],"_revision_number":2},{"id":"45402058bb7ba92e0c9b95d917c01d34782e7c11","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-18 16:45:16.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"80c414ab4d078b9d243791a8f3a92b84390863a3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-18 16:46:54.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e5e1887fc2041ccbc81ff1c39ac610e : SUCCESS in 1m 04s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"66918f977ba5f3d55b4a1b797e784dc98867ec07","revisions":{"fe7a4712d1cc401de99588f9033c84b308635b81":{"kind":"REWORK","_number":1,"created":"2021-05-17 13:40:55.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/89/791589/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/89/791589/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/89/791589/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/89/791589/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/89/791589/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/89/791589/1"}}},"commit":{"parents":[{"commit":"d0cc49248da613c5cb4a485f077ea3ab9d6cbbee","subject":"Add an active wait in the \"Backend.lookup\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/d0cc49248da613c5cb4a485f077ea3ab9d6cbbee"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2021-05-13 19:14:20.000000000","tz":-300},"committer":{"name":"Flavio Fernandes","email":"flavio@flaviof.com","date":"2021-05-17 13:40:55.000000000","tz":0},"subject":"Don\u0027t spam retries 100s of times a second","message":"Don\u0027t spam retries 100s of times a second\n\nWe can get a TRY_AGAIN because the DB is down or because the\na command is doing an update that will never pass the verify()\ncondition. If this happens, we shouldn\u0027t send 100s of requests per\nsecond until we eventually time out or things come back up, so add\na simple exponential backoff with reasonable max retry sleep of 8s.\n\nCloses-Bug: #1907836\nChange-Id: Ib757f182a4c7fdf25ed9374963ac5210e4d35523\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/fe7a4712d1cc401de99588f9033c84b308635b81"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/fe7a4712d1cc401de99588f9033c84b308635b81"}]},"branch":"refs/heads/stable/wallaby"},"66918f977ba5f3d55b4a1b797e784dc98867ec07":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-05-17 17:01:43.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/89/791589/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/89/791589/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/89/791589/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/89/791589/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/89/791589/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/89/791589/2"}}},"commit":{"parents":[{"commit":"d0cc49248da613c5cb4a485f077ea3ab9d6cbbee","subject":"Add an active wait in the \"Backend.lookup\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/d0cc49248da613c5cb4a485f077ea3ab9d6cbbee"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2021-05-13 19:14:20.000000000","tz":-300},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2021-05-17 17:01:43.000000000","tz":0},"subject":"Don\u0027t spam retries 100s of times a second","message":"Don\u0027t spam retries 100s of times a second\n\nWe can get a TRY_AGAIN because the DB is down or because the\na command is doing an update that will never pass the verify()\ncondition. If this happens, we shouldn\u0027t send 100s of requests per\nsecond until we eventually time out or things come back up, so add\na simple exponential backoff with reasonable max retry sleep of 8s.\n\nCloses-Bug: #1907836\nChange-Id: Ib757f182a4c7fdf25ed9374963ac5210e4d35523\n(cherry picked form commit 6596164f51217cc7fabf302ce14ccc9d9beaff1f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/66918f977ba5f3d55b4a1b797e784dc98867ec07"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/66918f977ba5f3d55b4a1b797e784dc98867ec07"}]},"branch":"refs/heads/stable/wallaby","description":"Edit commit message"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
