)]}'
{"id":"openstack%2Fovsdbapp~791594","triplet_id":"openstack%2Fovsdbapp~stable%2Ftrain~Ib757f182a4c7fdf25ed9374963ac5210e4d35523","project":"openstack/ovsdbapp","branch":"stable/train","topic":"bug/1907836-stable/train","hashtags":[],"change_id":"Ib757f182a4c7fdf25ed9374963ac5210e4d35523","subject":"Don\u0027t spam retries 100s of times a second","status":"MERGED","created":"2021-05-17 14:01:19.000000000","updated":"2021-05-26 15:45:04.000000000","submitted":"2021-05-26 15:45:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"791594-bug/1907836-stable/train","cherry_pick_of_change":791309,"cherry_pick_of_patch_set":4,"meta_rev_id":"9ca80ef7cbb7c459da23f44c9c8911ace4961074","_number":791594,"virtual_id_number":791594,"owner":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-26 15:45:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2021-05-18 15:39:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2021-05-26 13:29:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2021-05-26 13:29:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-17 14:01:19.000000000","updated_by":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"reviewer":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"state":"REVIEWER"},{"updated":"2021-05-17 16:31:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-18 15:39:39.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2021-05-26 13:29:23.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"8e7152e33ac8bb554443117a57271e157f10a205","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2021-05-17 14:01:19.000000000","message":"Patch Set 1: Cherry Picked from branch master.\n\nThe following files contain Git conflicts:\n* ovsdbapp/backend/ovs_idl/transaction.py","accounts_in_message":[],"_revision_number":1},{"id":"bc7f1c3ed82ac944ff0d28e34f16c885d4d7bf72","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2021-05-17 14:14:56.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"0245ab050a17ed197725ee4c000ffb9005969a32","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2021-05-17 14:19:18.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"4ed43f1e92bff3af9545cbaccf6df7d3536683b8","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2021-05-17 14:20:01.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":3},{"id":"b23405b711920da35c2998513520e50c151c07ee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-17 16:31:02.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a4f7d723390349ca93e0a107e58822a4 : SUCCESS in 5m 19s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f8b1e9a872b640c8b57100b4d9249a7c : SUCCESS in 4m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d994ba08753049659e7b0747da3f89fb : SUCCESS in 5m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/51bcf4e7e5c04ade9f2f4ba70eeb5ea5 : SUCCESS in 4m 52s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/a0626658748940ed8a7d92ef9dd4446b : SUCCESS in 7m 55s\n- ovsdbapp-tempest-dsvm-networking-ovn-ovs-release https://zuul.opendev.org/t/openstack/build/75ada6034b0d4652ab9fcef9250255b9 : SUCCESS in 2h 10m 04s","accounts_in_message":[],"_revision_number":3},{"id":"9d3df254585f77ee60a0c42dce3eb09d290f476b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2021-05-17 18:49:51.000000000","message":"Uploaded patch set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"0fb0d07db382fa9b45d81b59d0253f708f8aa410","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-17 20:49:13.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/906e818ce0c54b71b654269d2e1e11f6 : SUCCESS in 3m 56s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7654cdc3d857458aa7ece09e2be46a5d : SUCCESS in 3m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a4246ac4c3e743288a85f1045a0222ea : SUCCESS in 4m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/067418d63a504057a73e566568430a5d : SUCCESS in 3m 59s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/9ee1d33318534273886fc40b0fddaae6 : SUCCESS in 6m 39s\n- ovsdbapp-tempest-dsvm-networking-ovn-ovs-release https://zuul.opendev.org/t/openstack/build/6dd36a151d2342549b12c2839f6cff6f : SUCCESS in 1h 58m 20s","accounts_in_message":[],"_revision_number":4},{"id":"9feb77e029b586d332f191325af39b5b6d61e3c2","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-05-18 15:39:39.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"6b420652b9d0359ccca59a0f6f136ac4f5b8dccb","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-05-26 13:29:23.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"2a24bec105da1caeb27343647c2f3f3202792572","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-26 13:29:44.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"b1a7cd95c2e0e2f11ed8409109b5bd343a2db04c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-26 15:45:04.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/44c771df1cb540cbad35dbd376f81d02 : SUCCESS in 5m 08s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f5fbf2b0b68248b8bc31ffbe1acee873 : SUCCESS in 4m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6e9927777ccd4a079634192cbc6cb8b2 : SUCCESS in 4m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1194fd4ce5bc408298c49ba6e920d1fa : SUCCESS in 4m 34s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/96f69bdcbe8649379e07b564311c610f : SUCCESS in 6m 44s\n- ovsdbapp-tempest-dsvm-networking-ovn-ovs-release https://zuul.opendev.org/t/openstack/build/311be2b8afff49398e865c7881491c6e : SUCCESS in 2h 04m 56s","accounts_in_message":[],"_revision_number":4},{"id":"9ca80ef7cbb7c459da23f44c9c8911ace4961074","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-26 15:45:04.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"0dd8208f03aaabf8ce093de7322d408be293f4b4","revisions":{"e24a01f162bc7e9ec5569f0e66adf92935997487":{"kind":"REWORK","_number":1,"created":"2021-05-17 14:01:19.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/94/791594/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/94/791594/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/94/791594/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/94/791594/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/94/791594/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/94/791594/1"}}},"commit":{"parents":[{"commit":"b193d191c2accaedac188641ea1256eccb79429a","subject":"Add an active wait in the \"Backend.lookup\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/b193d191c2accaedac188641ea1256eccb79429a"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2021-05-13 19:14:20.000000000","tz":-300},"committer":{"name":"Flavio Fernandes","email":"flavio@flaviof.com","date":"2021-05-17 14:01:19.000000000","tz":0},"subject":"Don\u0027t spam retries 100s of times a second","message":"Don\u0027t spam retries 100s of times a second\n\nWe can get a TRY_AGAIN because the DB is down or because the\na command is doing an update that will never pass the verify()\ncondition. If this happens, we shouldn\u0027t send 100s of requests per\nsecond until we eventually time out or things come back up, so add\na simple exponential backoff with reasonable max retry sleep of 8s.\n\nCloses-Bug: #1907836\nChange-Id: Ib757f182a4c7fdf25ed9374963ac5210e4d35523\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/e24a01f162bc7e9ec5569f0e66adf92935997487"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/e24a01f162bc7e9ec5569f0e66adf92935997487"}]},"branch":"refs/heads/stable/train"},"a83f62cb87a91bbbbb4c363051f9174063a28563":{"kind":"REWORK","_number":2,"created":"2021-05-17 14:14:56.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/94/791594/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/94/791594/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/94/791594/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/94/791594/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/94/791594/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/94/791594/2"}}},"commit":{"parents":[{"commit":"b193d191c2accaedac188641ea1256eccb79429a","subject":"Add an active wait in the \"Backend.lookup\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/b193d191c2accaedac188641ea1256eccb79429a"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2021-05-13 19:14:20.000000000","tz":-300},"committer":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2021-05-17 14:14:01.000000000","tz":-240},"subject":"Don\u0027t spam retries 100s of times a second","message":"Don\u0027t spam retries 100s of times a second\n\nWe can get a TRY_AGAIN because the DB is down or because the\na command is doing an update that will never pass the verify()\ncondition. If this happens, we shouldn\u0027t send 100s of requests per\nsecond until we eventually time out or things come back up, so add\na simple exponential backoff with reasonable max retry sleep of 8s.\n\nCloses-Bug: #1907836\nChange-Id: Ib757f182a4c7fdf25ed9374963ac5210e4d35523\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/a83f62cb87a91bbbbb4c363051f9174063a28563"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/a83f62cb87a91bbbbb4c363051f9174063a28563"}]},"branch":"refs/heads/stable/train"},"148ec63d3f39abca85562731ca125ba7631a5a34":{"kind":"REWORK","_number":3,"created":"2021-05-17 14:19:18.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/94/791594/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/94/791594/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/94/791594/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/94/791594/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/94/791594/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/94/791594/3"}}},"commit":{"parents":[{"commit":"d472c8f5a79b254e700cb49c3118d1a5a5ad7448","subject":"Do not try to execute transaction if command list is empty","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/d472c8f5a79b254e700cb49c3118d1a5a5ad7448"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2021-05-13 19:14:20.000000000","tz":-300},"committer":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2021-05-17 14:19:00.000000000","tz":-240},"subject":"Don\u0027t spam retries 100s of times a second","message":"Don\u0027t spam retries 100s of times a second\n\nWe can get a TRY_AGAIN because the DB is down or because the\na command is doing an update that will never pass the verify()\ncondition. If this happens, we shouldn\u0027t send 100s of requests per\nsecond until we eventually time out or things come back up, so add\na simple exponential backoff with reasonable max retry sleep of 8s.\n\nCloses-Bug: #1907836\nChange-Id: Ib757f182a4c7fdf25ed9374963ac5210e4d35523\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/148ec63d3f39abca85562731ca125ba7631a5a34"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/148ec63d3f39abca85562731ca125ba7631a5a34"}]},"branch":"refs/heads/stable/train"},"0dd8208f03aaabf8ce093de7322d408be293f4b4":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2021-05-17 18:49:51.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/94/791594/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/94/791594/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/94/791594/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/94/791594/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/94/791594/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/94/791594/4"}}},"commit":{"parents":[{"commit":"d472c8f5a79b254e700cb49c3118d1a5a5ad7448","subject":"Do not try to execute transaction if command list is empty","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/d472c8f5a79b254e700cb49c3118d1a5a5ad7448"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2021-05-13 19:14:20.000000000","tz":-300},"committer":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2021-05-17 18:49:33.000000000","tz":-240},"subject":"Don\u0027t spam retries 100s of times a second","message":"Don\u0027t spam retries 100s of times a second\n\nWe can get a TRY_AGAIN because the DB is down or because the\na command is doing an update that will never pass the verify()\ncondition. If this happens, we shouldn\u0027t send 100s of requests per\nsecond until we eventually time out or things come back up, so add\na simple exponential backoff with reasonable max retry sleep of 8s.\n\nCloses-Bug: #1907836\nChange-Id: Ib757f182a4c7fdf25ed9374963ac5210e4d35523\n(cherry picked form commit 6596164f51217cc7fabf302ce14ccc9d9beaff1f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/0dd8208f03aaabf8ce093de7322d408be293f4b4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/0dd8208f03aaabf8ce093de7322d408be293f4b4"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
