)]}'
{"id":"openstack%2Fovsdbapp~791595","triplet_id":"openstack%2Fovsdbapp~stable%2Ftrain~I99181078af46540e2e58feeea161509d62e22045","project":"openstack/ovsdbapp","branch":"stable/train","hashtags":[],"change_id":"I99181078af46540e2e58feeea161509d62e22045","subject":"Do not try to execute transaction if command list is empty","status":"MERGED","created":"2021-05-17 14:09:11.000000000","updated":"2021-05-19 10:40:52.000000000","submitted":"2021-05-19 10:40:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"791595","cherry_pick_of_change":711632,"cherry_pick_of_patch_set":1,"meta_rev_id":"bebd4ea483ad96437475d550123d33942181c1cd","_number":791595,"virtual_id_number":791595,"owner":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-19 10:40:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2021-05-17 15:54:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2021-05-17 18:59:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2021-05-17 18:59:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-17 14:10:22.000000000","updated_by":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2021-05-17 14:10:22.000000000","updated_by":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"reviewer":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"state":"REVIEWER"},{"updated":"2021-05-17 16:03:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-17 18:59:54.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"5d192a53b410f3b0a9cbb79fb87d50e92eac321f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2021-05-17 14:09:11.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"b2c7f59369c24b8c4d1d747eb4dcc2ba376129ba","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-05-17 15:54:38.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"25e7681b3a8ce5dca07575f1dd32d66484809aeb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-17 16:03:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/93f6b5df166242d68f7e1d16fc627911 : SUCCESS in 6m 20s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9b7c766c123d4860b76c989d64535cb1 : SUCCESS in 6m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9235dc828ad844789c7d7bda5c7c820c : SUCCESS in 6m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/47e3de8e137c49cbad801c0ebe5198e7 : SUCCESS in 6m 43s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/fdbf35b0e1bf45dda136b4c943bbf262 : SUCCESS in 8m 02s\n- ovsdbapp-tempest-dsvm-networking-ovn-ovs-release https://zuul.opendev.org/t/openstack/build/556e6056ae374a88bba0f47354b563a7 : SUCCESS in 1h 52m 10s","accounts_in_message":[],"_revision_number":1},{"id":"b61feeba659f764d7d4cbf22e1b9243b19c10d88","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-05-17 18:59:54.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f66ea3bd8a5335dabfd55e1ccf47176b80b9331b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-17 19:00:18.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3f55247656d677652c14d2761bd48a9f9614d829","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-17 22:12:01.000000000","message":"Patch Set 1:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9e35574515ccfd4dec86082852cc968d50c83a0d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-18 01:16:01.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/825898dce6684f3b9d870c8f23f65a3b : SUCCESS in 6m 00s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/19358e4d12e742a4a1c0eb2a55343c69 : SUCCESS in 5m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fb914c853c1340e198177740242f0d34 : SUCCESS in 5m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e5f78ad72ae64eb7b506b85d74384ead : SUCCESS in 5m 56s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/4328f59a4b06437bb76b4d10fb6d1015 : SUCCESS in 7m 55s\n- ovsdbapp-tempest-dsvm-networking-ovn-ovs-release https://zuul.opendev.org/t/openstack/build/5b2f5db8daa2452ba5f4172e06c04074 : FAILURE in 2h 57m 55s","accounts_in_message":[],"_revision_number":1},{"id":"4bc0dd747902c6868b20b68b49dc7a106699400c","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-05-18 15:38:12.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"91872553f9359de5106b02ffc7cbc87145c1ad53","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-18 18:16:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0b7bb84eec8e44a69fba9fe2ffa9c9a2 : SUCCESS in 4m 17s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/853667d208c24003a8102a15a6422473 : SUCCESS in 4m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a9399b62e1d34eec937fd9d1690de8b3 : SUCCESS in 4m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4e9786f9261f4944bb946d961ec38005 : SUCCESS in 4m 52s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/c759942466814f949042c7932c0c63c2 : SUCCESS in 7m 25s\n- ovsdbapp-tempest-dsvm-networking-ovn-ovs-release https://zuul.opendev.org/t/openstack/build/b9222b46b6ff4506a309b6d53ee1393f : SUCCESS in 2h 35m 55s","accounts_in_message":[],"_revision_number":1},{"id":"03e334fc92326037e7c6c86b5d552120bde0bfca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-18 18:16:20.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f534851bd1cf26a97ff515ef96f69aa951c31947","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-18 21:27:23.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/57b45f0f1d5741e2a09498e83793db40 : SUCCESS in 4m 41s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c7584607120f4a09befab2301bd8ca33 : SUCCESS in 4m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/db9dfa2aeff34b7192412caf91674c5c : SUCCESS in 4m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/81b378d514224a8a8d222cbbcaaad8d5 : SUCCESS in 4m 26s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/03cf752757ee43a691bd03b746305dbe : SUCCESS in 7m 50s\n- ovsdbapp-tempest-dsvm-networking-ovn-ovs-release https://zuul.opendev.org/t/openstack/build/db75318bf5934853b6e21e12f8f6f182 : FAILURE in 3h 05m 13s","accounts_in_message":[],"_revision_number":1},{"id":"3cc7b2c4e8160ab978519cca8a45310300f633c4","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-05-18 21:39:32.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"b1a1d4f8717778f09f2ecf547734dc2ab7d6b736","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-18 23:49:45.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/707b12a3f9d449f9b48adcaf5094486b : SUCCESS in 5m 01s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1f2bae7c572f449c95c221e0dbb47803 : SUCCESS in 3m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d264a84aaa0c47d2ba5b25318e3475b5 : SUCCESS in 4m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/243f19f3f6314d9baba6516bb55a55ee : SUCCESS in 5m 33s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/6e1fe2bcf5e343f38c14c6fc4aac4797 : SUCCESS in 6m 19s\n- ovsdbapp-tempest-dsvm-networking-ovn-ovs-release https://zuul.opendev.org/t/openstack/build/4d1a7d3b5a53480b9e825638dab23bb7 : FAILURE in 2h 08m 53s","accounts_in_message":[],"_revision_number":1},{"id":"63b8fc7f345d4809d3c6c7677fa4167772441f78","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-05-19 00:10:04.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"67f9006cdce90a2ba4f23ffbe693bac0a9409576","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-19 02:32:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d1b489e3c30d478892c4aa21695f9421 : SUCCESS in 3m 48s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/909507a47d174bdcad4d447dee0d8864 : SUCCESS in 3m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e0d63d2bd0484cdfa0cc1f6a0ef994a5 : SUCCESS in 3m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/927e332d4b36434c9282bdf77e05aeb4 : SUCCESS in 5m 00s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/bdc1421bef5d43f0ba326ee69b0d141e : SUCCESS in 5m 44s\n- ovsdbapp-tempest-dsvm-networking-ovn-ovs-release https://zuul.opendev.org/t/openstack/build/b2deaab1b0df42a5a034db401102bbea : SUCCESS in 2h 18m 15s","accounts_in_message":[],"_revision_number":1},{"id":"1fbc00336adb2920f15f67c2010de4e9a77c047d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-19 02:32:30.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6f401ad5d60beba4e9c3104bad08ede1fc3b0c1c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-19 05:31:33.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/923e08000a2d4baeb5801bd0599ec384 : SUCCESS in 4m 06s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/39aec251aefc466cb00dc54ee2e47b9d : SUCCESS in 3m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/00271fb9dce844b4837af34784c67f3d : SUCCESS in 4m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/41941558332247b2bcb487eeb3216de0 : SUCCESS in 4m 13s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/b4321d3aa4c24c749b487dc634506843 : SUCCESS in 6m 48s\n- ovsdbapp-tempest-dsvm-networking-ovn-ovs-release https://zuul.opendev.org/t/openstack/build/8b07b6703b704c8987ea4c89833be67d : FAILURE in 2h 55m 14s","accounts_in_message":[],"_revision_number":1},{"id":"26b0d103a29b9cf4bb18204890fcdb355bc976d9","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-05-19 05:41:18.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"65cc54d35c21310a05cc7b229676fd199fe47d78","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-19 07:28:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bc579994ced14537a9840389ddcebe6b : SUCCESS in 3m 52s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3e2ba1f8bdf34960b2ae02e254b9f8b1 : SUCCESS in 4m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8585e42c67f8499b9f5815de035fef90 : SUCCESS in 3m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8e35a842ee9443dc8bd000ca6d83054c : SUCCESS in 3m 55s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/92c02135f4614565a139e5c84d2a8ba4 : SUCCESS in 6m 29s\n- ovsdbapp-tempest-dsvm-networking-ovn-ovs-release https://zuul.opendev.org/t/openstack/build/12cb04bca6974d3d9637dcdf414a572e : SUCCESS in 1h 45m 13s","accounts_in_message":[],"_revision_number":1},{"id":"276da06add9fc716362d84ac5ce12579e2ce98c3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-19 07:28:50.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"08e9c4309b3c6acecc93eb7d6f0f485b5bd06f44","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-19 10:40:51.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/694c4f50adf84b968a58f7b6981fb0c1 : SUCCESS in 5m 16s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/149fa496f1104fb4aca7a5dae9741445 : SUCCESS in 4m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9eb68790d9754b7caca84e267fa8fb7d : SUCCESS in 5m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fff01b3ffbc341e98061a818d8dd918c : SUCCESS in 6m 15s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/f0ee49af34df4d298a88409878647310 : SUCCESS in 7m 29s\n- ovsdbapp-tempest-dsvm-networking-ovn-ovs-release https://zuul.opendev.org/t/openstack/build/86f1dfde68f7496a8e32750a82d3da60 : SUCCESS in 2h 51m 56s","accounts_in_message":[],"_revision_number":1},{"id":"bebd4ea483ad96437475d550123d33942181c1cd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-19 10:40:52.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d472c8f5a79b254e700cb49c3118d1a5a5ad7448","revisions":{"d472c8f5a79b254e700cb49c3118d1a5a5ad7448":{"kind":"REWORK","_number":1,"created":"2021-05-17 14:09:11.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/95/791595/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/95/791595/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/95/791595/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/95/791595/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/95/791595/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/95/791595/1"}}},"commit":{"parents":[{"commit":"b193d191c2accaedac188641ea1256eccb79429a","subject":"Add an active wait in the \"Backend.lookup\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/b193d191c2accaedac188641ea1256eccb79429a"}]}],"author":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2020-03-06 14:43:38.000000000","tz":0},"committer":{"name":"Flavio Fernandes","email":"flavio@flaviof.com","date":"2021-05-17 14:09:11.000000000","tz":0},"subject":"Do not try to execute transaction if command list is empty","message":"Do not try to execute transaction if command list is empty\n\nIf command list is empty we can just early exit.\n\nChange-Id: I99181078af46540e2e58feeea161509d62e22045\n(cherry picked from commit 8275af1726cb2079afd8f5230377e064221ebcf3)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/d472c8f5a79b254e700cb49c3118d1a5a5ad7448"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/d472c8f5a79b254e700cb49c3118d1a5a5ad7448"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
