)]}'
{"id":"openstack%2Fovsdbapp~802359","triplet_id":"openstack%2Fovsdbapp~master~I1623ead29a48eef70ce9e1efc516ef3d461d093d","project":"openstack/ovsdbapp","branch":"master","hashtags":[],"change_id":"I1623ead29a48eef70ce9e1efc516ef3d461d093d","subject":"Make exceptions in the event logs easier to read","status":"MERGED","created":"2021-07-26 16:44:46.000000000","updated":"2021-07-26 23:09:01.000000000","submitted":"2021-07-26 23:07:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"802359","meta_rev_id":"908293c0cb3bdb49edd02f05dcd1ca3ef0bd92a5","_number":802359,"virtual_id_number":802359,"owner":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-26 23:07:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2021-07-26 19:36:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2021-07-26 17:31:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2021-07-26 21:40:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-26 17:31:15.000000000","updated_by":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"reviewer":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"state":"REVIEWER"},{"updated":"2021-07-26 18:42:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-26 19:36:46.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"}],"messages":[{"id":"5c3b2a3a8d494d4f6547786fb9efa549a0d81ba5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"date":"2021-07-26 16:44:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"306127bbf127f7dd2dbd7559646ffac5bea10c3c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"date":"2021-07-26 16:49:52.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"41819b702b152b94b17872c708165e48b3ea84b1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"date":"2021-07-26 17:12:14.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"68b66ac46ee7b61a13f9f87effbecf90287cb5b4","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2021-07-26 17:31:15.000000000","message":"Patch Set 3: Code-Review+2\n\nThanks!","accounts_in_message":[],"_revision_number":3},{"id":"91e04894a7a197a2e1327df15ab94613e50f67ba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-26 18:42:18.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1c8e89be61eb43af9ccf1ae03518c8c1 : SUCCESS in 4m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/47daf1f894f44371a08efa2d2625260e : SUCCESS in 4m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6989587ffd6f412ba4babd80072c94ef : SUCCESS in 4m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/07426d49690647cba8ef0c9bc73f50f3 : SUCCESS in 4m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0491fd09b4584398ae739f58aa8fc96d : SUCCESS in 4m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7da8cf637ad74c98a5b70e260b1b8d00 : SUCCESS in 5m 38s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/1f2584e6b50d41e4949996e37e5e1aa5 : SUCCESS in 7m 55s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/fbac0a458b674329b2bae815ed876a9b : SUCCESS in 1h 27m 08s","accounts_in_message":[],"_revision_number":3},{"id":"7ffa9c65d88e51f88db7c74665100c3c1d1421ca","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-07-26 19:36:46.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"1e0d8323d4c2452d8f6a1a339e0f12e1519b6a0e","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2021-07-26 21:40:35.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"5c307de8d9c434ae851ce404fab4ad4d84150ff3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-26 21:41:01.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"5bc01a462cecab9e82036d4d17e0cfc79b7e62ab","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-26 23:07:46.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3587587d9e5c4e57acc1e73e671268b7 : SUCCESS in 4m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f8dfe5ec36af41bbad85119dc8c5449b : SUCCESS in 3m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1eeb8e7913824668ae463e3a3e0cd4b8 : SUCCESS in 4m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/31a0c17a0f25427e973221280f178248 : SUCCESS in 3m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e199d5c530c74997aeaca736d670017e : SUCCESS in 5m 12s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/1c129383dd354d669da235da0400126c : SUCCESS in 5m 41s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/310376bc553546cc950c72258b19df71 : SUCCESS in 1h 24m 46s","accounts_in_message":[],"_revision_number":3},{"id":"b36a1d8d1588a34854bda263a0227f74da042824","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-26 23:07:46.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"908293c0cb3bdb49edd02f05dcd1ca3ef0bd92a5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-26 23:09:01.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93bdd5c40b3b4fecab096d03a97b8902 : SUCCESS in 57s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"f02b1c20c062cca190d9e39bbf8612f1663bf604","revisions":{"c5ba89006fc077e900b6f548e6cc8bc1f76664c9":{"kind":"REWORK","_number":1,"created":"2021-07-26 16:44:46.000000000","uploader":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"ref":"refs/changes/59/802359/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/59/802359/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/59/802359/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/59/802359/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/59/802359/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/59/802359/1"}}},"commit":{"parents":[{"commit":"5d24facd36ca06b1f5f793c834da2efa4fa3a508","subject":"Merge \"Actually close the connection in Connection.stop()\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/5d24facd36ca06b1f5f793c834da2efa4fa3a508"}]}],"author":{"name":"Elvira García","email":"egarciar@redhat.com","date":"2021-07-26 16:41:34.000000000","tz":120},"committer":{"name":"Elvira García","email":"egarciar@redhat.com","date":"2021-07-26 16:44:44.000000000","tz":120},"subject":"Make exceptions in the event logs easier to read","message":"Make exceptions in the event logs easier to read\n\nOnly a part of an exception was thrown when an event could not be\ntriggered. With LOG.exception it is possible to read the full traceback\nof the problem now.\n\nSigned-off-by: Elvira García \u003cegarciar@redhat.com\u003e\nChange-Id: I1623ead29a48eef70ce9e1efc516ef3d461d093d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/c5ba89006fc077e900b6f548e6cc8bc1f76664c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/c5ba89006fc077e900b6f548e6cc8bc1f76664c9"}]},"branch":"refs/heads/master"},"9f01395d9125871d379822f6f7faa80aa5af8b0c":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-07-26 16:49:52.000000000","uploader":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"ref":"refs/changes/59/802359/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/59/802359/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/59/802359/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/59/802359/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/59/802359/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/59/802359/2"}}},"commit":{"parents":[{"commit":"5d24facd36ca06b1f5f793c834da2efa4fa3a508","subject":"Merge \"Actually close the connection in Connection.stop()\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/5d24facd36ca06b1f5f793c834da2efa4fa3a508"}]}],"author":{"name":"Elvira García","email":"egarciar@redhat.com","date":"2021-07-26 16:41:34.000000000","tz":120},"committer":{"name":"Elvira García","email":"egarciar@redhat.com","date":"2021-07-26 16:48:22.000000000","tz":120},"subject":"Make exceptions in the event logs easier to read","message":"Make exceptions in the event logs easier to read\n\nOnly a part of an exception was thrown when an event could not be\ntriggered. With LOG.exception it is possible to read the full traceback\nof the problem now.\n\nCloses-bug: #1938038\nSigned-off-by: Elvira García \u003cegarciar@redhat.com\u003e\nChange-Id: I1623ead29a48eef70ce9e1efc516ef3d461d093d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/9f01395d9125871d379822f6f7faa80aa5af8b0c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/9f01395d9125871d379822f6f7faa80aa5af8b0c"}]},"branch":"refs/heads/master"},"f02b1c20c062cca190d9e39bbf8612f1663bf604":{"kind":"REWORK","_number":3,"created":"2021-07-26 17:12:14.000000000","uploader":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"ref":"refs/changes/59/802359/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/59/802359/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/59/802359/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/59/802359/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/59/802359/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/59/802359/3"}}},"commit":{"parents":[{"commit":"5d24facd36ca06b1f5f793c834da2efa4fa3a508","subject":"Merge \"Actually close the connection in Connection.stop()\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/5d24facd36ca06b1f5f793c834da2efa4fa3a508"}]}],"author":{"name":"Elvira García","email":"egarciar@redhat.com","date":"2021-07-26 16:41:34.000000000","tz":120},"committer":{"name":"Elvira García","email":"egarciar@redhat.com","date":"2021-07-26 17:12:00.000000000","tz":120},"subject":"Make exceptions in the event logs easier to read","message":"Make exceptions in the event logs easier to read\n\nOnly a part of an exception was thrown when an event could not be\ntriggered. With LOG.exception it is possible to read the full traceback\nof the problem now.\n\nCloses-bug: #1938038\nSigned-off-by: Elvira García \u003cegarciar@redhat.com\u003e\nChange-Id: I1623ead29a48eef70ce9e1efc516ef3d461d093d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/f02b1c20c062cca190d9e39bbf8612f1663bf604"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/f02b1c20c062cca190d9e39bbf8612f1663bf604"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
