)]}'
{"id":"openstack%2Fovsdbapp~804252","triplet_id":"openstack%2Fovsdbapp~master~I5fa20962f48a49aa5eb137a40861cf2ebc9e42d0","project":"openstack/ovsdbapp","branch":"master","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2022-04-14 19:35:23.000000000","reason":"Change was abandoned"},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2022-02-07 12:15:54.000000000","reason":"Lajos Katona replied on the change"},"5756":{"account":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"last_update":"2022-04-14 19:35:23.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I5fa20962f48a49aa5eb137a40861cf2ebc9e42d0","subject":"Use setkey for DbSetCommand maps","status":"ABANDONED","created":"2021-08-11 14:36:08.000000000","updated":"2022-04-14 19:35:23.000000000","total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"429f1032583b9edcd4a72af02e0d9474fe790f45","_number":804252,"virtual_id_number":804252,"owner":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-02-16 20:19:00.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-11 15:35:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-11 16:02:57.000000000","updated_by":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"reviewer":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"state":"REVIEWER"},{"updated":"2021-08-12 05:41:36.000000000","updated_by":{"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},"reviewer":{"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},"state":"REVIEWER"},{"updated":"2021-08-20 14:07:38.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2021-10-14 16:12:50.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"CC"},{"updated":"2021-12-13 09:12:24.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"CC"}],"messages":[{"id":"762e794668b1855354ba1a4bb1248784572cd95f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2021-08-11 14:36:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"554adcff1ba2c09513a5492869ad02ec2e08703c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-11 15:35:42.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c51972c8a09b48ec919e4feb458c0722 : SUCCESS in 3m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/373f854e0af64a878741ae3d4320ac25 : SUCCESS in 4m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/07956fa58ead467daae54713bb82b08e : SUCCESS in 4m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/daa726a04ee24deb9b975580f69d5307 : SUCCESS in 4m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6b48d621ebc8455ea2e24438fd57b87d : SUCCESS in 4m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e681a22bcdc4898bf81a02e6a739c83 : SUCCESS in 5m 21s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/af15733b1bd245e1a64780a47f73cac8 : SUCCESS in 6m 15s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/45d566c255934c4c87b1b636b6d9a365 : FAILURE in 53m 59s","accounts_in_message":[],"_revision_number":1},{"id":"a723607179c59c7c624842b332a7fca8d6651c4f","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2021-08-11 18:17:29.000000000","message":"Patch Set 1:\n\nrecheck\n\nInteresting. It looks like the tempest logs are showing an issue with neutron:device_id not being set, which could somehow be related since db_set is often used on external_ids. though the place I see it being used is in AddLSwitchPortCommand and SetLSwitchPortComamnd, which unilaterally overwrite external_ids using setattr (though the update_port code copies over the existing external_ids). Neither appear to call db_set(). rechecking to verify.","accounts_in_message":[],"_revision_number":1},{"id":"8e635ddffccc0ca0bc88a06d815e22e8c2c67c4a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-11 19:17:10.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/482e5d0b308b42c285dce1a92e27456f : SUCCESS in 4m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9983f48f6dfe499ba37a2abc2d6a0237 : SUCCESS in 4m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a9aac269bffe47da8555d49715734a5f : SUCCESS in 4m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2937750765ed47af99ef73218d7a7f81 : SUCCESS in 3m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f9d2049c445642ea88a453814034b823 : SUCCESS in 3m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/19ff732c93234a128112c4a1cf96f6fb : SUCCESS in 5m 20s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/375c9274c60e44b9b52284c5c799341c : SUCCESS in 6m 02s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/16d0c0585a5b4e27baacfb61ae2e10c6 : FAILURE in 55m 23s","accounts_in_message":[],"_revision_number":1},{"id":"0b2bfbe1d749584cff1881b4af7acb026546b53a","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2021-09-02 16:05:59.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ba8ce86c04d42963c6cb44715eacf806f5e1eb61","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-02 17:08:28.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0dbbb8cdca31431286bcfb2d38cddc85 : SUCCESS in 5m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e4a1a45b1c6b433c8db483f2dfdcb4be : SUCCESS in 4m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2ff025fd54284790b0e8ffc773f2d224 : SUCCESS in 4m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3edff9aa83df41148cd35e16ec8f6c23 : SUCCESS in 5m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/12104f0db04e4dfaaf51dde4c0fd0abd : SUCCESS in 4m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/936b12756b4a43a3bd0f1307ab1919d2 : SUCCESS in 6m 30s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/110c930088ea43058192cc925214485e : SUCCESS in 7m 59s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/8478987d23474fbd92654983449ab53c : FAILURE in 53m 25s","accounts_in_message":[],"_revision_number":1},{"id":"143e424b0f65fda76caa07b40faf7acc23f0a44f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2021-09-10 21:37:42.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"f336e84a6dcdc036a1fd4f00b34925b8aae80f0b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-10 22:43:57.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f08ab8698cbc4dc5bf46bb29601d3c5b : SUCCESS in 5m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9fb266414bc24c0b861332ff612cca0f : SUCCESS in 5m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6371aa6113584ec0b42da260061b3f7d : SUCCESS in 4m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/57581bc3016f4926b8942a6ec536df2b : SUCCESS in 4m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d9210611c69e4021a3120c6c4a290b1e : SUCCESS in 6m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/173867286cb6459b9432e56a59b7940a : SUCCESS in 5m 39s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/04dc97a84e7844c3b97ddca542b8f098 : SUCCESS in 7m 10s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/974fde1416d8422888262b10020350bb : FAILURE in 1h 01m 34s","accounts_in_message":[],"_revision_number":2},{"id":"a26a10cde4e0c20d5d5d3253d23439d70c4e153f","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-10-14 16:12:50.000000000","message":"Abandoned\n\nThis review is \u003e 4 weeks without comment, and failed Zuul jobs the last time it was checked. We are abandoning this for now. Feel free to reactivate the review by pressing the restore button and leaving a \u0027recheck\u0027 comment to get fresh test results.","accounts_in_message":[],"_revision_number":2},{"id":"4bb9b167fe02c8bcd3f542be9a74baf00ac2d4e3","tag":"autogenerated:gerrit:restore","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-10-14 17:17:43.000000000","message":"Restored","accounts_in_message":[],"_revision_number":2},{"id":"9f737cd79cca0aa1f5ce6d082d327dfbd966c0ac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-14 19:01:02.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0a4584f198b2427a981af1d7f35237d7 : SUCCESS in 4m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e707ad297424973aa8c6b21ff7cfc6a : SUCCESS in 4m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/221f667d2fe44fac911fd24b34ed01e5 : SUCCESS in 4m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/78f04fe7d1b94c3597b7bf5e4c514a91 : SUCCESS in 4m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/12dcf8d8f6cd4c2a84c754b37c2ebd34 : SUCCESS in 6m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06724af3a619494bbc27cd178cdfb0bf : SUCCESS in 7m 56s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/a33028f8e3f74ea29038c63b63cee487 : SUCCESS in 6m 58s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/d9ce86f8ec0e4cf5a6f84fa12fdcc338 : FAILURE in 1h 36m 43s","accounts_in_message":[],"_revision_number":2},{"id":"1e86d6d57c829705661f8931bcfa49fbae8bf295","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2021-11-18 14:53:01.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"97af5eb3ec3db54ab2fabfd4abccd9680c9ba616","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2021-11-18 14:57:11.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"535ad8e772296bf3793b829fab5e67443f3ca303","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-18 16:09:46.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/42a75da078a844258492459993ad6ade : SUCCESS in 5m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/19e17d41ba9c46a1b20d74ab6a6399b7 : SUCCESS in 5m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a1d3fac491a34ca1be8513ee07ee01b1 : SUCCESS in 5m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/487290b483544e67afb0797d7334a37e : SUCCESS in 5m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/091d7d7903944c2aa8fc490d676a981e : SUCCESS in 5m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f1c328caf804563a80dc5979527ea43 : SUCCESS in 7m 17s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/d784b56509774b2eaaf743886ce98f6b : SUCCESS in 6m 57s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/77f6a18e1b00491591b108cb2060ccb6 : FAILURE in 1h 09m 02s","accounts_in_message":[],"_revision_number":3},{"id":"d9c27d733dd95d5486295ab0255bf510c6a6f7c6","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-11-18 16:16:24.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"d039426e179341e93e28f3d4a3456008af22e994","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-18 17:30:17.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/58742a1da7604324b7653cfcef7fc7c7 : SUCCESS in 3m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f9082635b92a4b0eb9446d6017d7cb9f : SUCCESS in 5m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bb6bdfb15ee54ed6a35ac67cf3acf824 : SUCCESS in 4m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4671b2091cba4323bfc39ba0f9f0e264 : SUCCESS in 4m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/eec2c59dd8f347dca3bfe61faa634dbe : SUCCESS in 5m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ea3fe762c4247b7bf344a46ff9f822e : SUCCESS in 6m 23s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/0742196f20384926b6a080d15b05ccae : SUCCESS in 6m 55s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/1716940f81254ab38447e740e86e2d17 : FAILURE in 57m 49s","accounts_in_message":[],"_revision_number":3},{"id":"25d463db68d2a3c31b13f58c8c04d7955c4ae562","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2021-11-18 18:21:31.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"79f7fbf003f2efd81ec1a7555453d2188df8ad82","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-18 19:30:44.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/97931fd3609c4204bfe35c370221378a : FAILURE in 3m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e85496cdc941409ea37a57bedb312a6c : FAILURE in 3m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff865ce9e2ff4835a7d2c8263cf7b769 : SUCCESS in 3m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ec8fd0220ef74ad9a22a23239117a704 : SUCCESS in 3m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c40671ead57043768e45c131ea59ead2 : SUCCESS in 4m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/eeb571d73a3342199c3564eda868b655 : SUCCESS in 4m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e98b80ad52f4c1596219fc23ccf281d : SUCCESS in 5m 26s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/f008f296dea541f59f55c18c3c62c48c : SUCCESS in 7m 01s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/243f3eb30d3947ebb1138eda9b961ffc : FAILURE in 1h 01m 03s","accounts_in_message":[],"_revision_number":4},{"id":"759ec1a01109dcc5437cc1b3df62329a8d811834","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-12-13 09:12:24.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"a991a330f19de42a11e13de48e6440e050063b35","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-12-13 09:16:58.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"167a418a4f1ab668560b8f0ef1b754b5d7397970","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-12-13 09:17:17.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"1eecab637e77abbbbabba83f41b5d708841ac9ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-13 10:08:50.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/a81dfbf0d1f14b93996a29d6666ab3bd : SUCCESS in 2m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5e3bef1847b54839b835d2711d56d4b5 : FAILURE in 3m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d559a87724d64a45b199f21bd60f343b : SUCCESS in 4m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1f2c46cf3870486ea3980a1fc0607352 : SUCCESS in 4m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/67d3175b17fc4cc4be40d1eb66db157d : SUCCESS in 3m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/54309529f67f4a8485596703421a2fdc : SUCCESS in 5m 30s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/dbc5f2c7620543aa8f91c9fa23361f01 : SUCCESS in 6m 37s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/76c3da5628494441956f44c4e8fec777 : FAILURE in 49m 38s","accounts_in_message":[],"_revision_number":5},{"id":"ac814c12bc5d386ac8d7e4ca99e2bdbe39c15ab7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-01-05 14:44:21.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"c49f74558f7a8e76b7f6e8db03c48591263e418c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-05 15:50:22.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/06e7e80d87ab4c97821cfb069dcb71b8 : SUCCESS in 3m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/34310426b9d64bd78d8f53fd0375dc84 : FAILURE in 3m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/45f6258f86f046f08fc04538019d3934 : SUCCESS in 4m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/afb145f462dd4d89b35917705926e232 : SUCCESS in 4m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/34c9254a421647e4a26ff88b93195a76 : SUCCESS in 3m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f263ade955f4aa0997d01f0752ccabd : SUCCESS in 5m 15s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/5332b33602304f5bb2b0be6397287ca3 : SUCCESS in 6m 17s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/6f8b1daad8134d0f8cde90ca35324645 : FAILURE in 59m 11s","accounts_in_message":[],"_revision_number":6},{"id":"49b34e305a47d1f84f08150ca4c388c1b2091c61","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-02-07 12:15:54.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"df2b9b3bf90af21579e8e4d1313b9e74b5c06076","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-07 13:32:29.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/844877d0d8854c34b226cfb14c423b34 : SUCCESS in 2m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/484c62ab9983403288d2e0b671ff3f94 : FAILURE in 3m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/45a684a5f968494a84d4bb2aee5c54ae : SUCCESS in 4m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/274cfdffbf8c4251a99b27345a0e2a2f : SUCCESS in 3m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b82be948b669493d8126a6cec8b2b209 : SUCCESS in 3m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/385ea08bb53e4899ae7d1b4839987773 : SUCCESS in 4m 48s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/18d9b1775baa47c780df2c47f23233cd : SUCCESS in 5m 51s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/981c4ce0f9f143ca8153c79b5152913f : FAILURE in 1h 10m 51s","accounts_in_message":[],"_revision_number":6},{"id":"e6f68ae3c5c616de65424f07e238d8cebc30e7c3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2022-02-16 19:03:40.000000000","message":"Patch Set 7: Patch Set 6 was rebased","accounts_in_message":[],"_revision_number":7},{"id":"6866408e62df896b15f2730b926c1e48ed0e5d98","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-16 20:19:00.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/db8d7746d53b4ec28be988ba3673184d : SUCCESS in 2m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/462091df220c419ea5bd895b6ec9f185 : FAILURE in 3m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a5391b6971ed43498a7e9118801cfdcc : SUCCESS in 3m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/070244d433764b02881c7ba0b31a1f85 : SUCCESS in 3m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/aef97b63160b4081b4e9a5fb449e3719 : SUCCESS in 3m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ae414fd784145fc992ae0d365e24439 : SUCCESS in 4m 32s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/010a5d68df0d41d7ab5e97b430d0b589 : SUCCESS in 6m 17s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/6f11385954ae4d988688125ee01c9fcb : FAILURE in 1h 07m 17s","accounts_in_message":[],"_revision_number":7},{"id":"429f1032583b9edcd4a72af02e0d9474fe790f45","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-04-14 19:35:23.000000000","message":"Abandoned\n\nThis review is \u003e 4 weeks without comment, and failed Zuul jobs the last time it was checked. We are abandoning this for now. Feel free to reactivate the review by pressing the restore button and leaving a \u0027recheck\u0027 comment to get fresh test results.","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"da45d18b1dd498ef753d2ed334fab61d31b92dc9","revisions":{"9bebe628b670b1c879f232d3f0c8e16ca688ad5e":{"kind":"REWORK","_number":1,"created":"2021-08-11 14:36:08.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/52/804252/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/52/804252/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/52/804252/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/52/804252/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/52/804252/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/52/804252/1"}}},"commit":{"parents":[{"commit":"35c8eae020ef5300e08b7869e5da11dd0e2c500e","subject":"Merge \"setup.cfg: Replace dashes with underscores\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/35c8eae020ef5300e08b7869e5da11dd0e2c500e"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2021-08-11 14:30:19.000000000","tz":-300},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2021-08-11 14:30:19.000000000","tz":-300},"subject":"Use setkey for DbSetCommand maps","message":"Use setkey for DbSetCommand maps\n\nsetkey() has been available for a long time now, so we can safely\nuse it and drop the verify() call, which in certain circumstances\ncan lead to a busy loop until the connection timeout passes, e.g.\nml2/ovn\u0027s agent health check timestamp mechanism.\n\nChange-Id: I5fa20962f48a49aa5eb137a40861cf2ebc9e42d0\nRelated-Bug: #1907836\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/9bebe628b670b1c879f232d3f0c8e16ca688ad5e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/9bebe628b670b1c879f232d3f0c8e16ca688ad5e"}]},"branch":"refs/heads/master"},"7123d28b9e44e6297de4d4045642364cd07c84c8":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-09-10 21:37:42.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/52/804252/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/52/804252/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/52/804252/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/52/804252/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/52/804252/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/52/804252/2"}}},"commit":{"parents":[{"commit":"06d4175eab1570857c915fe0c7b27ffc243795a3","subject":"Add support for OVN NB Logical_Router_Policy table","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/06d4175eab1570857c915fe0c7b27ffc243795a3"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2021-08-11 14:30:19.000000000","tz":-300},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2021-09-10 21:37:42.000000000","tz":0},"subject":"Use setkey for DbSetCommand maps","message":"Use setkey for DbSetCommand maps\n\nsetkey() has been available for a long time now, so we can safely\nuse it and drop the verify() call, which in certain circumstances\ncan lead to a busy loop until the connection timeout passes, e.g.\nml2/ovn\u0027s agent health check timestamp mechanism.\n\nChange-Id: I5fa20962f48a49aa5eb137a40861cf2ebc9e42d0\nRelated-Bug: #1907836\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/7123d28b9e44e6297de4d4045642364cd07c84c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/7123d28b9e44e6297de4d4045642364cd07c84c8"}]},"branch":"refs/heads/master","description":"Rebase"},"ecdc2c1089ce8888ce35e5c13e7e5031f7423440":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-11-18 14:53:01.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/52/804252/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/52/804252/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/52/804252/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/52/804252/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/52/804252/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/52/804252/3"}}},"commit":{"parents":[{"commit":"0e4c7a056c8c9ca9e91ab43ebb30f9d37a330d83","subject":"nb: set chassis_name for new gateway chassis","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/0e4c7a056c8c9ca9e91ab43ebb30f9d37a330d83"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2021-08-11 14:30:19.000000000","tz":-300},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2021-11-18 14:53:01.000000000","tz":0},"subject":"Use setkey for DbSetCommand maps","message":"Use setkey for DbSetCommand maps\n\nsetkey() has been available for a long time now, so we can safely\nuse it and drop the verify() call, which in certain circumstances\ncan lead to a busy loop until the connection timeout passes, e.g.\nml2/ovn\u0027s agent health check timestamp mechanism.\n\nChange-Id: I5fa20962f48a49aa5eb137a40861cf2ebc9e42d0\nRelated-Bug: #1907836\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/ecdc2c1089ce8888ce35e5c13e7e5031f7423440"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/ecdc2c1089ce8888ce35e5c13e7e5031f7423440"}]},"branch":"refs/heads/master","description":"Rebase"},"37e99a2569fd060282fb38f1e03f5e852af4e337":{"kind":"REWORK","_number":4,"created":"2021-11-18 18:21:31.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/52/804252/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/52/804252/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/52/804252/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/52/804252/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/52/804252/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/52/804252/4"}}},"commit":{"parents":[{"commit":"0e4c7a056c8c9ca9e91ab43ebb30f9d37a330d83","subject":"nb: set chassis_name for new gateway chassis","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/0e4c7a056c8c9ca9e91ab43ebb30f9d37a330d83"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2021-08-11 14:30:19.000000000","tz":-300},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2021-11-18 18:21:22.000000000","tz":-360},"subject":"Use setkey for DbSetCommand maps","message":"Use setkey for DbSetCommand maps\n\nsetkey() has been available for a long time now, so we can safely\nuse it and drop the verify() call, which in certain circumstances\ncan lead to a busy loop until the connection timeout passes, e.g.\nml2/ovn\u0027s agent health check timestamp mechanism.\n\nChange-Id: I5fa20962f48a49aa5eb137a40861cf2ebc9e42d0\nRelated-Bug: #1907836\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/37e99a2569fd060282fb38f1e03f5e852af4e337"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/37e99a2569fd060282fb38f1e03f5e852af4e337"}]},"branch":"refs/heads/master"},"f3e0b0f72fcdd020bc38414eb195c99857acf08f":{"kind":"REWORK","_number":5,"created":"2021-12-13 09:16:58.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/52/804252/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/52/804252/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/52/804252/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/52/804252/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/52/804252/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/52/804252/5"}}},"commit":{"parents":[{"commit":"0e4c7a056c8c9ca9e91ab43ebb30f9d37a330d83","subject":"nb: set chassis_name for new gateway chassis","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/0e4c7a056c8c9ca9e91ab43ebb30f9d37a330d83"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2021-08-11 14:30:19.000000000","tz":-300},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2021-12-13 09:16:58.000000000","tz":0},"subject":"Use setkey for DbSetCommand maps","message":"Use setkey for DbSetCommand maps\n\nsetkey() has been available for a long time now, so we can safely\nuse it and drop the verify() call, which in certain circumstances\ncan lead to a busy loop until the connection timeout passes, e.g.\nml2/ovn\u0027s agent health check timestamp mechanism.\n\nChange-Id: I5fa20962f48a49aa5eb137a40861cf2ebc9e42d0\nRelated-Bug: #1907836\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/f3e0b0f72fcdd020bc38414eb195c99857acf08f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/f3e0b0f72fcdd020bc38414eb195c99857acf08f"}]},"branch":"refs/heads/master"},"c0f85c965022da4672d55d9245d20803c74778b4":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2022-01-05 14:44:21.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/52/804252/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/52/804252/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/52/804252/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/52/804252/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/52/804252/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/52/804252/6"}}},"commit":{"parents":[{"commit":"2c6b6c9c8effc3e3cca3d151a51292b552762ab0","subject":"Merge \"Allow functional tests to pass on older OVN w/o IC\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/2c6b6c9c8effc3e3cca3d151a51292b552762ab0"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2021-08-11 14:30:19.000000000","tz":-300},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2022-01-05 14:44:21.000000000","tz":0},"subject":"Use setkey for DbSetCommand maps","message":"Use setkey for DbSetCommand maps\n\nsetkey() has been available for a long time now, so we can safely\nuse it and drop the verify() call, which in certain circumstances\ncan lead to a busy loop until the connection timeout passes, e.g.\nml2/ovn\u0027s agent health check timestamp mechanism.\n\nChange-Id: I5fa20962f48a49aa5eb137a40861cf2ebc9e42d0\nRelated-Bug: #1907836\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/c0f85c965022da4672d55d9245d20803c74778b4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/c0f85c965022da4672d55d9245d20803c74778b4"}]},"branch":"refs/heads/master","description":"Rebase"},"da45d18b1dd498ef753d2ed334fab61d31b92dc9":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2022-02-16 19:03:40.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/52/804252/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/52/804252/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/52/804252/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/52/804252/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/52/804252/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/52/804252/7"}}},"commit":{"parents":[{"commit":"dc15bfd5729eb9f2b404288a4aa901c318274507","subject":"Log what exception caused abort of the transaction","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/dc15bfd5729eb9f2b404288a4aa901c318274507"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2021-08-11 14:30:19.000000000","tz":-300},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2022-02-16 19:03:40.000000000","tz":0},"subject":"Use setkey for DbSetCommand maps","message":"Use setkey for DbSetCommand maps\n\nsetkey() has been available for a long time now, so we can safely\nuse it and drop the verify() call, which in certain circumstances\ncan lead to a busy loop until the connection timeout passes, e.g.\nml2/ovn\u0027s agent health check timestamp mechanism.\n\nChange-Id: I5fa20962f48a49aa5eb137a40861cf2ebc9e42d0\nRelated-Bug: #1907836\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/da45d18b1dd498ef753d2ed334fab61d31b92dc9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/da45d18b1dd498ef753d2ed334fab61d31b92dc9"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
