)]}'
{"id":"openstack%2Fovsdbapp~813622","triplet_id":"openstack%2Fovsdbapp~master~I4309b793c5d7b5b335e4d37fa7ced30b9f034b48","project":"openstack/ovsdbapp","branch":"master","topic":"set-chassis-name","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2021-10-13 08:04:47.000000000","reason":"Rodolfo Alonso replied on the change"},"6773":{"account":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"last_update":"2021-10-13 10:17:07.000000000","reason":"Lucas Alvares Gomes replied on the change"},"27623":{"account":{"_account_id":27623,"name":"Zuul","email":"test.openstack.test@gmail.com","username":"openstack.test.test"},"last_update":"2021-10-13 11:18:47.000000000","reason":"Change was submitted"},"5756":{"account":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"last_update":"2021-10-13 11:18:47.000000000","reason":"Change was submitted"},"33871":{"account":{"_account_id":33871,"name":"Anton Vazhnetsov","display_name":"Anton Vazhnetsov","email":"dragen15051@gmail.com","username":"0x5b"},"last_update":"2021-10-13 11:18:47.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4309b793c5d7b5b335e4d37fa7ced30b9f034b48","subject":"nb: set chassis_name for new gateway chassis","status":"MERGED","created":"2021-10-12 12:33:41.000000000","updated":"2021-10-13 11:20:26.000000000","submitted":"2021-10-13 11:18:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"813622-set-chassis-name","meta_rev_id":"e40ba2d346b283847bfc494016641dfd5f1056dd","_number":813622,"virtual_id_number":813622,"owner":{"_account_id":33871,"name":"Anton Vazhnetsov","display_name":"Anton Vazhnetsov","email":"dragen15051@gmail.com","username":"0x5b"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-10-13 11:18:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2021-10-13 08:04:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2021-10-13 10:17:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2021-10-13 10:17:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":27623,"name":"Zuul","email":"test.openstack.test@gmail.com","username":"openstack.test.test"},{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}],"reviewers":{"REVIEWER":[{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27623,"name":"Zuul","email":"test.openstack.test@gmail.com","username":"openstack.test.test"}],"CC":[{"_account_id":33896,"name":"Vladislav Odintsov","email":"odivlad@gmail.com","username":"odivlad"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-12 13:10:04.000000000","updated_by":{"_account_id":33871,"name":"Anton Vazhnetsov","display_name":"Anton Vazhnetsov","email":"dragen15051@gmail.com","username":"0x5b"},"reviewer":{"_account_id":33896,"name":"Vladislav Odintsov","email":"odivlad@gmail.com","username":"odivlad"},"state":"CC"},{"updated":"2021-10-12 13:10:04.000000000","updated_by":{"_account_id":33871,"name":"Anton Vazhnetsov","display_name":"Anton Vazhnetsov","email":"dragen15051@gmail.com","username":"0x5b"},"reviewer":{"_account_id":27623,"name":"Zuul","email":"test.openstack.test@gmail.com","username":"openstack.test.test"},"state":"REVIEWER"},{"updated":"2021-10-12 13:10:04.000000000","updated_by":{"_account_id":33871,"name":"Anton Vazhnetsov","display_name":"Anton Vazhnetsov","email":"dragen15051@gmail.com","username":"0x5b"},"reviewer":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"state":"REVIEWER"},{"updated":"2021-10-12 13:10:04.000000000","updated_by":{"_account_id":33871,"name":"Anton Vazhnetsov","display_name":"Anton Vazhnetsov","email":"dragen15051@gmail.com","username":"0x5b"},"reviewer":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"state":"REVIEWER"},{"updated":"2021-10-12 13:10:04.000000000","updated_by":{"_account_id":33871,"name":"Anton Vazhnetsov","display_name":"Anton Vazhnetsov","email":"dragen15051@gmail.com","username":"0x5b"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2021-10-12 14:14:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"72641c1ff41ac4c7b13f6c8f9469321e229a752a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33871,"name":"Anton Vazhnetsov","display_name":"Anton Vazhnetsov","email":"dragen15051@gmail.com","username":"0x5b"},"date":"2021-10-12 12:33:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"648c3946d6448464052ce8fe4a80bf0791a8c0df","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33871,"name":"Anton Vazhnetsov","display_name":"Anton Vazhnetsov","email":"dragen15051@gmail.com","username":"0x5b"},"date":"2021-10-12 13:08:00.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c38656a82316fb20917f341ef762a7838da0dad9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-12 14:14:28.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b1bd16a60be049168c8f2c3c593bd0d7 : SUCCESS in 4m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bca67221355f47758c95bd1f4ce8e28b : SUCCESS in 4m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eb0499f690504ba38ccdd2241d5260fd : SUCCESS in 3m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/86feeeda71d745c2bcf493a36e340304 : SUCCESS in 4m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b0c13da55afe4d729044c3658fcca017 : SUCCESS in 3m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef25e72874a543c0b75d1abb389f61f7 : SUCCESS in 5m 36s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/298d1ddfc26a4e1ba30af785fd75b4e8 : SUCCESS in 6m 10s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/3311ae01609a4fa084a054d36d0c2f24 : SUCCESS in 59m 40s","accounts_in_message":[],"_revision_number":2},{"id":"b83ae28508c3219b9a06c5d6d764de92d217d12a","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-10-13 08:04:47.000000000","message":"Patch Set 2: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"ac8e44179e255cc1924da2e91c7d4b4083ae8fc2","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2021-10-13 10:17:07.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"996f3465ae7f944c81271a489d5e433b75df28e9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-13 10:17:40.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"28c75436fe3eb5783e2c6ca81c277ff56a3be247","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-13 11:18:47.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b3e171a2bb2e47c1b414ef5a314fdc69 : SUCCESS in 5m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/339aeed19033408c962d5e8d19dbb8e2 : SUCCESS in 5m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d203aa54a2e44212b68dd61cb5e00b28 : SUCCESS in 4m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/df69d95c877449aeae314129547fd12d : SUCCESS in 5m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/419c7f19b4a446d7b227ee51629e3e00 : SUCCESS in 7m 26s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/ac4a4c80b17d4bb2b7b9870b73f630e0 : SUCCESS in 7m 33s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/52c3039cda36460395b3fb53b2fb11ad : SUCCESS in 58m 39s","accounts_in_message":[],"_revision_number":2},{"id":"e61531dab65b7b8209650b28aae4df3e0359dba2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-13 11:18:47.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"e40ba2d346b283847bfc494016641dfd5f1056dd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-13 11:20:26.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f48a3e0dc185447489b99195ec60cd5a : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0e4c7a056c8c9ca9e91ab43ebb30f9d37a330d83","revisions":{"6eb543d5a29dc574074b63561cc94eb30ce55b7a":{"kind":"REWORK","_number":1,"created":"2021-10-12 12:33:41.000000000","uploader":{"_account_id":33871,"name":"Anton Vazhnetsov","display_name":"Anton Vazhnetsov","email":"dragen15051@gmail.com","username":"0x5b"},"ref":"refs/changes/22/813622/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/22/813622/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/22/813622/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/22/813622/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/22/813622/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/22/813622/1"}}},"commit":{"parents":[{"commit":"47a01029f5ce70c11078c9d79ae1b95825e9cc50","subject":"Merge \"nb: fix route.output_port name\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/47a01029f5ce70c11078c9d79ae1b95825e9cc50"}]}],"author":{"name":"Anton Vazhnetsov","email":"AVazhnetsov@croc.ru","date":"2021-07-14 10:43:02.000000000","tz":180},"committer":{"name":"Anton Vazhnetsov","email":"AVazhnetsov@croc.ru","date":"2021-10-12 12:22:13.000000000","tz":180},"subject":"nb: set chassis_name for new gateway chassis","message":"nb: set chassis_name for new gateway chassis\n\nSet the passed \u0027chassis_name\u0027 on created gateway chassis.\nAccording to https://github.com/ovn-org/ovn/blob/main/ovn-nb.ovsschema#L538\n\nClosses-Bug: #1946746\nChange-Id: I4309b793c5d7b5b335e4d37fa7ced30b9f034b48\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/6eb543d5a29dc574074b63561cc94eb30ce55b7a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/6eb543d5a29dc574074b63561cc94eb30ce55b7a"}]},"branch":"refs/heads/master"},"0e4c7a056c8c9ca9e91ab43ebb30f9d37a330d83":{"kind":"REWORK","_number":2,"created":"2021-10-12 13:08:00.000000000","uploader":{"_account_id":33871,"name":"Anton Vazhnetsov","display_name":"Anton Vazhnetsov","email":"dragen15051@gmail.com","username":"0x5b"},"ref":"refs/changes/22/813622/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/22/813622/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/22/813622/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/22/813622/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/22/813622/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/22/813622/2"}}},"commit":{"parents":[{"commit":"47a01029f5ce70c11078c9d79ae1b95825e9cc50","subject":"Merge \"nb: fix route.output_port name\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/47a01029f5ce70c11078c9d79ae1b95825e9cc50"}]}],"author":{"name":"Anton Vazhnetsov","email":"AVazhnetsov@croc.ru","date":"2021-07-14 10:43:02.000000000","tz":180},"committer":{"name":"Anton Vazhnetsov","email":"AVazhnetsov@croc.ru","date":"2021-10-12 13:07:29.000000000","tz":180},"subject":"nb: set chassis_name for new gateway chassis","message":"nb: set chassis_name for new gateway chassis\n\nSet the passed \u0027chassis_name\u0027 on created gateway chassis.\nAccording to https://github.com/ovn-org/ovn/blob/main/ovn-nb.ovsschema#L538\n\nClosses-Bug: #1946746\nChange-Id: I4309b793c5d7b5b335e4d37fa7ced30b9f034b48\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/0e4c7a056c8c9ca9e91ab43ebb30f9d37a330d83"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/0e4c7a056c8c9ca9e91ab43ebb30f9d37a330d83"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
