)]}'
{"id":"openstack%2Fovsdbapp~826567","triplet_id":"openstack%2Fovsdbapp~stable%2Fussuri~Ic67575ff26c689ec8ab1a5c619ae0b0a2846b05f","project":"openstack/ovsdbapp","branch":"stable/ussuri","topic":"improve-errors-logging-stable/xena-stable/wallaby-stable/victoria-stable/ussuri","attention_set":{},"removed_from_attention_set":{"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2022-02-18 17:38:26.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic67575ff26c689ec8ab1a5c619ae0b0a2846b05f","subject":"Log what exception caused abort of the transaction","status":"MERGED","created":"2022-01-27 07:33:07.000000000","updated":"2022-02-18 17:39:53.000000000","submitted":"2022-02-18 17:38:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"826567-improve-errors-logging-stable/xena-stable/wallaby-stable/victoria-stable/ussuri","cherry_pick_of_change":826566,"cherry_pick_of_patch_set":1,"meta_rev_id":"2c09cb7b04986a567a519c35b8d475d754766d9b","_number":826567,"virtual_id_number":826567,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-18 17:38:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-02-04 19:50:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-02-18 16:25:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-27 08:39:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-04 19:50:41.000000000","updated_by":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"reviewer":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"state":"REVIEWER"}],"messages":[{"id":"5c9fda18ddca8ec31a0a6ab438ad75387d307e63","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-01-27 07:33:07.000000000","message":"Patch Set 1: Cherry Picked from branch stable/victoria.","accounts_in_message":[],"_revision_number":1},{"id":"173161b664297d0a3c85efb862eef523d8a6351e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-27 08:39:04.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/968b9ea998114abda1a4d454d341ba9b : SUCCESS in 4m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3f2a812867174f1285cc4e12df3b536e : SUCCESS in 7m 26s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5b9f3bbffade4fe29e8256355f1bcab9 : SUCCESS in 3m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32c9bb0185fa4c4ba67575acae26be72 : SUCCESS in 8m 06s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/adfc327ee37945818e8619f2ba10881c : FAILURE in 6m 05s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/f456acc58bd04daba923566537553c42 : SUCCESS in 1h 02m 52s","accounts_in_message":[],"_revision_number":1},{"id":"500339fadedda2ea4c6f000f7ca8df5e82c8da24","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2022-02-04 17:59:21.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"03fd75f6d01e25b5772e0b050b18bee4d7561c21","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-04 19:18:46.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a54f7058151f4bd9a5f1720ba4f7d1e6 : SUCCESS in 3m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/73ce2af911e04f8abf8d33df9fe1eae1 : SUCCESS in 3m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/de9e7f7afa704527809ab52c62c540d2 : SUCCESS in 3m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f5455a27b232435dbc32cbcfd7158ff5 : SUCCESS in 5m 30s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/2e0062a5d89d4a10aa2805a2b14db012 : SUCCESS in 5m 43s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/ff1a444dab6a41e1821ed151c9858028 : SUCCESS in 1h 13m 42s","accounts_in_message":[],"_revision_number":2},{"id":"7245bee020f010d40d382a1c07ae49a507e02f25","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2022-02-04 19:50:41.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"1e681c93602e8cff764df07b220d33783d12a998","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2022-02-18 16:25:05.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"6133053549f0ab768422b3340064e40a651b5b22","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-18 16:25:47.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"8e9d0afcae165b2ecb0574c689380e59b76e286a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-18 17:38:26.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6e773bf3a0124fcbb4a79512a70683c5 : SUCCESS in 3m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6603801223614cafa6a57a4aa7cc883d : SUCCESS in 3m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/29c5377a6ab54703ac4a9054a71450df : SUCCESS in 4m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/73debc13b51c40bc8fdcfc3de47e24b3 : SUCCESS in 5m 46s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/df58d16dce034053bde95c35ea6a2efe : SUCCESS in 6m 00s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/1eff3b33461f47c28f733448eaae1a0d : SUCCESS in 1h 08m 59s","accounts_in_message":[],"_revision_number":2},{"id":"b3c02d4acbf08679a6f78c7daf2ac758db00b6b9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-18 17:38:26.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"2c09cb7b04986a567a519c35b8d475d754766d9b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-18 17:39:53.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e77e2a6dc444d5a85319ff3bdb49d3e : SUCCESS in 54s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"4a6df23b25c6758b89076066c6befab2b2146bb3","revisions":{"325d0ec8d059c8f0279c5f354276cf726b693157":{"kind":"REWORK","_number":1,"created":"2022-01-27 07:33:07.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/67/826567/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/67/826567/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/67/826567/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/67/826567/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/67/826567/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/67/826567/1"}}},"commit":{"parents":[{"commit":"7e400f117f5269930fe11caff153a4966dbf614c","subject":"Actually close the connection in Connection.stop()","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/7e400f117f5269930fe11caff153a4966dbf614c"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2022-01-21 09:06:33.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2022-01-27 07:33:07.000000000","tz":0},"subject":"Log what exception caused abort of the transaction","message":"Log what exception caused abort of the transaction\n\nIn some cases, when ovsdb transaction is aborted, knowing what error\ncaused that can help to understand the root cause of the problem\nso this patch adds such logging on the DEBUG level.\n\nChange-Id: Ic67575ff26c689ec8ab1a5c619ae0b0a2846b05f\n(cherry picked from commit dc15bfd5729eb9f2b404288a4aa901c318274507)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/325d0ec8d059c8f0279c5f354276cf726b693157"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/325d0ec8d059c8f0279c5f354276cf726b693157"}]},"branch":"refs/heads/stable/ussuri"},"4a6df23b25c6758b89076066c6befab2b2146bb3":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-02-04 17:59:21.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/67/826567/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/67/826567/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/67/826567/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/67/826567/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/67/826567/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/67/826567/2"}}},"commit":{"parents":[{"commit":"ce5c016bfecad70be47c55af27b200a1e6cfea88","subject":"tools: run functional tests against ovn/main branch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/ce5c016bfecad70be47c55af27b200a1e6cfea88"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2022-01-21 09:06:33.000000000","tz":60},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2022-02-04 17:59:21.000000000","tz":0},"subject":"Log what exception caused abort of the transaction","message":"Log what exception caused abort of the transaction\n\nIn some cases, when ovsdb transaction is aborted, knowing what error\ncaused that can help to understand the root cause of the problem\nso this patch adds such logging on the DEBUG level.\n\nChange-Id: Ic67575ff26c689ec8ab1a5c619ae0b0a2846b05f\n(cherry picked from commit dc15bfd5729eb9f2b404288a4aa901c318274507)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/4a6df23b25c6758b89076066c6befab2b2146bb3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/4a6df23b25c6758b89076066c6befab2b2146bb3"}]},"branch":"refs/heads/stable/ussuri","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
