)]}'
{"id":"openstack%2Fovsdbapp~828376","triplet_id":"openstack%2Fovsdbapp~master~Ib40baa942b7213af226a093969f5da0445d43d3f","project":"openstack/ovsdbapp","branch":"master","topic":"bug/1959975","attention_set":{},"removed_from_attention_set":{"5756":{"account":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"last_update":"2022-03-24 19:27:41.000000000","reason":"Terry Wilson replied on the change"},"8655":{"account":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"last_update":"2022-03-25 20:43:28.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ib40baa942b7213af226a093969f5da0445d43d3f","subject":"Change lookup() default parameter behavior","status":"ABANDONED","created":"2022-02-08 16:35:59.000000000","updated":"2022-03-25 20:43:28.000000000","total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"677df8d048daf726e15fda3ee17be1941031c06d","_number":828376,"virtual_id_number":828376,"owner":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-02-08 19:58:03.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"all":[{"value":-1,"date":"2022-03-25 20:42:02.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-08 17:05:37.000000000","updated_by":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"reviewer":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"state":"CC"},{"updated":"2022-02-08 19:58:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-25 20:42:02.000000000","updated_by":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"reviewer":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"state":"REVIEWER"}],"messages":[{"id":"c4991e9dbb29949a3a0a32cc9f0b4bed23e513c6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2022-02-08 16:35:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ec043bff519b2676b0fbb184b576db7e771f3f75","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2022-02-08 17:05:37.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5114b9fadc2a309b4c3db1f85e8368c7191f9ea5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2022-02-08 17:09:04.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"91ca6c8a3491ac67541dc87716b362a019010192","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2022-02-08 17:09:08.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8f5b8455e261fa9f2377033f19e2a965d8fbce53","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-08 19:58:03.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a8b0a6d5c67c4758a657a47673ec3f84 : SUCCESS in 4m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bebd3c1590a5476d898236ba74e9127e : SUCCESS in 5m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a54cb8d370ac4996a5e1fc5be7dfe744 : SUCCESS in 4m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0603b6de2fb045979b19a7dc30e4e764 : SUCCESS in 5m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b7107403c3b44ff1a2eeda12dc459519 : SUCCESS in 6m 27s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/6d58e10f95a644bba8304e9c2fae6b52 : SUCCESS in 7m 04s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/8b3f4659dd864738a457632e870ae54b : FAILURE in 2h 24m 22s","accounts_in_message":[],"_revision_number":2},{"id":"d1739c1861e660b1998928704dc07e1ff40ab0db","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2022-03-24 19:27:41.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d9582ff22c79160c71909201da0475b17f85254c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-24 22:36:04.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/83d41e28e0664d348351c708c5f57b14 : SUCCESS in 5m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/902a2ec01ac24b1ea455993deef2eed6 : SUCCESS in 5m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e1770b8b11f04f45b8cc4fb875b9d0a2 : SUCCESS in 5m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ae08efb388674b81be4cc6ca3884e490 : SUCCESS in 5m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa6fdd7731434f69aef22878a18a8d3f : SUCCESS in 7m 48s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/b951030380f8440699c3c314e1b5ea90 : SUCCESS in 8m 00s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/35b5ff1beaf24501b9b2134a0bc13353 : TIMED_OUT in 3h 05m 26s","accounts_in_message":[],"_revision_number":2},{"id":"ae4d7f4af7204552bbcdee78ae0a221cb6c7a128","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2022-03-25 20:42:02.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"677df8d048daf726e15fda3ee17be1941031c06d","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2022-03-25 20:43:28.000000000","message":"Abandoned\n\nThe lookup() wait will be removed","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"beef6ab35b69272de2e1697f081faec1f935b5e0","revisions":{"a8babf150a75258f0d11243dd63cd9fdf090c9f7":{"kind":"REWORK","_number":1,"created":"2022-02-08 16:35:59.000000000","uploader":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"ref":"refs/changes/76/828376/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/76/828376/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/76/828376/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/76/828376/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/76/828376/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/76/828376/1"}}},"commit":{"parents":[{"commit":"bb1a1463f3caf681193d7bb94d1752287dc77f0f","subject":"Fix dynamic get of nofity_handler in lookup()","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/bb1a1463f3caf681193d7bb94d1752287dc77f0f"}]}],"author":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2022-02-08 16:34:46.000000000","tz":0},"committer":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2022-02-08 16:35:45.000000000","tz":0},"subject":"Change lookup() default parameter behavior","message":"Change lookup() default parameter behavior\n\nPreviously, the passed default was considered only if the immediate\nlookup failed and didn\u0027t use the WaitEvent. This patch changes it so\nwhenever default is passed and exception would be raised we return\ndefault instead.\n\nChange-Id: Ib40baa942b7213af226a093969f5da0445d43d3f\nSigned-off-by: Jakub Libosvar \u003clibosvar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/a8babf150a75258f0d11243dd63cd9fdf090c9f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/a8babf150a75258f0d11243dd63cd9fdf090c9f7"}]},"branch":"refs/heads/master"},"beef6ab35b69272de2e1697f081faec1f935b5e0":{"kind":"REWORK","_number":2,"created":"2022-02-08 17:09:04.000000000","uploader":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"ref":"refs/changes/76/828376/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/76/828376/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/76/828376/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/76/828376/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/76/828376/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/76/828376/2"}}},"commit":{"parents":[{"commit":"bb1a1463f3caf681193d7bb94d1752287dc77f0f","subject":"Fix dynamic get of nofity_handler in lookup()","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/bb1a1463f3caf681193d7bb94d1752287dc77f0f"}]}],"author":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2022-02-08 16:34:46.000000000","tz":0},"committer":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2022-02-08 17:09:04.000000000","tz":0},"subject":"Change lookup() default parameter behavior","message":"Change lookup() default parameter behavior\n\nPreviously, the passed default was considered only if the immediate\nlookup failed and didn\u0027t use the WaitEvent. This patch changes it so\nwhenever default is passed and exception would be raised we return\ndefault instead.\n\nChange-Id: Ib40baa942b7213af226a093969f5da0445d43d3f\nSigned-off-by: Jakub Libosvar \u003clibosvar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/beef6ab35b69272de2e1697f081faec1f935b5e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/beef6ab35b69272de2e1697f081faec1f935b5e0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
