)]}'
{"id":"openstack%2Fovsdbapp~828720","triplet_id":"openstack%2Fovsdbapp~master~I37138e91f9dd3e42908ac1e7e7c123323177a8e8","project":"openstack/ovsdbapp","branch":"master","topic":"vtep-support","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2022-07-20 14:04:17.000000000","reason":"\u003cGERRIT_ACCOUNT_16688\u003e replied on the change","reason_account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},"6773":{"account":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"last_update":"2022-07-20 15:08:43.000000000","reason":"Change was submitted"},"27623":{"account":{"_account_id":27623,"name":"Zuul","email":"test.openstack.test@gmail.com","username":"openstack.test.test"},"last_update":"2022-07-20 15:08:43.000000000","reason":"Change was submitted"},"33896":{"account":{"_account_id":33896,"name":"Vladislav Odintsov","email":"odivlad@gmail.com","username":"odivlad"},"last_update":"2022-07-20 15:08:43.000000000","reason":"Change was submitted"},"23804":{"account":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"last_update":"2022-07-20 15:08:43.000000000","reason":"Change was submitted"},"5756":{"account":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"last_update":"2022-03-30 13:42:11.000000000","reason":"Terry Wilson replied on the change"},"33871":{"account":{"_account_id":33871,"name":"Anton Vazhnetsov","display_name":"Anton Vazhnetsov","email":"dragen15051@gmail.com","username":"0x5b"},"last_update":"2022-07-20 15:08:43.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I37138e91f9dd3e42908ac1e7e7c123323177a8e8","subject":"vtep: add support for \u0027hardware_vtep\u0027 schema","status":"MERGED","created":"2022-02-10 14:33:10.000000000","updated":"2022-07-20 15:09:43.000000000","submitted":"2022-07-20 15:08:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":47,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"828720-vtep-support","meta_rev_id":"511fa2036054576f71ea7c04ffc7d67dd898dcd4","_number":828720,"virtual_id_number":828720,"owner":{"_account_id":33871,"name":"Anton Vazhnetsov","display_name":"Anton Vazhnetsov","email":"dragen15051@gmail.com","username":"0x5b"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-20 15:08:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2022-07-20 14:04:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-03-30 13:42:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2022-07-20 14:04:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":27623,"name":"Zuul","email":"test.openstack.test@gmail.com","username":"openstack.test.test"},{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"}],"reviewers":{"REVIEWER":[{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"_account_id":27623,"name":"Zuul","email":"test.openstack.test@gmail.com","username":"openstack.test.test"}],"CC":[{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"_account_id":33896,"name":"Vladislav Odintsov","email":"odivlad@gmail.com","username":"odivlad"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-10 14:46:00.000000000","updated_by":{"_account_id":33871,"name":"Anton Vazhnetsov","display_name":"Anton Vazhnetsov","email":"dragen15051@gmail.com","username":"0x5b"},"reviewer":{"_account_id":33896,"name":"Vladislav Odintsov","email":"odivlad@gmail.com","username":"odivlad"},"state":"CC"},{"updated":"2022-02-10 14:46:00.000000000","updated_by":{"_account_id":33871,"name":"Anton Vazhnetsov","display_name":"Anton Vazhnetsov","email":"dragen15051@gmail.com","username":"0x5b"},"reviewer":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"state":"REVIEWER"},{"updated":"2022-02-10 14:46:00.000000000","updated_by":{"_account_id":33871,"name":"Anton Vazhnetsov","display_name":"Anton Vazhnetsov","email":"dragen15051@gmail.com","username":"0x5b"},"reviewer":{"_account_id":27623,"name":"Zuul","email":"test.openstack.test@gmail.com","username":"openstack.test.test"},"state":"REVIEWER"},{"updated":"2022-02-10 15:48:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-11 10:48:54.000000000","updated_by":{"_account_id":33871,"name":"Anton Vazhnetsov","display_name":"Anton Vazhnetsov","email":"dragen15051@gmail.com","username":"0x5b"},"reviewer":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"state":"REVIEWER"},{"updated":"2022-03-02 13:54:30.000000000","updated_by":{"_account_id":33871,"name":"Anton Vazhnetsov","display_name":"Anton Vazhnetsov","email":"dragen15051@gmail.com","username":"0x5b"},"reviewer":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"state":"CC"},{"updated":"2022-03-02 13:54:30.000000000","updated_by":{"_account_id":33871,"name":"Anton Vazhnetsov","display_name":"Anton Vazhnetsov","email":"dragen15051@gmail.com","username":"0x5b"},"reviewer":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"state":"CC"},{"updated":"2022-03-02 13:54:30.000000000","updated_by":{"_account_id":33871,"name":"Anton Vazhnetsov","display_name":"Anton Vazhnetsov","email":"dragen15051@gmail.com","username":"0x5b"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2022-03-03 12:29:56.000000000","updated_by":{"_account_id":33871,"name":"Anton Vazhnetsov","display_name":"Anton Vazhnetsov","email":"dragen15051@gmail.com","username":"0x5b"},"reviewer":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"state":"REVIEWER"},{"updated":"2022-03-09 15:24:59.000000000","updated_by":{"_account_id":33871,"name":"Anton Vazhnetsov","display_name":"Anton Vazhnetsov","email":"dragen15051@gmail.com","username":"0x5b"},"reviewer":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"state":"CC"},{"updated":"2022-03-14 12:00:55.000000000","updated_by":{"_account_id":33871,"name":"Anton Vazhnetsov","display_name":"Anton Vazhnetsov","email":"dragen15051@gmail.com","username":"0x5b"},"reviewer":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"state":"REVIEWER"},{"updated":"2022-07-12 12:11:53.000000000","updated_by":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"reviewer":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"state":"CC"},{"updated":"2022-07-13 10:00:26.000000000","updated_by":{"_account_id":33871,"name":"Anton Vazhnetsov","display_name":"Anton Vazhnetsov","email":"dragen15051@gmail.com","username":"0x5b"},"reviewer":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"state":"REVIEWER"}],"messages":[{"id":"3f545b53dcf7f93d99670cf6c35666b925981bc4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33871,"name":"Anton Vazhnetsov","display_name":"Anton Vazhnetsov","email":"dragen15051@gmail.com","username":"0x5b"},"date":"2022-02-10 14:33:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6fc4509c55e0022245f9212ac0bcf6a4734b6c63","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33871,"name":"Anton Vazhnetsov","display_name":"Anton Vazhnetsov","email":"dragen15051@gmail.com","username":"0x5b"},"date":"2022-02-10 14:34:28.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"7a1bd31c0a20e70b0d038c12cf27068537b63081","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-10 15:48:56.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/91342cf0eaf64197a0a95ee118a987c3 : SUCCESS in 4m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/88d8b0598ec141ddba093d66f67b4c6f : SUCCESS in 4m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/68af70912e34407583f0c7dfbb671d39 : SUCCESS in 4m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b1a6837e8db8460ebaac59aae0d639c5 : SUCCESS in 3m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d04b001f7eb42d7b47d57656e4bcb0f : SUCCESS in 5m 33s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/a0a331bc3331451e9947a2b482e2782f : FAILURE in 6m 21s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/f3dcbedc64f1469193e86508441be294 : SUCCESS in 1h 08m 56s","accounts_in_message":[],"_revision_number":2},{"id":"137951243dc1bcd04a729f043e9721974262238d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33871,"name":"Anton Vazhnetsov","display_name":"Anton Vazhnetsov","email":"dragen15051@gmail.com","username":"0x5b"},"date":"2022-02-11 09:32:52.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"fa9e6e67d064b9606eb8605f5d326c8d44d5f358","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-11 10:34:01.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0f648bb69077490ea88a52ec0223ce28 : SUCCESS in 8m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/599751ff115d447cbae49b201dcd0d43 : SUCCESS in 4m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/96eb7d0050274f34abb798671f1bd7a7 : SUCCESS in 4m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2e71dbb4aecc4535ac0f85dd17e33404 : SUCCESS in 4m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d0e9bced2954b609bce2c28cd821516 : SUCCESS in 5m 48s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/2522a245434a40ddbe8668ac8e80f37f : SUCCESS in 6m 48s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/aec1269139fb4876b44208fd93e0b08a : SUCCESS in 56m 14s","accounts_in_message":[],"_revision_number":3},{"id":"6d6a141bacf2dffb5173084d92ad9eae8af6a660","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-03-04 09:12:32.000000000","message":"Patch Set 3: Code-Review-1\n\n(9 comments)","accounts_in_message":[],"_revision_number":3},{"id":"b66f1eb7f4824c71ba48ceb32886d1579ff5ce3d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33871,"name":"Anton Vazhnetsov","display_name":"Anton Vazhnetsov","email":"dragen15051@gmail.com","username":"0x5b"},"date":"2022-03-04 11:54:18.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"f60e7364223f4b92d51bf9eb21ed7d337ca11208","author":{"_account_id":33871,"name":"Anton Vazhnetsov","display_name":"Anton Vazhnetsov","email":"dragen15051@gmail.com","username":"0x5b"},"date":"2022-03-04 12:08:50.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"ee2496964afd74ead817705903b6d5ba798ee33f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-04 12:57:46.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/70a1e0de3c7443219e98d6a892141a53 : SUCCESS in 6m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/917ae000aa864c1bb99419e32dab22ed : SUCCESS in 5m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6414d4e0192d4384bd37890bada41b93 : SUCCESS in 6m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/eca7d746a55d4bb1b07a7453c269ee37 : SUCCESS in 5m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a5469cac14aa483780bf9f70e71d02a0 : SUCCESS in 7m 11s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/520e048099774946bf42184f2f9f0781 : SUCCESS in 7m 33s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/a031b121319c4d2ebc46515f0ad4492b : SUCCESS in 58m 03s","accounts_in_message":[],"_revision_number":4},{"id":"cdbad0c6c0106dd41840e2a7cc6ded6d9cfe5c25","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33871,"name":"Anton Vazhnetsov","display_name":"Anton Vazhnetsov","email":"dragen15051@gmail.com","username":"0x5b"},"date":"2022-03-05 22:00:57.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"6cd7338dec122dfd0beed06ba45c5b0ffb55111c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-05 22:59:22.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bd5f1cb4c61c40c0b6f3b2b8e5c837af : SUCCESS in 5m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b53538e08b7f48c7a34771c67fc15a55 : SUCCESS in 4m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/66ce488bff6c42699a28b9a9c908a41b : SUCCESS in 4m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/41a02ab257fc4c4ebb074f58b7951e6d : SUCCESS in 5m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9c0bec379414e33b5ef5c1969ccb0de : SUCCESS in 6m 16s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/348440fd5fbf451ca71d7a021812a85f : SUCCESS in 7m 28s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/c3916f679d7b4ee19c28aecfc35cd7a3 : SUCCESS in 57m 06s","accounts_in_message":[],"_revision_number":5},{"id":"c73d09a73dcae38b0e11adf026317473bd202918","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-03-08 08:03:05.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"6eb8bfe10b5c627e04a7a8ff7d6ef5971ebb03f4","author":{"_account_id":33871,"name":"Anton Vazhnetsov","display_name":"Anton Vazhnetsov","email":"dragen15051@gmail.com","username":"0x5b"},"date":"2022-03-14 12:00:55.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"9bca5dce74bff5320b73f462eff7312505faf6df","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2022-03-22 03:35:43.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"6405fc41d188c5393d5ec94aea7441acb41529f1","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2022-03-25 15:38:50.000000000","message":"Patch Set 5: Code-Review-1\n\n(14 comments)","accounts_in_message":[],"_revision_number":5},{"id":"610b5b77b8a571ef970393aaa67000c59ba49f0b","author":{"_account_id":33871,"name":"Anton Vazhnetsov","display_name":"Anton Vazhnetsov","email":"dragen15051@gmail.com","username":"0x5b"},"date":"2022-03-28 09:16:49.000000000","message":"Patch Set 5:\n\n(3 comments)","accounts_in_message":[],"_revision_number":5},{"id":"cf2c9682824fc79c67339c5196ab624c347b1d19","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33871,"name":"Anton Vazhnetsov","display_name":"Anton Vazhnetsov","email":"dragen15051@gmail.com","username":"0x5b"},"date":"2022-03-28 09:18:00.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"57d264f742801485048f1aa1c8c3eb202ae02d78","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-28 10:14:34.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/00fa5f42054e4c78a26b6f4e52ee1a61 : SUCCESS in 6m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8844f84b8a59431baa98fb113893138a : SUCCESS in 5m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/83d891ecbe8846df98381386c3110bfb : SUCCESS in 5m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/98500971c1a14c3791ea68a1dc797d8d : SUCCESS in 5m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/107d419692814fbf8bbb72647a06a57c : SUCCESS in 7m 01s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/1ce9f55b3eb84768be9d2373aeaa9519 : SUCCESS in 7m 16s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/e3c7de4ec5914d23ac4d49b9988d139c : SUCCESS in 54m 00s","accounts_in_message":[],"_revision_number":6},{"id":"a5fbee08bd821f4f1921cb10cfd0882f0fb86b0d","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2022-03-29 20:29:17.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"72128564cdb05d4164b8115bc920ee86ba166279","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33871,"name":"Anton Vazhnetsov","display_name":"Anton Vazhnetsov","email":"dragen15051@gmail.com","username":"0x5b"},"date":"2022-03-30 12:09:14.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"28a72379b30ca88bd88cf6a5b784335a48b8e110","author":{"_account_id":33871,"name":"Anton Vazhnetsov","display_name":"Anton Vazhnetsov","email":"dragen15051@gmail.com","username":"0x5b"},"date":"2022-03-30 12:55:14.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"42795cf98b331c5cc9dfab2f49c6d9790d89a3aa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-30 12:57:55.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a57b6be6c2ff44e69e5de076c92a4317 : SUCCESS in 5m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b6d970e663d549e092c624066af18274 : SUCCESS in 5m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a78f9cdb6a8a48a19c5ff83cd0e8ebd7 : SUCCESS in 6m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/afbb0dd10cda4e3b9bef84b114b43326 : SUCCESS in 5m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84074eaa9a8c4a348d09cd82ddfe9a4f : SUCCESS in 6m 27s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/6ca64d6f445543bc86042549f9f30b8d : SUCCESS in 7m 50s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/285957c6df5f4043bf368b26c21640ef : SUCCESS in 46m 19s","accounts_in_message":[],"_revision_number":7},{"id":"c8cd7d50bd6645ca45f53e9af10bd6236bed986b","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2022-03-30 13:42:11.000000000","message":"Patch Set 7: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"eb772801c7127962f34a81d3a37c0a0b21fef3e8","author":{"_account_id":33871,"name":"Anton Vazhnetsov","display_name":"Anton Vazhnetsov","email":"dragen15051@gmail.com","username":"0x5b"},"date":"2022-03-30 13:46:29.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"e8050c663c218af93899412736b93e6e02e9d1f8","author":{"_account_id":33871,"name":"Anton Vazhnetsov","display_name":"Anton Vazhnetsov","email":"dragen15051@gmail.com","username":"0x5b"},"date":"2022-05-04 10:52:09.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"d85f4309d6ccd16da411468026f550f9e5853a67","author":{"_account_id":33896,"name":"Vladislav Odintsov","email":"odivlad@gmail.com","username":"odivlad"},"date":"2022-06-30 12:30:55.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"e72b05b67807bffe56e01d2c900c37bcf020fc18","author":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"date":"2022-07-12 12:11:53.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"d0dbc5f75854126404ad6299dfaa7c813bb73df1","author":{"_account_id":33871,"name":"Anton Vazhnetsov","display_name":"Anton Vazhnetsov","email":"dragen15051@gmail.com","username":"0x5b"},"date":"2022-07-12 14:23:03.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"f84a8392b23f589a1a564ada0462a1115a4de5b4","author":{"_account_id":33871,"name":"Anton Vazhnetsov","display_name":"Anton Vazhnetsov","email":"dragen15051@gmail.com","username":"0x5b"},"date":"2022-07-13 09:50:17.000000000","message":"Patch Set 7:\n\n(7 comments)","accounts_in_message":[],"_revision_number":7},{"id":"22ee96cddb31fc395e29b86b78c7f6c4922bb503","author":{"_account_id":33871,"name":"Anton Vazhnetsov","display_name":"Anton Vazhnetsov","email":"dragen15051@gmail.com","username":"0x5b"},"date":"2022-07-13 09:54:58.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"62afe52c6e273284ebe2ea2cc68f69ddef36ecaa","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-07-20 14:04:17.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"641fa8743e6b26df56f2f05c77593897be2f43ee","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-20 14:06:11.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"211f2361a2e6be06a675daa9dc415b1d929a84a7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-20 15:08:43.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c94710ed56c048c8a64823e82dbc8839\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eb6fec028d9a42c0a9e51205f66016ae : SUCCESS in 4m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2b157400588a4e50919aed2ba8849413 : SUCCESS in 4m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/69adf07b78f04eb1811c47d4f0253d87 : SUCCESS in 4m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b5a6766794cd43c9b8792e65baecb6e4 : SUCCESS in 6m 25s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/29a619bc83cd43818620e7541e00f986 : SUCCESS in 6m 46s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/58fba3dc0f6c43c2860c01bde49b9d2f : SUCCESS in 1h 01m 51s","accounts_in_message":[],"_revision_number":7},{"id":"f5d8285c1f7792cf772525af866062b4be3d57f4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-20 15:08:43.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7},{"id":"511fa2036054576f71ea7c04ffc7d67dd898dcd4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-20 15:09:43.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1f7e4a1936ee4b2ea484e029814ed440\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a59effb8fb3b4ef0b9835a2ea68841f9 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"06d25c29800dabc66d6727ef1161bebec7d21ad1","revisions":{"b0ccd001221def53a9c050d50e4ffd4abb6dfe77":{"kind":"REWORK","_number":1,"created":"2022-02-10 14:33:10.000000000","uploader":{"_account_id":33871,"name":"Anton Vazhnetsov","display_name":"Anton Vazhnetsov","email":"dragen15051@gmail.com","username":"0x5b"},"ref":"refs/changes/20/828720/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/20/828720/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/20/828720/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/20/828720/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/20/828720/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/20/828720/1"}}},"commit":{"parents":[{"commit":"dc15bfd5729eb9f2b404288a4aa901c318274507","subject":"Log what exception caused abort of the transaction","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/dc15bfd5729eb9f2b404288a4aa901c318274507"}]}],"author":{"name":"Anton Vazhnetsov","email":"AVazhnetsov@croc.ru","date":"2022-02-06 23:16:07.000000000","tz":180},"committer":{"name":"Anton Vazhnetsov","email":"AVazhnetsov@croc.ru","date":"2022-02-10 14:26:25.000000000","tz":180},"subject":"vtep: add support for \u0027hardware_vtep\u0027 schema","message":"vtep: add support for \u0027hardware_vtep\u0027 schema\n\nAdded support for the \"hardware_vtep\" schema. The ``HwVtepApiIdlImpl`` class provides\ninteraction with \"hw_vtep.db\".\n\nCloses-Bug: 1960513\n\nChange-Id: I37138e91f9dd3e42908ac1e7e7c123323177a8e8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/b0ccd001221def53a9c050d50e4ffd4abb6dfe77"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/b0ccd001221def53a9c050d50e4ffd4abb6dfe77"}]},"branch":"refs/heads/master"},"2cd1aa2c3189dadfe6246d18ef90d803ff0184c8":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-02-10 14:34:28.000000000","uploader":{"_account_id":33871,"name":"Anton Vazhnetsov","display_name":"Anton Vazhnetsov","email":"dragen15051@gmail.com","username":"0x5b"},"ref":"refs/changes/20/828720/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/20/828720/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/20/828720/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/20/828720/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/20/828720/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/20/828720/2"}}},"commit":{"parents":[{"commit":"dc15bfd5729eb9f2b404288a4aa901c318274507","subject":"Log what exception caused abort of the transaction","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/dc15bfd5729eb9f2b404288a4aa901c318274507"}]}],"author":{"name":"Anton Vazhnetsov","email":"AVazhnetsov@croc.ru","date":"2022-02-06 23:16:07.000000000","tz":180},"committer":{"name":"Anton Vazhnetsov","email":"AVazhnetsov@croc.ru","date":"2022-02-10 14:33:53.000000000","tz":180},"subject":"vtep: add support for \u0027hardware_vtep\u0027 schema","message":"vtep: add support for \u0027hardware_vtep\u0027 schema\n\nAdded support for the \"hardware_vtep\" schema. The ``HwVtepApiIdlImpl`` class provides interaction with \"hw_vtep.db\".\n\nCloses-Bug: 1960513\n\nChange-Id: I37138e91f9dd3e42908ac1e7e7c123323177a8e8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/2cd1aa2c3189dadfe6246d18ef90d803ff0184c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/2cd1aa2c3189dadfe6246d18ef90d803ff0184c8"}]},"branch":"refs/heads/master"},"48f8398a4cd262182ff5e0ceaeb2b23ec0adb73d":{"kind":"REWORK","_number":3,"created":"2022-02-11 09:32:52.000000000","uploader":{"_account_id":33871,"name":"Anton Vazhnetsov","display_name":"Anton Vazhnetsov","email":"dragen15051@gmail.com","username":"0x5b"},"ref":"refs/changes/20/828720/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/20/828720/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/20/828720/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/20/828720/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/20/828720/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/20/828720/3"}}},"commit":{"parents":[{"commit":"dc15bfd5729eb9f2b404288a4aa901c318274507","subject":"Log what exception caused abort of the transaction","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/dc15bfd5729eb9f2b404288a4aa901c318274507"}]}],"author":{"name":"Anton Vazhnetsov","email":"AVazhnetsov@croc.ru","date":"2022-02-06 23:16:07.000000000","tz":180},"committer":{"name":"Anton Vazhnetsov","email":"AVazhnetsov@croc.ru","date":"2022-02-11 09:32:04.000000000","tz":180},"subject":"vtep: add support for \u0027hardware_vtep\u0027 schema","message":"vtep: add support for \u0027hardware_vtep\u0027 schema\n\nAdded support for the \"hardware_vtep\" schema. The ``HwVtepApiIdlImpl`` class provides interaction with \"hw_vtep.db\".\n\nCloses-Bug: 1960513\n\nChange-Id: I37138e91f9dd3e42908ac1e7e7c123323177a8e8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/48f8398a4cd262182ff5e0ceaeb2b23ec0adb73d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/48f8398a4cd262182ff5e0ceaeb2b23ec0adb73d"}]},"branch":"refs/heads/master"},"ba7f334fbfb686c2a58d9a7816cf453c4f72e6cd":{"kind":"REWORK","_number":4,"created":"2022-03-04 11:54:18.000000000","uploader":{"_account_id":33871,"name":"Anton Vazhnetsov","display_name":"Anton Vazhnetsov","email":"dragen15051@gmail.com","username":"0x5b"},"ref":"refs/changes/20/828720/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/20/828720/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/20/828720/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/20/828720/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/20/828720/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/20/828720/4"}}},"commit":{"parents":[{"commit":"dc15bfd5729eb9f2b404288a4aa901c318274507","subject":"Log what exception caused abort of the transaction","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/dc15bfd5729eb9f2b404288a4aa901c318274507"}]}],"author":{"name":"Anton Vazhnetsov","email":"AVazhnetsov@croc.ru","date":"2022-02-06 23:16:07.000000000","tz":180},"committer":{"name":"Anton Vazhnetsov","email":"dragen15051@gmail.com","date":"2022-03-04 11:50:04.000000000","tz":180},"subject":"vtep: add support for \u0027hardware_vtep\u0027 schema","message":"vtep: add support for \u0027hardware_vtep\u0027 schema\n\nAdded support for the \"hardware_vtep\" schema. The ``HwVtepApiIdlImpl`` class provides interaction with \"hw_vtep.db\".\n\nCloses-Bug: 1960513\n\nChange-Id: I37138e91f9dd3e42908ac1e7e7c123323177a8e8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/ba7f334fbfb686c2a58d9a7816cf453c4f72e6cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/ba7f334fbfb686c2a58d9a7816cf453c4f72e6cd"}]},"branch":"refs/heads/master"},"9b699eea902f5216a6baec63479af7f1e22edb35":{"kind":"REWORK","_number":5,"created":"2022-03-05 22:00:57.000000000","uploader":{"_account_id":33871,"name":"Anton Vazhnetsov","display_name":"Anton Vazhnetsov","email":"dragen15051@gmail.com","username":"0x5b"},"ref":"refs/changes/20/828720/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/20/828720/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/20/828720/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/20/828720/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/20/828720/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/20/828720/5"}}},"commit":{"parents":[{"commit":"04d1ca0ab6376f93197d3fbb98d431fbf8c0ea59","subject":"Merge \"remove unicode from code\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/04d1ca0ab6376f93197d3fbb98d431fbf8c0ea59"}]}],"author":{"name":"Anton Vazhnetsov","email":"AVazhnetsov@croc.ru","date":"2022-02-06 23:16:07.000000000","tz":180},"committer":{"name":"Anton Vazhnetsov","email":"dragen15051@gmail.com","date":"2022-03-05 21:58:52.000000000","tz":180},"subject":"vtep: add support for \u0027hardware_vtep\u0027 schema","message":"vtep: add support for \u0027hardware_vtep\u0027 schema\n\nAdded support for the \"hardware_vtep\" schema. The ``HwVtepApiIdlImpl`` class provides interaction with \"hw_vtep.db\".\n\nCloses-Bug: 1960513\n\nChange-Id: I37138e91f9dd3e42908ac1e7e7c123323177a8e8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/9b699eea902f5216a6baec63479af7f1e22edb35"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/9b699eea902f5216a6baec63479af7f1e22edb35"}]},"branch":"refs/heads/master"},"ce79424d4c63eb6cfcb63e977876c00fa2ca8646":{"kind":"REWORK","_number":6,"created":"2022-03-28 09:18:00.000000000","uploader":{"_account_id":33871,"name":"Anton Vazhnetsov","display_name":"Anton Vazhnetsov","email":"dragen15051@gmail.com","username":"0x5b"},"ref":"refs/changes/20/828720/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/20/828720/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/20/828720/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/20/828720/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/20/828720/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/20/828720/6"}}},"commit":{"parents":[{"commit":"04d1ca0ab6376f93197d3fbb98d431fbf8c0ea59","subject":"Merge \"remove unicode from code\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/04d1ca0ab6376f93197d3fbb98d431fbf8c0ea59"}]}],"author":{"name":"Anton Vazhnetsov","email":"AVazhnetsov@croc.ru","date":"2022-02-06 23:16:07.000000000","tz":180},"committer":{"name":"Anton Vazhnetsov","email":"dragen15051@gmail.com","date":"2022-03-28 09:17:46.000000000","tz":180},"subject":"vtep: add support for \u0027hardware_vtep\u0027 schema","message":"vtep: add support for \u0027hardware_vtep\u0027 schema\n\nAdded support for the \"hardware_vtep\" schema. The ``HwVtepApiIdlImpl`` class provides interaction with \"hw_vtep.db\".\n\nCloses-Bug: 1960513\n\nChange-Id: I37138e91f9dd3e42908ac1e7e7c123323177a8e8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/ce79424d4c63eb6cfcb63e977876c00fa2ca8646"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/ce79424d4c63eb6cfcb63e977876c00fa2ca8646"}]},"branch":"refs/heads/master"},"06d25c29800dabc66d6727ef1161bebec7d21ad1":{"kind":"REWORK","_number":7,"created":"2022-03-30 12:09:14.000000000","uploader":{"_account_id":33871,"name":"Anton Vazhnetsov","display_name":"Anton Vazhnetsov","email":"dragen15051@gmail.com","username":"0x5b"},"ref":"refs/changes/20/828720/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/20/828720/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/20/828720/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/20/828720/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/20/828720/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/20/828720/7"}}},"commit":{"parents":[{"commit":"06207b33e3317b984868aeb67aaba4bb8423e710","subject":"Fix broken behavior after recent OVS update","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/06207b33e3317b984868aeb67aaba4bb8423e710"}]}],"author":{"name":"Anton Vazhnetsov","email":"AVazhnetsov@croc.ru","date":"2022-02-06 23:16:07.000000000","tz":180},"committer":{"name":"Anton Vazhnetsov","email":"dragen15051@gmail.com","date":"2022-03-30 12:05:19.000000000","tz":180},"subject":"vtep: add support for \u0027hardware_vtep\u0027 schema","message":"vtep: add support for \u0027hardware_vtep\u0027 schema\n\nAdded support for the \"hardware_vtep\" schema. The ``HwVtepApiIdlImpl`` class provides interaction with \"hw_vtep.db\".\n\nCloses-Bug: 1960513\n\nChange-Id: I37138e91f9dd3e42908ac1e7e7c123323177a8e8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/06d25c29800dabc66d6727ef1161bebec7d21ad1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/06d25c29800dabc66d6727ef1161bebec7d21ad1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}}]}],"submit_requirements":[]}
