)]}'
{"id":"openstack%2Fovsdbapp~834544","triplet_id":"openstack%2Fovsdbapp~master~Iad0d7472a7adce4a79111f94a2f33fc5b1a5c530","project":"openstack/ovsdbapp","branch":"master","topic":"bug/1965819","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2022-03-23 13:25:08.000000000","reason":"Rodolfo Alonso replied on the change"},"5756":{"account":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"last_update":"2022-03-23 14:44:18.000000000","reason":"Change was submitted"},"8655":{"account":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"last_update":"2022-03-22 20:32:43.000000000","reason":"Jakub Libosvar replied on the change"}},"hashtags":[],"change_id":"Iad0d7472a7adce4a79111f94a2f33fc5b1a5c530","subject":"Fix broken behavior after recent OVS update","status":"MERGED","created":"2022-03-21 20:12:05.000000000","updated":"2022-03-23 14:45:55.000000000","submitted":"2022-03-23 14:44:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"834544-bug/1965819","meta_rev_id":"43d7862e95e258a2aac33994af14cf566fdeddcd","_number":834544,"virtual_id_number":834544,"owner":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-23 14:44:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2022-03-23 13:25:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":2,"date":"2022-03-22 20:32:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2022-03-23 13:25:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-21 23:00:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-22 04:09:39.000000000","updated_by":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2022-03-22 04:09:39.000000000","updated_by":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"reviewer":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"state":"REVIEWER"}],"messages":[{"id":"ca141224a3fcbeac5989190fcdd5f20a5de5bbff","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2022-03-21 20:12:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0ac224d284acc41109e9c85ed4d54e34d7b6ab4c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2022-03-21 22:02:46.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"d84c6157e3a3658858c212e4d2d3d365400abb65","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-21 23:00:11.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/af649baad61c4be296a14944479eb402 : SUCCESS in 6m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/00623811232e4e7c9fa1efa44fc0e25c : SUCCESS in 5m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/428e54c45c2b4422addce373bf309fcf : SUCCESS in 6m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/56e646279b9341ccb7c1f54391301d4b : SUCCESS in 6m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f96c5e28236944c5b8651b309474e05a : SUCCESS in 7m 29s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/9eea52bcc57e4541aa445ae9435f1f70 : SUCCESS in 8m 36s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/73f4521a6ae74d11bb2568bcbf92a2e4 : SUCCESS in 55m 18s","accounts_in_message":[],"_revision_number":2},{"id":"948ba6a29f8bf79d070c939f6de3e246b71c8e1d","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2022-03-21 23:28:47.000000000","message":"Patch Set 2: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7f2df9cc07c39357945b71703059b4d03e616590","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2022-03-22 04:03:01.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"7c0c37dc1e3e70dd41dc60d02a104bd7082cdabf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2022-03-22 04:10:05.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"3748b30d0c911867a9e9719bc8905d95eb87e039","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2022-03-22 04:11:21.000000000","message":"Topic set to bug/1965819","accounts_in_message":[],"_revision_number":4},{"id":"329d908ef4703f27a08405c73385d556c141c437","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-22 05:05:50.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a8abda3257974722b2e3f18657426a4a : SUCCESS in 5m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a66523d078c84f79be61be9c89d36a7b : SUCCESS in 5m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/35fb6310ffa641da89a0bb619d028196 : SUCCESS in 5m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/23b6b11815d949aca644acb1ee5689c1 : SUCCESS in 5m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c117336d5144b6d9d33543bd839cf77 : SUCCESS in 7m 28s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/785aa325dd1d44eca214f2d6ebe70474 : SUCCESS in 7m 52s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/a31fbe902e2c484b9ac2db0da7f445b4 : SUCCESS in 54m 07s","accounts_in_message":[],"_revision_number":4},{"id":"6a5462d319a5a47f3af381786e55f885cea0d566","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-03-22 15:00:51.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"10cbdbc07b77f5531c54e74f2097760927bddb6f","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2022-03-22 15:44:20.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"af02ba558508b3305077c24761fa849f4b4b2223","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2022-03-22 16:07:53.000000000","message":"Patch Set 4: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"a5ff49ef1795d717cb296d5d37ef71fc7bdf40c9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2022-03-22 18:55:50.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"31445ba9e6861d3deadba5c66b0fd1ff2141f780","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2022-03-22 19:04:01.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"ab3c9444b5316dbd7c0d8e3b8e2e35541d04e020","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-22 20:04:04.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fd02ee6361ff4700a1dd374fca69a73c : SUCCESS in 5m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5e3a3f7f7fdb4b7b933ea9be8e9752ef : SUCCESS in 5m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bbe11aa623f644fea63661b45e32e573 : SUCCESS in 5m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/566cebb91b7d42ca90466024fe6671b8 : SUCCESS in 5m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7fb1702d245d436b88ae727d1696e715 : SUCCESS in 7m 12s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/40fdafacb65e433bb1680026c4db53ca : SUCCESS in 7m 56s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/e1ed14a498fb4d48ae0fad9cf4ea14e9 : SUCCESS in 1h 05m 36s","accounts_in_message":[],"_revision_number":5},{"id":"b0cab8a72f7c14df239f5cff87def9079fc7c994","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2022-03-22 20:32:43.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"cba612689c7b4a25e21d8aac7969cdadf7d9360c","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-03-23 13:25:08.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"fb1ac7211a33306ed077875630dcf2f58f192dd0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-23 13:25:50.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"ef86887ba6b935e8d637c5f9e3ac3df98b22775d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-23 14:44:18.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7147cccbfe9742a5a9739dd8e3706716 : SUCCESS in 5m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d805fade45394c17924a20c427979cbc : SUCCESS in 5m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6e201575301249aeb871b908e5ca70a6 : SUCCESS in 6m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/530c212393ab46f3b9828d44916d2b18 : SUCCESS in 4m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f96b91e93b34b2bba2c8d6830363ec4 : SUCCESS in 6m 55s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/d5d8ca50646e4c778bec5733b1894c68 : SUCCESS in 8m 10s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/81876127025846fc8d54ad06338b5b8c : SUCCESS in 1h 14m 15s","accounts_in_message":[],"_revision_number":5},{"id":"c43dc7980a3cde65a90d5c5875e54ce6a9363bc7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-23 14:44:18.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"43d7862e95e258a2aac33994af14cf566fdeddcd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-23 14:45:55.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/acb9a3ffd7714e95bdd814b81565ca96 : SUCCESS in 1m 03s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"06207b33e3317b984868aeb67aaba4bb8423e710","revisions":{"2ca645c0daaa6a53e1a508494f57e9087ae7e11a":{"kind":"REWORK","_number":1,"created":"2022-03-21 20:12:05.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/44/834544/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/44/834544/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/44/834544/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/44/834544/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/44/834544/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/44/834544/1"}}},"commit":{"parents":[{"commit":"dd66e2762ec7486a86d6938369eaaf6e79037db8","subject":"Update master for stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/dd66e2762ec7486a86d6938369eaaf6e79037db8"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2022-03-21 19:36:42.000000000","tz":0},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2022-03-21 20:12:04.000000000","tz":0},"subject":"Fix broken behavior after recent OVS update","message":"Fix broken behavior after recent OVS update\n\nAfter the monitor_cond_since/update3 support patch in ovs\n(46d44cf3be0), directly setting table.condition is broken. This\nisn\u0027t something that was every truly supported. Ultimately, it\nwould be good if we could make this code less dependent on exactly\nwhat happens in Idl.__init__()--maybe adding an update_tables()\nmethod in python-ovs.\n\nCloses-Bug: 1965819\nChange-Id: Iad0d7472a7adce4a79111f94a2f33fc5b1a5c530\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/2ca645c0daaa6a53e1a508494f57e9087ae7e11a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/2ca645c0daaa6a53e1a508494f57e9087ae7e11a"}]},"branch":"refs/heads/master"},"20e209a9c68a2d7a527eb7da1e6d604d09fb6a60":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-03-21 22:02:46.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/44/834544/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/44/834544/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/44/834544/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/44/834544/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/44/834544/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/44/834544/2"}}},"commit":{"parents":[{"commit":"7d1f3180bd7900f4becc75de44d7fd2a8eba9692","subject":"github no longer accepts git:// connections","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/7d1f3180bd7900f4becc75de44d7fd2a8eba9692"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2022-03-21 19:36:42.000000000","tz":0},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2022-03-21 22:02:46.000000000","tz":0},"subject":"Fix broken behavior after recent OVS update","message":"Fix broken behavior after recent OVS update\n\nAfter the monitor_cond_since/update3 support patch in ovs\n(46d44cf3be0), directly setting table.condition is broken. This\nisn\u0027t something that was every truly supported. Ultimately, it\nwould be good if we could make this code less dependent on exactly\nwhat happens in Idl.__init__()--maybe adding an update_tables()\nmethod in python-ovs.\n\nCloses-Bug: 1965819\nChange-Id: Iad0d7472a7adce4a79111f94a2f33fc5b1a5c530\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/20e209a9c68a2d7a527eb7da1e6d604d09fb6a60"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/20e209a9c68a2d7a527eb7da1e6d604d09fb6a60"}]},"branch":"refs/heads/master","description":"Rebase"},"6619784c0439204aeec1eceed09f7d87e3740c61":{"kind":"REWORK","_number":3,"created":"2022-03-22 04:03:01.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/44/834544/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/44/834544/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/44/834544/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/44/834544/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/44/834544/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/44/834544/3"}}},"commit":{"parents":[{"commit":"7d1f3180bd7900f4becc75de44d7fd2a8eba9692","subject":"github no longer accepts git:// connections","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/7d1f3180bd7900f4becc75de44d7fd2a8eba9692"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2022-03-21 19:36:42.000000000","tz":0},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2022-03-22 04:00:29.000000000","tz":0},"subject":"Fix broken behavior after recent OVS update","message":"Fix broken behavior after recent OVS update\n\nAfter the monitor_cond_since/update3 support patch in ovs\n(46d44cf3be0), directly setting table.condition is broken. This\nisn\u0027t something that was every truly supported. Ultimately, it\nwould be good if we could make this code less dependent on exactly\nwhat happens in Idl.__init__()--maybe adding an update_tables()\nmethod in python-ovs.\n\nTo avoid stomping on the state stored in ConditionState, instead\nof replacing Idl.tables, this removes deletions and adds new\ntables were passed. Existing tables are left alone.\n\nCloses-Bug: 1965819\nChange-Id: Iad0d7472a7adce4a79111f94a2f33fc5b1a5c530\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/6619784c0439204aeec1eceed09f7d87e3740c61"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/6619784c0439204aeec1eceed09f7d87e3740c61"}]},"branch":"refs/heads/master"},"128db494b7f069ce6d422df3e62832b3f3c89c49":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2022-03-22 04:10:05.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/44/834544/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/44/834544/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/44/834544/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/44/834544/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/44/834544/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/44/834544/4"}}},"commit":{"parents":[{"commit":"7d1f3180bd7900f4becc75de44d7fd2a8eba9692","subject":"github no longer accepts git:// connections","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/7d1f3180bd7900f4becc75de44d7fd2a8eba9692"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2022-03-21 19:36:42.000000000","tz":0},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2022-03-22 04:10:05.000000000","tz":0},"subject":"Fix broken behavior after recent OVS update","message":"Fix broken behavior after recent OVS update\n\nAfter the monitor_cond_since/update3 support patch in ovs\n(46d44cf3be0), directly setting table.condition is broken. This\nisn\u0027t something that was every truly supported. Ultimately, it\nwould be good if we could make this code less dependent on exactly\nwhat happens in Idl.__init__()--maybe adding an update_tables()\nmethod in python-ovs.\n\nTo avoid stomping on the state stored in ConditionState, instead\nof replacing Idl.tables, this removes deletions and adds new\ntables that were passed. Existing tables are left alone.\n\nCloses-Bug: 1965819\nChange-Id: Iad0d7472a7adce4a79111f94a2f33fc5b1a5c530\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/128db494b7f069ce6d422df3e62832b3f3c89c49"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/128db494b7f069ce6d422df3e62832b3f3c89c49"}]},"branch":"refs/heads/master","description":"Edit commit message"},"06207b33e3317b984868aeb67aaba4bb8423e710":{"kind":"REWORK","_number":5,"created":"2022-03-22 18:55:50.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/44/834544/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/44/834544/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/44/834544/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/44/834544/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/44/834544/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/44/834544/5"}}},"commit":{"parents":[{"commit":"7d1f3180bd7900f4becc75de44d7fd2a8eba9692","subject":"github no longer accepts git:// connections","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/7d1f3180bd7900f4becc75de44d7fd2a8eba9692"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2022-03-21 19:36:42.000000000","tz":0},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2022-03-22 18:55:43.000000000","tz":0},"subject":"Fix broken behavior after recent OVS update","message":"Fix broken behavior after recent OVS update\n\nAfter the monitor_cond_since/update3 support patch in ovs\n(46d44cf3be0), directly setting table.condition is broken. This\nisn\u0027t something that was every truly supported. Ultimately, it\nwould be good if we could make this code less dependent on exactly\nwhat happens in Idl.__init__()--maybe adding an update_tables()\nmethod in python-ovs.\n\nTo avoid stomping on the state stored in ConditionState, instead\nof replacing Idl.tables, this removes deletions and adds new\ntables that were passed. Existing tables are left alone.\n\nCloses-Bug: 1965819\nChange-Id: Iad0d7472a7adce4a79111f94a2f33fc5b1a5c530\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/06207b33e3317b984868aeb67aaba4bb8423e710"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/06207b33e3317b984868aeb67aaba4bb8423e710"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
