)]}'
{"id":"openstack%2Fovsdbapp~834727","triplet_id":"openstack%2Fovsdbapp~stable%2Fyoga~Iad0d7472a7adce4a79111f94a2f33fc5b1a5c530","project":"openstack/ovsdbapp","branch":"stable/yoga","topic":"bug/1965819-stable/yoga","attention_set":{},"removed_from_attention_set":{"5756":{"account":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"last_update":"2022-03-30 14:48:07.000000000","reason":"Terry Wilson replied on the change"}},"hashtags":[],"change_id":"Iad0d7472a7adce4a79111f94a2f33fc5b1a5c530","subject":"Fix broken behavior after recent OVS update","status":"MERGED","created":"2022-03-23 15:28:46.000000000","updated":"2022-03-30 15:53:41.000000000","submitted":"2022-03-30 15:52:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"834727-bug/1965819-stable/yoga","cherry_pick_of_change":834544,"cherry_pick_of_patch_set":5,"meta_rev_id":"9050a222de1b91a389fdb21079eccb13a0128c44","_number":834727,"virtual_id_number":834727,"owner":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-30 15:52:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":2,"date":"2022-03-25 15:50:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-03-30 14:48:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-23 15:39:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-23 17:59:46.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"CC"},{"updated":"2022-03-25 15:50:15.000000000","updated_by":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"reviewer":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"state":"REVIEWER"}],"messages":[{"id":"8ab36e74e6cc171dff0f0031ed9dad86d69271fc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2022-03-23 15:28:46.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"ba95504674af559e61c3800454e01e9bfff26bdb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-23 15:39:00.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f15c8400d3494ec9aee4713441f6391d : SUCCESS in 5m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6beb7c76871c4388b2ae6824e3e0777f : SUCCESS in 5m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d9194d9c1ba144dcb8c82398fbe1accc : SUCCESS in 5m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/613452e7f06c4bc1aecda614a54fe1e4 : SUCCESS in 5m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/953b51ec35924b958b2010d5f4a50aa7 : SUCCESS in 7m 26s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/5bccec3508f8473f806045db44b26182 : FAILURE in 5m 30s","accounts_in_message":[],"_revision_number":1},{"id":"05d4c94e53b0a651ef8cc65f9e2314b1b2fddcbc","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-03-23 17:59:46.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"761f7a53e81e040c3ac484a40a78f031816a7d92","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-23 18:15:45.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c1d9f01c4db8425db9e79a0b79e858eb : SUCCESS in 5m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/125cd7c0ca5047cb9eff1ad66e075549 : SUCCESS in 4m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/236dc7216b794f949024006cb938d7c4 : SUCCESS in 4m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cd76cf8dba9840f19ed7aaba73725b77 : SUCCESS in 4m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6e89ee2d1f849ca862b65c230cdd8c2 : SUCCESS in 6m 24s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/04f62fced8794ba2bbb9ffed2a65ca45 : FAILURE in 4m 40s","accounts_in_message":[],"_revision_number":1},{"id":"bedbf35efa0de5e1d911e513a715b093cd440800","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2022-03-24 20:11:43.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"8009ff2fbcd144e81b3d555c221aab42a9d44127","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-24 20:24:47.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ceb7c35a8c5b4b2883d247018c0157f7 : SUCCESS in 5m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/deef35c29b254e5481ea31bc0440e694 : SUCCESS in 5m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/adfe044fe1aa420da571b2f50f102ef5 : SUCCESS in 5m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3998671983fa45d5bdf6182600f85dc7 : SUCCESS in 5m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7665661fbdfd44ab8fef1c50326d8bed : SUCCESS in 6m 38s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/f31847ae63c44535b5d9b4dea6cfc912 : SUCCESS in 7m 48s","accounts_in_message":[],"_revision_number":2},{"id":"87f95844d3a92b42feb1bc43d85a29dbe78f32c5","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2022-03-25 15:50:15.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"3e36c59858fdcbb03d23d225e7c5473b57e98a5c","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2022-03-30 14:48:07.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"421c080682e5a4322c7cc9240ee2dc56abfc2d76","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-30 14:48:45.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"36192d40ecf6a671c141b5c4cb49815376aa472b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-30 15:52:09.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/322bd8a07ca24f7b88d19278577199d9 : SUCCESS in 3m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/56fa12db6da8469fa23a663491b694da : SUCCESS in 3m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eb25110848064ea5860d8d42ca2bd967 : SUCCESS in 4m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4bfe80024d2849fba73ed3ad895976ab : SUCCESS in 4m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca0b737a268d4d6aa87766e13ef8db1b : SUCCESS in 4m 53s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/53d38426a6d1434dac32b722400e19c2 : SUCCESS in 6m 59s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/10c7534f81134a48a9786a968704df6b : SUCCESS in 58m 26s","accounts_in_message":[],"_revision_number":2},{"id":"6637b0b5e79ae8c2dec92bfd9d7f29ca281c8e13","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-30 15:52:10.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"9050a222de1b91a389fdb21079eccb13a0128c44","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-30 15:53:41.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b3e0d23bdba94098855b8fb77c27fc71 : SUCCESS in 1m 08s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"de473cea9defe04a6f904fa63a8cf49746534916","revisions":{"619b0e1798ed4033e57c78b13bf82acd6a2340ca":{"kind":"REWORK","_number":1,"created":"2022-03-23 15:28:46.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/27/834727/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/27/834727/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/27/834727/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/27/834727/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/27/834727/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/27/834727/1"}}},"commit":{"parents":[{"commit":"7735d5cbe13981228568934765284eb97e1d533b","subject":"Update TOX_CONSTRAINTS_FILE for stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/7735d5cbe13981228568934765284eb97e1d533b"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2022-03-21 19:36:42.000000000","tz":0},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2022-03-23 15:28:46.000000000","tz":0},"subject":"Fix broken behavior after recent OVS update","message":"Fix broken behavior after recent OVS update\n\nAfter the monitor_cond_since/update3 support patch in ovs\n(46d44cf3be0), directly setting table.condition is broken. This\nisn\u0027t something that was every truly supported. Ultimately, it\nwould be good if we could make this code less dependent on exactly\nwhat happens in Idl.__init__()--maybe adding an update_tables()\nmethod in python-ovs.\n\nTo avoid stomping on the state stored in ConditionState, instead\nof replacing Idl.tables, this removes deletions and adds new\ntables that were passed. Existing tables are left alone.\n\nCloses-Bug: 1965819\nChange-Id: Iad0d7472a7adce4a79111f94a2f33fc5b1a5c530\n(cherry picked from commit 06207b33e3317b984868aeb67aaba4bb8423e710)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/619b0e1798ed4033e57c78b13bf82acd6a2340ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/619b0e1798ed4033e57c78b13bf82acd6a2340ca"}]},"branch":"refs/heads/stable/yoga"},"de473cea9defe04a6f904fa63a8cf49746534916":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-03-24 20:11:43.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/27/834727/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/27/834727/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/27/834727/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/27/834727/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/27/834727/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/27/834727/2"}}},"commit":{"parents":[{"commit":"10539ed9becbadfd6a727cd82713442e63969cd5","subject":"github no longer accepts git:// connections","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/10539ed9becbadfd6a727cd82713442e63969cd5"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2022-03-21 19:36:42.000000000","tz":0},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2022-03-24 20:11:43.000000000","tz":0},"subject":"Fix broken behavior after recent OVS update","message":"Fix broken behavior after recent OVS update\n\nAfter the monitor_cond_since/update3 support patch in ovs\n(46d44cf3be0), directly setting table.condition is broken. This\nisn\u0027t something that was every truly supported. Ultimately, it\nwould be good if we could make this code less dependent on exactly\nwhat happens in Idl.__init__()--maybe adding an update_tables()\nmethod in python-ovs.\n\nTo avoid stomping on the state stored in ConditionState, instead\nof replacing Idl.tables, this removes deletions and adds new\ntables that were passed. Existing tables are left alone.\n\nCloses-Bug: 1965819\nChange-Id: Iad0d7472a7adce4a79111f94a2f33fc5b1a5c530\n(cherry picked from commit 06207b33e3317b984868aeb67aaba4bb8423e710)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/de473cea9defe04a6f904fa63a8cf49746534916"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/de473cea9defe04a6f904fa63a8cf49746534916"}]},"branch":"refs/heads/stable/yoga","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
