)]}'
{"id":"openstack%2Fovsdbapp~834728","triplet_id":"openstack%2Fovsdbapp~stable%2Fxena~Iad0d7472a7adce4a79111f94a2f33fc5b1a5c530","project":"openstack/ovsdbapp","branch":"stable/xena","topic":"bug/1965819-stable/xena","attention_set":{},"removed_from_attention_set":{"5756":{"account":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"last_update":"2022-03-30 14:48:19.000000000","reason":"Terry Wilson replied on the change"}},"hashtags":[],"change_id":"Iad0d7472a7adce4a79111f94a2f33fc5b1a5c530","subject":"Fix broken behavior after recent OVS update","status":"MERGED","created":"2022-03-23 15:29:06.000000000","updated":"2022-03-30 16:02:48.000000000","submitted":"2022-03-30 16:01:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"834728-bug/1965819-stable/xena","cherry_pick_of_change":834544,"cherry_pick_of_patch_set":5,"meta_rev_id":"ff24b37aa9b427d4b67aec4dafce35f49fe8f4b5","_number":834728,"virtual_id_number":834728,"owner":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-30 16:01:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":2,"date":"2022-03-25 15:50:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-03-30 14:48:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-23 16:29:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-25 15:50:19.000000000","updated_by":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"reviewer":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"state":"REVIEWER"}],"messages":[{"id":"74155cd3dfa84ce5fa01b983f632627936b92b72","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2022-03-23 15:29:06.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"b5c31ccd2e8977c67a4056721768eb806c0029fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-23 16:29:18.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/301f0cd7ae38487eb8ebc8153f31ba1e : SUCCESS in 5m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6eb7a249fb314f71b7aece57f25012f8 : SUCCESS in 5m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4a07800dc4aa46edbcc8da9a22cf3e4a : SUCCESS in 5m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ef47e3b19c3242598d56991197feebac : SUCCESS in 5m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/db4dd3faa39b46ac98d83c8ef9ac542b : SUCCESS in 5m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f92ee9800644684a0b9818f394fa422 : SUCCESS in 7m 07s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/8165e9b01148408ba9c9def140721773 : FAILURE in 5m 20s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/c32a1d00bd8b4a5aa8a2c72bb6345d6c : SUCCESS in 56m 52s","accounts_in_message":[],"_revision_number":1},{"id":"dbdfe28f56945a261911621ec90d3a63bd851219","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2022-03-24 19:26:10.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"2ca94bdc600e3eaa3de12b85aed1687bc24ffff4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-24 20:31:02.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/31b30c83aa574316a6e6d4d33eaf845c : SUCCESS in 5m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9302fae7a31541d8b86f099e436b4ab2 : SUCCESS in 5m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/671760f865e048068e6e257192a683be : SUCCESS in 5m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e3f08216e3284c18a934f0c16be5fe0d : SUCCESS in 5m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/772a5d1bee334e68b5337ca6ce343a65 : SUCCESS in 5m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a506d426f2864a628b2f21ebbf475bad : SUCCESS in 6m 45s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/5794cff6e6474ac7954dde4c0c4ada15 : SUCCESS in 9m 20s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/9118322d5c99455d862c4e92fb840893 : SUCCESS in 59m 48s","accounts_in_message":[],"_revision_number":2},{"id":"eeb902def7eb9268133e723ab8bb48374e4ce811","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2022-03-25 15:50:19.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"66940f630fe559add4baa9ae1c8770026d9d1c47","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2022-03-30 14:48:19.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"98bec301876974f03c98bd176ed2b68ee8de5f59","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-30 14:48:58.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"febfd11179ee08bf67d4ba5af78eb4573207f9ac","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-30 16:01:10.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bcec014f646e4b388c5d23089c78b5a3 : SUCCESS in 4m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f3511c5d7de42d3b09bf73c38cb9eab : SUCCESS in 3m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aaab959104ef4a06ae213327bd727473 : SUCCESS in 3m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4c02a769306f45c694a2a2e6ca5eb796 : SUCCESS in 3m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/557246a903d549049e67dbe91d698987 : SUCCESS in 6m 09s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/6de6732893454a6cab24018b5431027f : SUCCESS in 6m 35s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/3e4b9cba67e2434eb0927be8b15a25fc : SUCCESS in 1h 05m 39s","accounts_in_message":[],"_revision_number":2},{"id":"4696894b0daa36d9ab04fd2a91c8f5e8e936e2f8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-30 16:01:10.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"ff24b37aa9b427d4b67aec4dafce35f49fe8f4b5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-30 16:02:48.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/48fe99de94d64aefa3b3135439d40d05 : SUCCESS in 1m 17s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a7a1d5cfd3bded883c982f646758d2ab23cfb2c7","revisions":{"7bac9e8555678469dc2eafb9690acaac7edf1838":{"kind":"REWORK","_number":1,"created":"2022-03-23 15:29:06.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/28/834728/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/28/834728/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/28/834728/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/28/834728/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/28/834728/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/28/834728/1"}}},"commit":{"parents":[{"commit":"3a0cd11c6517c227cbc123dacecd884a2bef657d","subject":"Log what exception caused abort of the transaction","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/3a0cd11c6517c227cbc123dacecd884a2bef657d"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2022-03-21 19:36:42.000000000","tz":0},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2022-03-23 15:29:06.000000000","tz":0},"subject":"Fix broken behavior after recent OVS update","message":"Fix broken behavior after recent OVS update\n\nAfter the monitor_cond_since/update3 support patch in ovs\n(46d44cf3be0), directly setting table.condition is broken. This\nisn\u0027t something that was every truly supported. Ultimately, it\nwould be good if we could make this code less dependent on exactly\nwhat happens in Idl.__init__()--maybe adding an update_tables()\nmethod in python-ovs.\n\nTo avoid stomping on the state stored in ConditionState, instead\nof replacing Idl.tables, this removes deletions and adds new\ntables that were passed. Existing tables are left alone.\n\nCloses-Bug: 1965819\nChange-Id: Iad0d7472a7adce4a79111f94a2f33fc5b1a5c530\n(cherry picked from commit 06207b33e3317b984868aeb67aaba4bb8423e710)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/7bac9e8555678469dc2eafb9690acaac7edf1838"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/7bac9e8555678469dc2eafb9690acaac7edf1838"}]},"branch":"refs/heads/stable/xena"},"a7a1d5cfd3bded883c982f646758d2ab23cfb2c7":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-03-24 19:26:10.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/28/834728/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/28/834728/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/28/834728/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/28/834728/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/28/834728/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/28/834728/2"}}},"commit":{"parents":[{"commit":"a73fbf1e5ed430f6487c282b1278d94c65f5b78a","subject":"github no longer accepts git:// connections","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/a73fbf1e5ed430f6487c282b1278d94c65f5b78a"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2022-03-21 19:36:42.000000000","tz":0},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2022-03-24 19:26:10.000000000","tz":0},"subject":"Fix broken behavior after recent OVS update","message":"Fix broken behavior after recent OVS update\n\nAfter the monitor_cond_since/update3 support patch in ovs\n(46d44cf3be0), directly setting table.condition is broken. This\nisn\u0027t something that was every truly supported. Ultimately, it\nwould be good if we could make this code less dependent on exactly\nwhat happens in Idl.__init__()--maybe adding an update_tables()\nmethod in python-ovs.\n\nTo avoid stomping on the state stored in ConditionState, instead\nof replacing Idl.tables, this removes deletions and adds new\ntables that were passed. Existing tables are left alone.\n\nCloses-Bug: 1965819\nChange-Id: Iad0d7472a7adce4a79111f94a2f33fc5b1a5c530\n(cherry picked from commit 06207b33e3317b984868aeb67aaba4bb8423e710)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/a7a1d5cfd3bded883c982f646758d2ab23cfb2c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/a7a1d5cfd3bded883c982f646758d2ab23cfb2c7"}]},"branch":"refs/heads/stable/xena","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
