)]}'
{"id":"openstack%2Fovsdbapp~834729","triplet_id":"openstack%2Fovsdbapp~stable%2Fwallaby~Iad0d7472a7adce4a79111f94a2f33fc5b1a5c530","project":"openstack/ovsdbapp","branch":"stable/wallaby","topic":"bug/1965819-stable/wallaby","attention_set":{},"removed_from_attention_set":{"5756":{"account":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"last_update":"2022-03-30 14:48:30.000000000","reason":"Terry Wilson replied on the change"}},"hashtags":[],"change_id":"Iad0d7472a7adce4a79111f94a2f33fc5b1a5c530","subject":"Fix broken behavior after recent OVS update","status":"MERGED","created":"2022-03-23 15:29:20.000000000","updated":"2022-03-30 16:33:05.000000000","submitted":"2022-03-30 16:31:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"834729-bug/1965819-stable/wallaby","cherry_pick_of_change":834544,"cherry_pick_of_patch_set":5,"meta_rev_id":"da7db4e98b0be6182f34edffb6488fa829c0e92f","_number":834729,"virtual_id_number":834729,"owner":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-30 16:31:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":2,"date":"2022-03-25 15:50:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-03-30 14:48:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-23 16:41:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-25 15:50:24.000000000","updated_by":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"reviewer":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"state":"REVIEWER"}],"messages":[{"id":"726355c4276158bda47253efbedb63a8f7862677","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2022-03-23 15:29:20.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"a5d705d62a347c5dd18c73963e04362c814e5c23","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-23 16:41:26.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3d766eefbbb34ecc9602529c473a4eb6 : SUCCESS in 5m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b49b4f37059241a1be98dd5c60eed921 : SUCCESS in 5m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0a2d86390cc5491dbd85d6ffb4245360 : SUCCESS in 6m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7f0933ce4d4a41a1a29af5d59e7fc2bc : SUCCESS in 5m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/88bb93540a4648c099fb1489e5c310e6 : SUCCESS in 5m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e2eb0fcdabf4ebfbce2f32a25fc3d84 : SUCCESS in 8m 18s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/c3a88f6e813444539dffc293e34c8fad : FAILURE in 5m 18s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/b4a675b3c3f64afcacd3fd2d7a59a4df : POST_FAILURE in 1h 06m 25s","accounts_in_message":[],"_revision_number":1},{"id":"f1b9f95ec0e1082da40b9fa73bdee42cc4abecab","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2022-03-24 19:26:14.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"df72f0f0a800ca5806ffd29e80377e36fae358fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-24 20:36:14.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/14d4c8d2562c4986a6c73001327da846 : SUCCESS in 4m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1131615826244d1f825211743fd89c7f : SUCCESS in 5m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c02330ce63604f5984cdf544094fed6d : SUCCESS in 4m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/90085f72f82d4e30844c460be5ea70c1 : SUCCESS in 5m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/47f77b30a0d541f4b4b3bc0952b69eab : SUCCESS in 5m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bcd81fbc7dd94596993d065f501c9e2c : SUCCESS in 6m 21s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/e7427880d32544bab18cdd62db61fd65 : SUCCESS in 7m 58s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/32f7a1c6284e40d1a62f8096eaa85b89 : SUCCESS in 1h 05m 05s","accounts_in_message":[],"_revision_number":2},{"id":"8a5baa01d7f4798fc195fcca67f6196708448a79","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2022-03-25 15:50:24.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"70d22a59c55283d0377d34f3e122f8e1e4535bcf","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2022-03-30 14:48:30.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"30cab7345dc3cc9d3f9a1fb1ad1eeac2a3ff9567","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-30 14:49:21.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"79e685cb60ad0613dc777af3c2064caeae02ace8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-30 16:31:11.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/633d9f1686f14cc396d942a583133e79 : SUCCESS in 4m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5cfe99e6043842c09b9e73239f7fe4d3 : SUCCESS in 3m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5f0af715519a49e3986f87a4abfb0095 : SUCCESS in 3m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/91d359f5de2e41f68ca948093d326883 : SUCCESS in 4m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/05f4588065d14563a3bc43e27940f019 : SUCCESS in 5m 10s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/600e3ac3c4734af39aa752fdcb3c3cc0 : SUCCESS in 6m 28s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/b48813edb1ee43d99cf7c133fbbb356e : SUCCESS in 1h 39m 42s","accounts_in_message":[],"_revision_number":2},{"id":"af5184ed55ad126dccc40b6c19b2c33c3d100819","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-30 16:31:11.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"da7db4e98b0be6182f34edffb6488fa829c0e92f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-30 16:33:05.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/884fa8ff7b2a475ebb261ca9d6a4a04e : SUCCESS in 1m 26s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"99905097ceab83a07af4a7980c870da0ed416cf2","revisions":{"eba5e90f40a29f8f05055b5ecccc10aff6a3552a":{"kind":"REWORK","_number":1,"created":"2022-03-23 15:29:20.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/29/834729/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/29/834729/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/29/834729/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/29/834729/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/29/834729/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/29/834729/1"}}},"commit":{"parents":[{"commit":"fa4a747f56d17713efdfc16cdba27e643d9ca3fa","subject":"Log what exception caused abort of the transaction","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/fa4a747f56d17713efdfc16cdba27e643d9ca3fa"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2022-03-21 19:36:42.000000000","tz":0},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2022-03-23 15:29:20.000000000","tz":0},"subject":"Fix broken behavior after recent OVS update","message":"Fix broken behavior after recent OVS update\n\nAfter the monitor_cond_since/update3 support patch in ovs\n(46d44cf3be0), directly setting table.condition is broken. This\nisn\u0027t something that was every truly supported. Ultimately, it\nwould be good if we could make this code less dependent on exactly\nwhat happens in Idl.__init__()--maybe adding an update_tables()\nmethod in python-ovs.\n\nTo avoid stomping on the state stored in ConditionState, instead\nof replacing Idl.tables, this removes deletions and adds new\ntables that were passed. Existing tables are left alone.\n\nCloses-Bug: 1965819\nChange-Id: Iad0d7472a7adce4a79111f94a2f33fc5b1a5c530\n(cherry picked from commit 06207b33e3317b984868aeb67aaba4bb8423e710)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/eba5e90f40a29f8f05055b5ecccc10aff6a3552a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/eba5e90f40a29f8f05055b5ecccc10aff6a3552a"}]},"branch":"refs/heads/stable/wallaby"},"99905097ceab83a07af4a7980c870da0ed416cf2":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-03-24 19:26:14.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/29/834729/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/29/834729/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/29/834729/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/29/834729/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/29/834729/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/29/834729/2"}}},"commit":{"parents":[{"commit":"fb5feeee05ea9233a6861852ad2efd3d22f4cae5","subject":"github no longer accepts git:// connections","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/fb5feeee05ea9233a6861852ad2efd3d22f4cae5"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2022-03-21 19:36:42.000000000","tz":0},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2022-03-24 19:26:14.000000000","tz":0},"subject":"Fix broken behavior after recent OVS update","message":"Fix broken behavior after recent OVS update\n\nAfter the monitor_cond_since/update3 support patch in ovs\n(46d44cf3be0), directly setting table.condition is broken. This\nisn\u0027t something that was every truly supported. Ultimately, it\nwould be good if we could make this code less dependent on exactly\nwhat happens in Idl.__init__()--maybe adding an update_tables()\nmethod in python-ovs.\n\nTo avoid stomping on the state stored in ConditionState, instead\nof replacing Idl.tables, this removes deletions and adds new\ntables that were passed. Existing tables are left alone.\n\nCloses-Bug: 1965819\nChange-Id: Iad0d7472a7adce4a79111f94a2f33fc5b1a5c530\n(cherry picked from commit 06207b33e3317b984868aeb67aaba4bb8423e710)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/99905097ceab83a07af4a7980c870da0ed416cf2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/99905097ceab83a07af4a7980c870da0ed416cf2"}]},"branch":"refs/heads/stable/wallaby","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
