)]}'
{"id":"openstack%2Fovsdbapp~835151","triplet_id":"openstack%2Fovsdbapp~master~I42072ecb6a658814c7ff48cbc6f2a37a91d1eba8","project":"openstack/ovsdbapp","branch":"master","attention_set":{},"removed_from_attention_set":{"5756":{"account":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"last_update":"2022-03-30 17:29:56.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I42072ecb6a658814c7ff48cbc6f2a37a91d1eba8","subject":"Revert \"Add an active wait in the \"Backend.lookup\"\"","status":"MERGED","created":"2022-03-24 21:51:33.000000000","updated":"2022-03-30 17:31:32.000000000","submitted":"2022-03-30 17:29:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"835151","meta_rev_id":"4ad90e8bfd37340972756fcbd8dbb10dc6eb95f0","_number":835151,"virtual_id_number":835151,"owner":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-30 17:29:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-03-30 16:06:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-03-30 16:06:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-24 22:54:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-29 20:15:57.000000000","updated_by":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"reviewer":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"state":"REVIEWER"}],"messages":[{"id":"9446e3e98935b7e9576446deb8f6f902d0511d3e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2022-03-24 21:51:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c831bf75e072756e5c5a470aad1fa515868a8332","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-24 22:54:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7a50711ec76e4cb2822066fa50109e99 : SUCCESS in 5m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/16faca02f3754780a63a9e1156b72f88 : SUCCESS in 5m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/745042f01cea40969233f09f9d0d5480 : SUCCESS in 4m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/25bcf12d026e4c2dae59d29c0ccdc597 : SUCCESS in 5m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/30e28bc7c95a422b9226e5916a31215a : SUCCESS in 6m 14s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/e74d4ffff0df43b68b17c589805dad0d : SUCCESS in 7m 48s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/22261f85e17543f48f0cc23a71e6d34e : SUCCESS in 1h 00m 59s","accounts_in_message":[],"_revision_number":1},{"id":"e141a7eb10792fd9bc727d2e3d25e996a44e6b3e","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2022-03-29 20:15:57.000000000","message":"Patch Set 1: Code-Review+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"d819cc55f0fe0ef3d660badb13342127919ede24","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2022-03-30 14:55:33.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"42ec59bcdffbf86547b9816245478d7725ceaa45","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-30 16:02:38.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/98dd19689a564b379bbc338ed120c818 : SUCCESS in 3m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/699fbcbe76884b2b8bd21dd314000d3c : SUCCESS in 4m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bc9a7c183fdc47abaee5ecec09b59ff0 : SUCCESS in 3m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/676ae0b3c7174f3784fdd5b3b5b96a8a : SUCCESS in 4m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8dfe598dbc140d0816694756ea2a624 : SUCCESS in 5m 31s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/5bacebc413004b34a3100768d6651e4f : SUCCESS in 5m 53s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/12799c931f0143b0bc9e58bb2c8eb436 : SUCCESS in 1h 01m 00s","accounts_in_message":[],"_revision_number":2},{"id":"1feea36e4cd21797856ebed64aaa8ad446b2c6b1","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2022-03-30 16:06:18.000000000","message":"Patch Set 2: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"125b8915e2898a39cde0a389dd8c24480d7a8473","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2022-03-30 16:06:30.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"25d31537716e6fba9eb5b438af810460cb3e3035","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-30 16:07:09.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"c762eba84a121038a021ac34ece23a77e9298f04","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-30 17:29:56.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/edcadd5a0c174b19bf491572a503a428 : SUCCESS in 4m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1f06e240d3b444b89f34e40ab9ccbe60 : SUCCESS in 3m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/da5df8e5c60242799bf5ba0be6f139bc : SUCCESS in 3m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c9d5766936ed45009d45d344a214f756 : SUCCESS in 3m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78969cdf164e4bdf9a9309c42f7cbe95 : SUCCESS in 5m 04s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/8d2bc7d335d845a1b390d4989606f1c1 : SUCCESS in 6m 30s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/f8cd3ff9e47a4487988e68bfc35f68af : SUCCESS in 1h 07m 56s","accounts_in_message":[],"_revision_number":2},{"id":"a0e1ddb396c567492fe9c20672fb4004af8a10d1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-30 17:29:56.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"4ad90e8bfd37340972756fcbd8dbb10dc6eb95f0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-30 17:31:32.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8ff29952e824a1fbc6ff18a75d3b0b6 : SUCCESS in 1m 11s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"be6622b552b4fe0c8e58a151e47f936587a9e89e","revisions":{"fef18708eaab2c4abffdd40e093dc8ad99fdbe55":{"kind":"REWORK","_number":1,"created":"2022-03-24 21:51:33.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/51/835151/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/51/835151/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/51/835151/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/51/835151/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/51/835151/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/51/835151/1"}}},"commit":{"parents":[{"commit":"06207b33e3317b984868aeb67aaba4bb8423e710","subject":"Fix broken behavior after recent OVS update","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/06207b33e3317b984868aeb67aaba4bb8423e710"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2022-03-24 21:06:57.000000000","tz":-300},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2022-03-24 21:51:33.000000000","tz":-300},"subject":"Revert \"Add an active wait in the \"Backend.lookup\"\"","message":"Revert \"Add an active wait in the \"Backend.lookup\"\"\n\nThis reverts commit bc004045cd30a708eadd64ded1af1b30ef1d1918.\n\nThere turned out to be some unresolvable issues with waiting for\nevents when lookup() is called from within a Command object in\nthe connection thread. Also the lookup of the notify_handler\nwas looking on the wrong object, so unless the notify_handler was\nspecifically passed (and we are not aware of any code that did),\nthis lookup timeout never actually was used.\n\nRemoving the feature, but leaving the arguments to avoid breaking\nany code and adding a warning to update code if they are passed.\n\nChange-Id: I42072ecb6a658814c7ff48cbc6f2a37a91d1eba8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/fef18708eaab2c4abffdd40e093dc8ad99fdbe55"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/fef18708eaab2c4abffdd40e093dc8ad99fdbe55"}]},"branch":"refs/heads/master"},"be6622b552b4fe0c8e58a151e47f936587a9e89e":{"kind":"REWORK","_number":2,"created":"2022-03-30 14:55:33.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/51/835151/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/51/835151/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/51/835151/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/51/835151/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/51/835151/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/51/835151/2"}}},"commit":{"parents":[{"commit":"06207b33e3317b984868aeb67aaba4bb8423e710","subject":"Fix broken behavior after recent OVS update","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/06207b33e3317b984868aeb67aaba4bb8423e710"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2022-03-24 21:06:57.000000000","tz":-300},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2022-03-30 14:55:21.000000000","tz":-300},"subject":"Revert \"Add an active wait in the \"Backend.lookup\"\"","message":"Revert \"Add an active wait in the \"Backend.lookup\"\"\n\nThis reverts commit bc004045cd30a708eadd64ded1af1b30ef1d1918.\n\nThere turned out to be some unresolvable issues with waiting for\nevents when lookup() is called from within a Command object in\nthe connection thread. Also the lookup of the notify_handler\nwas looking on the wrong object, so unless the notify_handler was\nspecifically passed (and we are not aware of any code that did),\nthis lookup timeout never actually was used.\n\nRemoving the feature, but leaving the arguments to avoid breaking\nany code and adding a warning to update code if they are passed.\n\nChange-Id: I42072ecb6a658814c7ff48cbc6f2a37a91d1eba8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/be6622b552b4fe0c8e58a151e47f936587a9e89e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/be6622b552b4fe0c8e58a151e47f936587a9e89e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
