)]}'
{"id":"openstack%2Fovsdbapp~855762","triplet_id":"openstack%2Fovsdbapp~master~Icb8604a4963b6c239425803560b4d326ab0349f7","project":"openstack/ovsdbapp","branch":"master","topic":"release-health-check","hashtags":[],"change_id":"Icb8604a4963b6c239425803560b4d326ab0349f7","subject":"DNM: dummy change to test gate health","status":"ABANDONED","created":"2022-09-03 20:39:44.000000000","updated":"2022-09-07 09:07:12.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"ed6066acd41fc8cf89e52479d1eb883ffd95f831","_number":855762,"virtual_id_number":855762,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2022-09-03 21:41:30.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-03 21:41:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2c5e11a0683370aed0183a682d7cdcdc7d91c1f0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-09-03 20:39:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"79953816c5eca4008182b052982a05828fb0a7c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-03 21:41:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c4cdcdde8a194c878f25fe8b32336570\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/408cec294d4d487c9261168b18a037ef : SUCCESS in 5m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/101ce1332b5b454181873ad4d0b85cfc : SUCCESS in 4m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/41026be473c84c79b263edf69ff26ad0 : SUCCESS in 4m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/82c32f5aa2a34763b25a5fb3911a6f7a : SUCCESS in 4m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ac3113f812d4b999a1dc889ec536e09 : SUCCESS in 7m 32s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/27e2ccb147514879895ff71b1475d3f8 : SUCCESS in 7m 04s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/7e5a357018664a59b0ec197bb94175e6 : SUCCESS in 50m 34s","accounts_in_message":[],"_revision_number":1},{"id":"ed6066acd41fc8cf89e52479d1eb883ffd95f831","tag":"autogenerated:gerrit:abandon","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-09-07 09:07:12.000000000","message":"Abandoned\n\nGate check passed, no further need for this patch.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"34d60881cab4c5a55615a49ff61a39f51214c3d8","revisions":{"34d60881cab4c5a55615a49ff61a39f51214c3d8":{"kind":"REWORK","_number":1,"created":"2022-09-03 20:39:44.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/62/855762/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/62/855762/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/62/855762/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/62/855762/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/62/855762/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/62/855762/1"}}},"commit":{"parents":[{"commit":"96cf8d6288587423e65d5149016e07fb51430724","subject":"Merge \"nb: add support for lb health checks API\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/96cf8d6288587423e65d5149016e07fb51430724"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2022-09-03 20:39:30.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2022-09-03 20:39:41.000000000","tz":120},"subject":"DNM: dummy change to test gate health","message":"DNM: dummy change to test gate health\n\nChange-Id: Icb8604a4963b6c239425803560b4d326ab0349f7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/34d60881cab4c5a55615a49ff61a39f51214c3d8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/34d60881cab4c5a55615a49ff61a39f51214c3d8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
