)]}'
{"id":"openstack%2Fovsdbapp~856748","triplet_id":"openstack%2Fovsdbapp~stable%2Fzed~Id0fbc09c2960f570d2eaad5c3163f355f9f9cff7","project":"openstack/ovsdbapp","branch":"stable/zed","topic":"create-zed","hashtags":[],"change_id":"Id0fbc09c2960f570d2eaad5c3163f355f9f9cff7","subject":"Update .gitreview for stable/zed","status":"MERGED","created":"2022-09-09 11:48:03.000000000","updated":"2022-09-14 13:35:39.000000000","submitted":"2022-09-14 13:32:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"856748-create-zed","meta_rev_id":"8f846ee23f306336a413542bb5d3e45d57f65da0","_number":856748,"virtual_id_number":856748,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-09-14 13:32:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"recommended":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2022-09-10 13:45:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":1,"date":"2022-09-13 17:40:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":2,"date":"2022-09-14 12:15:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":1,"date":"2022-09-14 12:15:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-09 12:02:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-09-10 13:45:12.000000000","updated_by":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"reviewer":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"state":"REVIEWER"},{"updated":"2022-09-13 17:40:30.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2022-09-14 12:15:45.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"}],"messages":[{"id":"2a37b39521e67233aefc786ba52cbdefc20cb45a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2022-09-09 11:48:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f86fcd3b303eb116abfc0f2262d5641e872bf110","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-09 12:02:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/847be57670de4819aebb531116036232\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/65e4b6a7311f449883e1d8cc4a094c72 : SUCCESS in 5m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e1aa1e807b294039a67397883523fbf0 : SUCCESS in 3m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3bc1471bb4574ccda13ac042e853a299 : SUCCESS in 7m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/461810f08bd54f1c8807edcee85a1fd3 : SUCCESS in 4m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f5134da9bf84ea389b7f60a8534a1ef : SUCCESS in 6m 08s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/c67e53444f1244b38eb41b5e8d48512e : SUCCESS in 8m 58s","accounts_in_message":[],"_revision_number":1},{"id":"37c9d0247d9e4fcb758c86ed1d1cdad93c04838c","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2022-09-10 13:45:12.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"d7096b1c90380a358d4a0821a6c96f982f20c8a4","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-09-13 17:40:30.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"610593879db3a5734061aa802facda9648fd2a5d","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-09-14 12:15:45.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5f3d85d94fcc45e0b1b200a83e5b563bdb1ecd4b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-14 12:16:16.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"218853b130a2d6e0a8519c4a6ec214299be42f45","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-14 13:32:21.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bc1a7f05087940ff8780dbb8a7beeffe\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb02e4f3b000468baf4dc3830577346f : SUCCESS in 5m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2fb721593677421aa19ba49700b0727c : SUCCESS in 4m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8b58e4496d7a41419ed77f48e40e7ec3 : SUCCESS in 3m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92f7ffee80424a9dbfe4dfe764901e61 : SUCCESS in 5m 54s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/0922b0fe47b44d27974b59756f16d4f8 : SUCCESS in 6m 59s","accounts_in_message":[],"_revision_number":1},{"id":"7f68b548757a773fc1a0e41d5cceec0718216d11","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-14 13:32:21.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"8f846ee23f306336a413542bb5d3e45d57f65da0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-14 13:35:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c638eeaa7a6441c4aa5c94a0a5364da6\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/127d29dbbaf842c0b39cc7f3eb54ccc7 : SUCCESS in 1m 07s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b58a340d97d362f0fa553a074357d50e6fc0ab28","revisions":{"b58a340d97d362f0fa553a074357d50e6fc0ab28":{"kind":"REWORK","_number":1,"created":"2022-09-09 11:48:03.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/48/856748/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/48/856748/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/48/856748/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/48/856748/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/48/856748/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/48/856748/1"}}},"commit":{"parents":[{"commit":"96cf8d6288587423e65d5149016e07fb51430724","subject":"Merge \"nb: add support for lb health checks API\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/96cf8d6288587423e65d5149016e07fb51430724"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2022-09-09 11:48:03.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2022-09-09 11:48:03.000000000","tz":0},"subject":"Update .gitreview for stable/zed","message":"Update .gitreview for stable/zed\n\nChange-Id: Id0fbc09c2960f570d2eaad5c3163f355f9f9cff7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/b58a340d97d362f0fa553a074357d50e6fc0ab28"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/b58a340d97d362f0fa553a074357d50e6fc0ab28"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}}]}],"submit_requirements":[]}
