)]}'
{"id":"openstack%2Fovsdbapp~862576","triplet_id":"openstack%2Fovsdbapp~stable%2Fyoga~Iab2177fb9fa653292a3805689895f98e0833dc4a","project":"openstack/ovsdbapp","branch":"stable/yoga","attention_set":{},"removed_from_attention_set":{"8655":{"account":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"last_update":"2022-10-25 20:18:55.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iab2177fb9fa653292a3805689895f98e0833dc4a","subject":"Don\u0027t force_reconnect() on unhandled Idl exception","status":"MERGED","created":"2022-10-25 13:09:58.000000000","updated":"2022-10-25 20:19:54.000000000","submitted":"2022-10-25 20:18:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"862576","cherry_pick_of_change":862575,"cherry_pick_of_patch_set":1,"meta_rev_id":"76490d717126b731eda4521b36bb6727cb29e3e2","_number":862576,"virtual_id_number":862576,"owner":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-10-25 20:18:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"all":[{"value":2,"date":"2022-10-25 19:11:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"all":[{"value":1,"date":"2022-10-25 19:11:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-25 13:09:58.000000000","updated_by":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"reviewer":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"state":"CC"},{"updated":"2022-10-25 14:26:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-25 19:11:24.000000000","updated_by":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"reviewer":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"state":"REVIEWER"}],"messages":[{"id":"aee381f316504df439047100fde9cba60778bf65","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2022-10-25 13:09:58.000000000","message":"Patch Set 1: Cherry Picked from branch stable/zed.","accounts_in_message":[],"_revision_number":1},{"id":"720fd5670176025bf5e0ce78d59438f167bf6cdb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-25 14:26:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1f6c493bd9824af0a1b28f9cf25bed86\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/64d1bc19aaf44411a2952df18dbd7be4 : SUCCESS in 2m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f9c5c811048b4fcfbefb8840a67a64c5 : SUCCESS in 2m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2f7c416766ed48a2869419e414abc3ad : SUCCESS in 2m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a0d0234394c640218ef836f8a3ffd816 : SUCCESS in 3m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6fbc1874f518449aa0add4bac4980b12 : SUCCESS in 4m 13s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/1513cbfd2fbb4e29aa6717e33a584dd6 : SUCCESS in 5m 58s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/1a4c61887b064c74ab1b011123727063 : SUCCESS in 1h 09m 22s","accounts_in_message":[],"_revision_number":1},{"id":"e91a38a4a8b6c0e58f6de066bae4bf4a0e59da7d","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2022-10-25 19:11:24.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"da18ec6e0003b5748861ff8e5e28aafdd5edd190","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-25 19:11:59.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b5ce4897f9be7c920ff15aecb9918d1fe306195d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-25 20:18:55.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b62469d885864d02b3ee162da59fa742\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/09e46e8a4300447988b8cbe2c4d5de8f : SUCCESS in 3m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/be825f31c35a490ab58e70ceb81830a6 : SUCCESS in 3m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e9b80dc8772a42979cf64afad9f6099b : SUCCESS in 3m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/577ce305df864ae898cb2d3cfb2ca7d9 : SUCCESS in 3m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/caf4047376ea4ff482bc63de99778f28 : SUCCESS in 4m 54s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/69655594aeb04f73861acc99053bd8d9 : SUCCESS in 5m 17s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/e612b4d6b6f4415695c8a3387e5bff5b : SUCCESS in 51m 33s","accounts_in_message":[],"_revision_number":1},{"id":"9436f58f213cc9859cb1ba6fd25b387fe9cec461","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-25 20:18:55.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"76490d717126b731eda4521b36bb6727cb29e3e2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-25 20:19:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/22460b4e0cb24e7db1070458b10dffe0\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ec52bef63b647539426772f7c820230 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ab3e0cb0d0865417efbf103f44954573a5ba92ac","revisions":{"ab3e0cb0d0865417efbf103f44954573a5ba92ac":{"kind":"REWORK","_number":1,"created":"2022-10-25 13:09:58.000000000","uploader":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"ref":"refs/changes/76/862576/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/76/862576/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/76/862576/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/76/862576/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/76/862576/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/76/862576/1"}}},"commit":{"parents":[{"commit":"92cbba4481a7c7acedf24d8250fc7cf4be6b46d7","subject":"Fix logging exception while handling transaction exception","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/92cbba4481a7c7acedf24d8250fc7cf4be6b46d7"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2022-10-24 19:58:46.000000000","tz":-300},"committer":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2022-10-25 13:09:58.000000000","tz":0},"subject":"Don\u0027t force_reconnect() on unhandled Idl exception","message":"Don\u0027t force_reconnect() on unhandled Idl exception\n\nThere\u0027s no reason to believe that reconnecting to ovsdb-server will\nresolve an unhandled exception in python-ovs. In addition, since users\noften subclass Idl and add their own notify() methods, there could be\nexceptions thrown from that code.\n\nThe best we can do is log what is going on and rely on users to fix\nthe issue. Delaying with sleep() is usually a bad idea since if there\nwas some kind of ovsdb reconnection, it will delay calls to Idl.run()\nwhich will handle that reconnection over several calls.\n\nChange-Id: Iab2177fb9fa653292a3805689895f98e0833dc4a\n(cherry picked from commit cd70d1e29033f50e23ccd97aacde6cc476baf52e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/ab3e0cb0d0865417efbf103f44954573a5ba92ac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/ab3e0cb0d0865417efbf103f44954573a5ba92ac"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}}]}],"submit_requirements":[]}
