)]}'
{"id":"openstack%2Fovsdbapp~873853","triplet_id":"openstack%2Fovsdbapp~master~Iabfc1bd0a1b2c6998a013d19c47cda10a3879c82","project":"openstack/ovsdbapp","branch":"master","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2023-02-23 11:03:09.000000000","reason":"\u003cGERRIT_ACCOUNT_16688\u003e replied on the change","reason_account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},"5756":{"account":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"last_update":"2023-03-22 15:21:10.000000000","reason":"\u003cGERRIT_ACCOUNT_5756\u003e replied on the change","reason_account":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}},"23567":{"account":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"last_update":"2023-04-12 09:13:46.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iabfc1bd0a1b2c6998a013d19c47cda10a3879c82","subject":"Add new function ls_get_localnet_ports","status":"MERGED","created":"2023-02-15 10:25:55.000000000","updated":"2023-04-12 09:14:44.000000000","submitted":"2023-04-12 09:13:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":31,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"873853","meta_rev_id":"f45a66f70a73bb58685e581e07011ed613b3c59b","_number":873853,"virtual_id_number":873853,"owner":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-04-12 09:13:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2023-04-04 13:36:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},{"value":2,"date":"2023-04-04 13:53:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},{"value":1,"date":"2023-04-04 13:53:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-15 10:52:37.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2023-02-15 11:57:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-15 14:28:22.000000000","updated_by":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"reviewer":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"state":"REVIEWER"},{"updated":"2023-02-16 11:41:53.000000000","updated_by":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},"reviewer":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},"state":"REVIEWER"}],"messages":[{"id":"2fac2e9385114b49a7f1d3839b11a098597228a4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2023-02-15 10:25:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e747e5c6750e706e0eaf58bdc86b78e1c867d67d","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-02-15 10:52:37.000000000","message":"Patch Set 1: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"8c5db583da4104b1c058ced58417310041d44768","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-15 11:57:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/18c744dad8254d37a3a7ee0ba235637d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba17eb207d184a84b7ef24943d05e0b9 : SUCCESS in 3m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e79c3dd8a7004d5fb3dd4b9c1336ecd3 : SUCCESS in 2m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/aacb55cbfd8943f9abafd65e2577681b : SUCCESS in 2m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25f9c444c9294e3fa46329245238187c : SUCCESS in 4m 36s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/40844667d038475688ae8b6ab0337c0d : SUCCESS in 6m 59s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/ca5e0a6b48a844f3a7ba022a51a4f21d : SUCCESS in 1h 23m 33s","accounts_in_message":[],"_revision_number":1},{"id":"cb27c8a5aed1fbb70f5af112d677a316b07928a2","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2023-02-15 14:28:22.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"6d57fa0104358657c579aca226be989ad5cbc8de","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2023-02-16 08:45:48.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"2300ebe33d61cc5b497f167b29d99ef5ea320af9","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2023-02-16 09:36:41.000000000","message":"Patch Set 2:\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"5866c84d71aa7341f157c0f615be965556b4c377","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 10:04:16.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b6769f15f7d3456198e67cda465df2c8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8be96f6ae33647979f9b5e6449f10990 : SUCCESS in 3m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0624afc570c0410f8fc642be167e0b2d : SUCCESS in 4m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7804591463704b9a8c4ebe82586b3d3e : SUCCESS in 3m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d52a12660f5c49e6b657fca2db1b37bf : SUCCESS in 4m 21s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/c9aad570a9a64594aa7b08ad1ee7d899 : SUCCESS in 6m 40s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/a3af67d02257425fbe0aad934983eb85 : SUCCESS in 1h 12m 48s","accounts_in_message":[],"_revision_number":2},{"id":"d8abc900edecbc15c6f88751b2c89d054d69828f","author":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},"date":"2023-02-16 11:41:53.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"32138e1ddf7def4f8ad3da075a420a582a7aecd9","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-02-23 11:03:09.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"e8bb1c76136d1401d2a786053db8c092cc80a020","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2023-03-09 11:20:29.000000000","message":"Patch Set 2: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a43e0b03db0f6a500abe08969fec8919658fb96c","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2023-03-09 11:23:09.000000000","message":"Patch Set 2: -Workflow\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"636d7af7c0868529af947d06f3c74f954bd50e65","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2023-03-16 14:09:36.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d09bdd7755eea627088af767c34b4b206e7d7e67","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-03-16 15:20:59.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"472bdb7258aa6844d89932d11518eb40a53e85ce","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-03-16 15:21:13.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"e08d5d3f09e3a5112d519c050307924d79fe5735","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2023-03-16 15:48:30.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"4501dccd6a766a23fb9a195e535c0be2d2557259","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2023-03-16 15:49:03.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"12745bbbed873a20cb21d1398f5afd56807de47b","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2023-03-16 16:01:35.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"9f355f0ddbaa77eea91ad8851bb4558fa4efc476","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2023-03-16 16:04:02.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"60616e68e745cb632f9b6bf1080843f6efe74d73","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-16 17:22:59.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/385d47a8aaf24474aeea4f8eb6ef91f1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/743e431d052f446b882db45c41863112 : SUCCESS in 3m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/42f68ab6333148cd9ac181ee52cbf52b : SUCCESS in 2m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/09a4586025d34382bbd0df6324f05dd8 : SUCCESS in 2m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/04bf12a859cd48fa934785d53a5c2214 : SUCCESS in 4m 39s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/bc4013e63cfe4fa2acb536a50722d486 : FAILURE in 5m 02s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/a3277d2381624b5981043fbdd73763ce : SUCCESS in 1h 28m 43s","accounts_in_message":[],"_revision_number":4},{"id":"10b8994c5a2150dbe1a3299da2558539aa78ba87","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2023-03-16 18:14:04.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"8b4f21bec701a196ea7b7aeb5e028ae6ac4f2723","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2023-03-16 18:14:19.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"8f0d60515d0b83969b397e0212944e22fab07849","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-16 19:04:20.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/584f56c0ebf543528a7ef6c78d1bf5e1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aa6949d3fe614180a1e6a99fc4be024c : SUCCESS in 3m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e67d781e030a4ddfb12f517017a36a41 : SUCCESS in 3m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8e39da8dc2704aa7b30f2aab1b499140 : SUCCESS in 3m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d6b8849d26440c3a0f866c51dbf1688 : SUCCESS in 4m 41s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/87c5037f1eb5450db0c65e842dd5922c : FAILURE in 5m 31s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/7e53ac4a2a4849f5a6e8663fda3ebb93 : SUCCESS in 49m 23s","accounts_in_message":[],"_revision_number":5},{"id":"9b3777c60f5bc35b2555343e857aab04dab9b416","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2023-03-16 23:17:38.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"3f94f85d46159e10b3db281929bcd64cea11414d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2023-03-17 06:39:33.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"d76ae90c31a07ad556068ae5ed6888065b138498","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2023-03-17 06:39:45.000000000","message":"Patch Set 6:\n\n(3 comments)","accounts_in_message":[],"_revision_number":6},{"id":"c182c6197d4cc752ae5e33a1fbdc74b7ef6e08cd","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2023-03-17 06:39:56.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"58ecbb5d9dc2211abfa86d29b54cb58d498d7724","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-17 08:06:39.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/137fed245e8b4df7a4bc6b97d48543ac\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cfb27dc18d044c9fa32272397e76334c : SUCCESS in 2m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e311f14ce1cf4c1e9dd3aa766fbc1a31 : SUCCESS in 2m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fbdf536295be493381eaa537b7da9cd7 : SUCCESS in 3m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9caeb49a5f2f46c290df226dd07fec36 : SUCCESS in 5m 44s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/3a92d62616de4e159e9a301e18b4ea2b : FAILURE in 5m 13s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/cc9923b01b5c4e2cbde7f7514bed7d8a : SUCCESS in 1h 21m 26s","accounts_in_message":[],"_revision_number":6},{"id":"751d904459f13c8dc0756782c01c33d6edc38626","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2023-03-17 13:45:33.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"3b6fac434ca00db0b075fe7c251c1fd6e4fd888f","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2023-03-17 14:23:55.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"3df4d31d7627004f5d52076cd3fbc2096605f1d7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2023-03-17 14:24:34.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"41f03d389efadf79af790a29a4d517260acc4478","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-17 15:27:07.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e8f697a37e474beab3ac8b3b8214e34b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ca951b667624156a21c7c73f66d13cb : SUCCESS in 4m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/19c3407ba06f468f8d402ad49f685cd6 : SUCCESS in 3m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5c519fd436424155aa5fa667f5cf9a17 : SUCCESS in 3m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb79de6db81e4ed1978f58dd49b4541b : SUCCESS in 6m 00s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/031e6721d02e45919286c643878014f0 : SUCCESS in 5m 07s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/89c325777a734f96997078493d6e472f : SUCCESS in 54m 11s","accounts_in_message":[],"_revision_number":7},{"id":"4c34a95b5219c80b8d3547d9bb6b310df2c5f862","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2023-03-22 15:21:10.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"43ab4d7e42ab4a1e3101df82b9401d8fd8df3e19","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-03-28 09:08:05.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"88421b2e68c881728292c3fc31299411843675c9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-28 09:08:34.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"5e59adb04eebc42f251f5d9ac4e059acc3b7622c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-28 10:23:20.000000000","message":"Patch Set 7: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/74242f45a4a04e86ab20fc3a9ade6b9a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/22fe8c37920a433bb19e184e80945739 : SUCCESS in 4m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/408bb5b9eca647a5b4a7daecad4871ef : SUCCESS in 2m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a1e50526fb3746b9b948f42f7cadceaa : SUCCESS in 3m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/efd81db21db84562a37d206dea3b25aa : SUCCESS in 4m 51s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/c1125c6f75ac4364ad120e4012b778ee : FAILURE in 5m 20s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/2517821847dc4eb68890b7d672b68b84 : SUCCESS in 1h 09m 48s","accounts_in_message":[],"_revision_number":7},{"id":"f6251ffc92988b51b72e9db95315444b468a8eec","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2023-03-28 12:19:43.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"afcd2ba6c48a2df630cf25e5d03d7c4d90a05be0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-28 13:47:07.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/76f6bd96c0f740f5b663b87caf782905\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae397555ed224822b599800e689b3c29 : SUCCESS in 3m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0ee453325dca4f7688e38460b99643f3 : SUCCESS in 2m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6566a3b1326741b797d2ac16c2abd00a : SUCCESS in 2m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de31da0e70b741e985729c6f9d48961b : SUCCESS in 4m 32s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/14438b8ba0e04e4c8092f07ad777d0c8 : FAILURE in 4m 13s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/118863273ac24112bf70a64d53ab1a36 : SUCCESS in 1h 11m 49s","accounts_in_message":[],"_revision_number":7},{"id":"3f7ebbba387f9ae2cfa00dff5ecce2a11f9f80d4","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2023-03-28 17:23:48.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"33cd706f16f9a0479313fd41e5a5d7026a533f25","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-28 18:24:30.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/799675ebe0e84401861fb6572dfe845e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b0566d9003a042699f563227c97b56de : SUCCESS in 3m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e652c09d107f458ab945dbb2c85a453d : SUCCESS in 4m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e8674d68206a48fa9b177cc74dab2915 : SUCCESS in 2m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aebcd66e30914759bf7e431f393077bb : SUCCESS in 5m 56s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/ac2dec5899444e88b8469571ce151a07 : FAILURE in 5m 08s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/88dae08490e042a8bf16f9f579bc7805 : SUCCESS in 58m 56s","accounts_in_message":[],"_revision_number":7},{"id":"8b34c38adac76796ad042ce24127163e8f72fe49","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2023-03-30 06:24:47.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"88ca594f25a9c602f5e4d90c745bdec18ced23ee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-30 08:01:22.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/35a5f670be84447e8a40d18372acf616\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/08ca83e71e2646d29b18de1eb1afcf40 : SUCCESS in 2m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3c94458918fb47be92c5d0e58cb4f2f0 : SUCCESS in 2m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a2c850a801994a4c83eaf804ad16dd93 : SUCCESS in 2m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0416a74dd76c473a9ff58a1828d8a0b8 : SUCCESS in 4m 24s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/a2ce0763d3a94d7897d215278ae8dd2e : FAILURE in 4m 10s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/67d4ce1685a64e198dfc071213ad5440 : SUCCESS in 1h 30m 42s","accounts_in_message":[],"_revision_number":7},{"id":"92453833b0456ccb40a51965c90e30ad71764209","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2023-03-31 06:14:46.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"9152fbf8c39a368c4136e9df3dde1f8304ecdb7c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-31 07:42:35.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a8db5bcdb591462e83fe19e26d431ace\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/26085458b5ff4b899ff572f388e530c2 : SUCCESS in 3m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2bdcafe797944ccb8fc79d40a59c4333 : SUCCESS in 2m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5b79b539b6c84252a251b902ffd8678e : SUCCESS in 3m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b72cfd22d2de41e49c397035e1174832 : SUCCESS in 5m 37s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/bc5acce9f3c14830baab8626f49bcccb : FAILURE in 4m 32s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/1f6fc123eaf642f1a64607c902fa8e87 : SUCCESS in 1h 21m 55s","accounts_in_message":[],"_revision_number":7},{"id":"5abf9a0b6d07cdc83f0ef42f451f3f977c9317de","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2023-04-03 05:30:48.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"54b2e5d7746be4de00831d5c03b33a08a54c7a96","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-03 06:25:35.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c142015487384cf5864f131619a5ea97\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e1676092196d4949b092e8fb86d3b05c : SUCCESS in 3m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/21cd06568a394aeaa571cc77c9a8f506 : SUCCESS in 2m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d353a14d24604b528f14076a9b6bfdbb : SUCCESS in 2m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a8daa88aef2420a95377f569772cd43 : SUCCESS in 4m 57s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/ba2a388f081f43caa7509b640bf49bf9 : FAILURE in 5m 25s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/b6c27d64774d4393912030fc8734d000 : SUCCESS in 53m 20s","accounts_in_message":[],"_revision_number":7},{"id":"d66a74efc2f123fa7d7285191c71cf5e219e170d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2023-04-04 13:36:11.000000000","message":"Patch Set 8: Patch Set 7 was rebased","accounts_in_message":[],"_revision_number":8},{"id":"0d78a559c047063cadd930a5dcd356e0ae4d1192","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2023-04-04 13:53:23.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"32a3ec2cec51318ae8c84e0df51d8df8cdbf38fd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-04 14:45:46.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/da9c2e4c94934229aac7af268521ceaa\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ae3478d27694414a5d2d129b1545658 : SUCCESS in 3m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1ddeee22d26e4d23bdef5c82120d5d6b : SUCCESS in 3m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7fc7ad6c6fef444fa39e226d5b70d460 : SUCCESS in 3m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1edbc1f7a9234a94901470c6a4803ef1 : SUCCESS in 4m 37s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/42e09cb094e4453396f032af636ac1f1 : SUCCESS in 6m 09s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/1ccf914716914d8a850e097a01e69385 : SUCCESS in 1h 02m 55s","accounts_in_message":[],"_revision_number":8},{"id":"6a9fbf8ccb3828426103bb265d878260667955f2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-12 07:56:47.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"4ff3e191ca971dcc07fa54d86838823f14f04dd9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-12 09:13:46.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/650f0566a2974b358f9acb0ba0f6e1da\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/05f48266295f4a52be4970bdd8456695 : SUCCESS in 3m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/291dfa5613ed4e3d94e78532cc9adb0a : SUCCESS in 3m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/acce074ebea24dc2a230492670dada81 : SUCCESS in 2m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d0e47c61b7c46e2b7155d177fdbf063 : SUCCESS in 5m 33s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/80ceab22b2174e98a71e4f5f452caceb : SUCCESS in 5m 20s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/4c0187af27284fffa5d97fae85bead47 : SUCCESS in 1h 14m 30s","accounts_in_message":[],"_revision_number":8},{"id":"d37ad60448f40735915a159e1907edbc3d7a2171","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-12 09:13:46.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":8},{"id":"f45a66f70a73bb58685e581e07011ed613b3c59b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-12 09:14:44.000000000","message":"Patch Set 8:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b63c571f5cc844319724a2308c8bedea\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf70ab448e624546982eb860b903117c : SUCCESS in 38s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"f2c5a39edd0e89b21455b49e9720be95fc0bd2dd","revisions":{"a27d294e2d9f9a30799d9bbdb2cd89f39ecaf1e3":{"kind":"REWORK","_number":1,"created":"2023-02-15 10:25:55.000000000","uploader":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"ref":"refs/changes/53/873853/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/53/873853/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/53/873853/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/53/873853/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/53/873853/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/53/873853/1"}}},"commit":{"parents":[{"commit":"7733f0c5364de3805d7f8d5be429e0a464a7c6b1","subject":"Define \"LbAddIpPortMappingCommand\" with latin chars","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/7733f0c5364de3805d7f8d5be429e0a464a7c6b1"}]}],"author":{"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","date":"2023-02-15 10:25:01.000000000","tz":60},"committer":{"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","date":"2023-02-15 10:25:01.000000000","tz":60},"subject":"Add new function ls_get_localnet_port","message":"Add new function ls_get_localnet_port\n\nThis is used to detect if a given logical_switch has any localnet\nport associated to it\n\nChange-Id: Iabfc1bd0a1b2c6998a013d19c47cda10a3879c82\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/a27d294e2d9f9a30799d9bbdb2cd89f39ecaf1e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/a27d294e2d9f9a30799d9bbdb2cd89f39ecaf1e3"}]},"branch":"refs/heads/master"},"1b3b7ba32427be099858ad0cb66de052308649a4":{"kind":"REWORK","_number":2,"created":"2023-02-16 08:45:48.000000000","uploader":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"ref":"refs/changes/53/873853/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/53/873853/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/53/873853/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/53/873853/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/53/873853/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/53/873853/2"}}},"commit":{"parents":[{"commit":"7733f0c5364de3805d7f8d5be429e0a464a7c6b1","subject":"Define \"LbAddIpPortMappingCommand\" with latin chars","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/7733f0c5364de3805d7f8d5be429e0a464a7c6b1"}]}],"author":{"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","date":"2023-02-15 10:25:01.000000000","tz":60},"committer":{"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","date":"2023-02-16 08:45:21.000000000","tz":60},"subject":"Add new function ls_get_localnet_ports","message":"Add new function ls_get_localnet_ports\n\nThis is used to detect if a given logical_switch has localnet\nports associated to it, and return them\n\nChange-Id: Iabfc1bd0a1b2c6998a013d19c47cda10a3879c82\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/1b3b7ba32427be099858ad0cb66de052308649a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/1b3b7ba32427be099858ad0cb66de052308649a4"}]},"branch":"refs/heads/master"},"d8d6df556b7120985189dc49d2cd867a310a303f":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2023-03-16 15:20:59.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/53/873853/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/53/873853/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/53/873853/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/53/873853/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/53/873853/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/53/873853/3"}}},"commit":{"parents":[{"commit":"2ed128014898c357063427e9347124835ff5384c","subject":"[CI] Set the default OS_TEST_PATH in the stestr conf file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/2ed128014898c357063427e9347124835ff5384c"}]}],"author":{"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","date":"2023-02-15 10:25:01.000000000","tz":60},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2023-03-16 15:20:59.000000000","tz":0},"subject":"Add new function ls_get_localnet_ports","message":"Add new function ls_get_localnet_ports\n\nThis is used to detect if a given logical_switch has localnet\nports associated to it, and return them\n\nChange-Id: Iabfc1bd0a1b2c6998a013d19c47cda10a3879c82\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/d8d6df556b7120985189dc49d2cd867a310a303f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/d8d6df556b7120985189dc49d2cd867a310a303f"}]},"branch":"refs/heads/master","description":"Rebase"},"cde9057e597dcd779326ac9882182f0565466995":{"kind":"REWORK","_number":4,"created":"2023-03-16 15:48:30.000000000","uploader":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"ref":"refs/changes/53/873853/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/53/873853/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/53/873853/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/53/873853/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/53/873853/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/53/873853/4"}}},"commit":{"parents":[{"commit":"2ed128014898c357063427e9347124835ff5384c","subject":"[CI] Set the default OS_TEST_PATH in the stestr conf file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/2ed128014898c357063427e9347124835ff5384c"}]}],"author":{"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","date":"2023-02-15 10:25:01.000000000","tz":60},"committer":{"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","date":"2023-03-16 15:47:55.000000000","tz":60},"subject":"Add new function ls_get_localnet_ports","message":"Add new function ls_get_localnet_ports\n\nThis is used to detect if a given logical_switch has localnet\nports associated to it, and return them\n\nChange-Id: Iabfc1bd0a1b2c6998a013d19c47cda10a3879c82\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/cde9057e597dcd779326ac9882182f0565466995"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/cde9057e597dcd779326ac9882182f0565466995"}]},"branch":"refs/heads/master"},"08d2425ddb074c0955ce468e3aa8b7028e211bf7":{"kind":"REWORK","_number":5,"created":"2023-03-16 18:14:19.000000000","uploader":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"ref":"refs/changes/53/873853/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/53/873853/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/53/873853/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/53/873853/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/53/873853/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/53/873853/5"}}},"commit":{"parents":[{"commit":"2ed128014898c357063427e9347124835ff5384c","subject":"[CI] Set the default OS_TEST_PATH in the stestr conf file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/2ed128014898c357063427e9347124835ff5384c"}]}],"author":{"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","date":"2023-02-15 10:25:01.000000000","tz":60},"committer":{"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","date":"2023-03-16 18:14:07.000000000","tz":60},"subject":"Add new function ls_get_localnet_ports","message":"Add new function ls_get_localnet_ports\n\nThis is used to detect if a given logical_switch has localnet\nports associated to it, and return them\n\nChange-Id: Iabfc1bd0a1b2c6998a013d19c47cda10a3879c82\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/08d2425ddb074c0955ce468e3aa8b7028e211bf7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/08d2425ddb074c0955ce468e3aa8b7028e211bf7"}]},"branch":"refs/heads/master"},"f92b164f6aa32c615849a6279c591b3e6a908164":{"kind":"REWORK","_number":6,"created":"2023-03-17 06:39:33.000000000","uploader":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"ref":"refs/changes/53/873853/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/53/873853/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/53/873853/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/53/873853/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/53/873853/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/53/873853/6"}}},"commit":{"parents":[{"commit":"2ed128014898c357063427e9347124835ff5384c","subject":"[CI] Set the default OS_TEST_PATH in the stestr conf file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/2ed128014898c357063427e9347124835ff5384c"}]}],"author":{"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","date":"2023-02-15 10:25:01.000000000","tz":60},"committer":{"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","date":"2023-03-17 06:39:23.000000000","tz":60},"subject":"Add new function ls_get_localnet_ports","message":"Add new function ls_get_localnet_ports\n\nThis is used to detect if a given logical_switch has localnet\nports associated to it, and return them\n\nChange-Id: Iabfc1bd0a1b2c6998a013d19c47cda10a3879c82\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/f92b164f6aa32c615849a6279c591b3e6a908164"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/f92b164f6aa32c615849a6279c591b3e6a908164"}]},"branch":"refs/heads/master"},"a6cdcabee8e2696524a9bdc38b7ad87d9566a9aa":{"kind":"REWORK","_number":7,"created":"2023-03-17 14:24:34.000000000","uploader":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"ref":"refs/changes/53/873853/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/53/873853/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/53/873853/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/53/873853/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/53/873853/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/53/873853/7"}}},"commit":{"parents":[{"commit":"2ed128014898c357063427e9347124835ff5384c","subject":"[CI] Set the default OS_TEST_PATH in the stestr conf file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/2ed128014898c357063427e9347124835ff5384c"}]}],"author":{"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","date":"2023-02-15 10:25:01.000000000","tz":60},"committer":{"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","date":"2023-03-17 14:24:23.000000000","tz":60},"subject":"Add new function ls_get_localnet_ports","message":"Add new function ls_get_localnet_ports\n\nThis is used to detect if a given logical_switch has localnet\nports associated to it, and return them\n\nChange-Id: Iabfc1bd0a1b2c6998a013d19c47cda10a3879c82\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/a6cdcabee8e2696524a9bdc38b7ad87d9566a9aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/a6cdcabee8e2696524a9bdc38b7ad87d9566a9aa"}]},"branch":"refs/heads/master"},"f2c5a39edd0e89b21455b49e9720be95fc0bd2dd":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2023-04-04 13:36:11.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/53/873853/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/53/873853/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/53/873853/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/53/873853/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/53/873853/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/53/873853/8"}}},"commit":{"parents":[{"commit":"ee28e381fb46a04be3451de65a5166dd457eede8","subject":"Use OVN\u0027s OVS submodule for functional tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/ee28e381fb46a04be3451de65a5166dd457eede8"}]}],"author":{"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","date":"2023-02-15 10:25:01.000000000","tz":60},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2023-04-04 13:36:11.000000000","tz":0},"subject":"Add new function ls_get_localnet_ports","message":"Add new function ls_get_localnet_ports\n\nThis is used to detect if a given logical_switch has localnet\nports associated to it, and return them\n\nChange-Id: Iabfc1bd0a1b2c6998a013d19c47cda10a3879c82\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/f2c5a39edd0e89b21455b49e9720be95fc0bd2dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/f2c5a39edd0e89b21455b49e9720be95fc0bd2dd"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
