)]}'
{"id":"openstack%2Fovsdbapp~877632","triplet_id":"openstack%2Fovsdbapp~stable%2F2023.1~I51564920cec916031aaebf8151b015a3d87e9471","project":"openstack/ovsdbapp","branch":"stable/2023.1","topic":"fix_functional_ci-stable/2023.1","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2023-03-17 09:57:47.000000000","reason":"Change was submitted"},"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2023-03-17 09:57:47.000000000","reason":"Change was submitted"},"5756":{"account":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"last_update":"2023-03-17 09:57:47.000000000","reason":"Change was submitted"},"8655":{"account":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"last_update":"2023-03-17 09:57:47.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I51564920cec916031aaebf8151b015a3d87e9471","subject":"[CI] Set the default OS_TEST_PATH in the stestr conf file","status":"MERGED","created":"2023-03-16 14:54:45.000000000","updated":"2023-03-17 09:58:55.000000000","submitted":"2023-03-17 09:57:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"877632-fix_functional_ci-stable/2023.1","cherry_pick_of_change":877661,"cherry_pick_of_patch_set":3,"meta_rev_id":"ad2cd46335e2a087c6405dff9a23ba328a22e9cc","_number":877632,"virtual_id_number":877632,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-17 09:57:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":2,"date":"2023-03-17 06:25:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":2,"date":"2023-03-17 08:28:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":1,"date":"2023-03-17 08:28:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"reviewers":{"REVIEWER":[{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-16 15:59:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-17 06:25:19.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2023-03-17 06:25:19.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"state":"REVIEWER"},{"updated":"2023-03-17 06:25:19.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"state":"REVIEWER"},{"updated":"2023-03-17 08:28:40.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"}],"messages":[{"id":"29e840be333ad0cee4144ba8b9d9ed9d5cc504ed","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-03-16 14:54:45.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"9d12b4701d0045bb480edc065be53e369a1d9359","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-16 15:59:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3e48cc67cd0f47c6931b3652c844e6a2\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/708c73f415074e989b30262a6d117024 : SUCCESS in 9m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2a406d2c576744c5961f6c6283fb46e4 : SUCCESS in 4m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e8433a3bb7044004989530dcc2fd02d5 : SUCCESS in 9m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e972b614b9cf457fac25a817bc1adac1 : SUCCESS in 8m 33s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/c6cb5932c9294d0fa6c92a16b2689f6b : SUCCESS in 6m 56s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/bc96c046f88646a8b1c9cf596ff94a02 : SUCCESS in 1h 02m 08s","accounts_in_message":[],"_revision_number":1},{"id":"b96cbb1eb9c5d622fa6b5f6a410c547a4b8becfe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-03-17 06:24:21.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"9089fa5481840d926ec877b36c158d995fef832d","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-03-17 06:25:19.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"14b7f9b7d19cfa5e98f41de13bbba76dc6921434","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-17 07:17:20.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/95f78a9335f4408082098f593a242a32\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7bb3510f7af3494fa4756be53a7205bb : SUCCESS in 3m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c03f9d2e95144fb0a7485f6350d58698 : SUCCESS in 2m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f865016f9f124dbe8539412dd4737bb7 : SUCCESS in 2m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c719b8b2b83f4996b0ef3de85fcd70ba : SUCCESS in 5m 13s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/e820f3b5ce6d4aed91b141e23c7e6ed8 : SUCCESS in 5m 19s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/f947e318377f444ba05f3e81981e405e : SUCCESS in 45m 43s","accounts_in_message":[],"_revision_number":2},{"id":"b2e57a1a06e849cda32d42ec5e1197b3c52f28f6","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-03-17 08:28:40.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e233d35825af3ce6f2307348620a31507aa55b46","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-17 08:29:06.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"fc51807b693bbf5f8a19ccfe806d694846642695","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-17 09:57:46.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/35cbd40071c249ae89d30d3b0076bdc9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2e4601fa2d144cbf961f909a9cea966d : SUCCESS in 2m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9c4b397c3d824263aa61f16ef34559e2 : SUCCESS in 2m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/81aca50d6e1e455c8b939d7a5a0ae070 : SUCCESS in 2m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/facc519f5724466a80ae8ade8d6f34e6 : SUCCESS in 4m 44s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/8b8f30133127426fa67b70525780c745 : SUCCESS in 5m 56s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/74894f656a1e49db8308c43452a0406b : SUCCESS in 1h 23m 22s","accounts_in_message":[],"_revision_number":2},{"id":"5d662ff6d65e80d883a426becb62f8e96565c7dc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-17 09:57:47.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"ad2cd46335e2a087c6405dff9a23ba328a22e9cc","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-17 09:58:55.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/15e2025749454ca89a1dbe86fe00f7ad\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a37e9d99a664ed49abb920142d6dbbf : SUCCESS in 45s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"cc45eb8268f570d3df744663ee889ea6f6a7288b","revisions":{"df5307685f87cb47c589547eca75ed4701233cc0":{"kind":"REWORK","_number":1,"created":"2023-03-16 14:54:45.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/32/877632/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/32/877632/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/32/877632/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/32/877632/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/32/877632/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/32/877632/1"}}},"commit":{"parents":[{"commit":"7733f0c5364de3805d7f8d5be429e0a464a7c6b1","subject":"Define \"LbAddIpPortMappingCommand\" with latin chars","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/7733f0c5364de3805d7f8d5be429e0a464a7c6b1"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-03-16 14:29:18.000000000","tz":60},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2023-03-16 14:54:45.000000000","tz":0},"subject":"[CI] Set the default OS_TEST_PATH in the stestr conf file","message":"[CI] Set the default OS_TEST_PATH in the stestr conf file\n\nSet the default OS_TEST_PATH in the stestr configuration file. Any\nother test environment can overwrite this variable in the \"setenv\"\nsection.\n\nChange-Id: I51564920cec916031aaebf8151b015a3d87e9471\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/df5307685f87cb47c589547eca75ed4701233cc0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/df5307685f87cb47c589547eca75ed4701233cc0"}]},"branch":"refs/heads/stable/2023.1"},"cc45eb8268f570d3df744663ee889ea6f6a7288b":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2023-03-17 06:24:21.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/32/877632/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/32/877632/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/32/877632/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/32/877632/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/32/877632/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/32/877632/2"}}},"commit":{"parents":[{"commit":"7733f0c5364de3805d7f8d5be429e0a464a7c6b1","subject":"Define \"LbAddIpPortMappingCommand\" with latin chars","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/7733f0c5364de3805d7f8d5be429e0a464a7c6b1"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-03-16 14:29:18.000000000","tz":60},"committer":{"name":"yatin","email":"ykarel@redhat.com","date":"2023-03-17 06:24:21.000000000","tz":0},"subject":"[CI] Set the default OS_TEST_PATH in the stestr conf file","message":"[CI] Set the default OS_TEST_PATH in the stestr conf file\n\nSet the default OS_TEST_PATH in the stestr configuration file. Any\nother test environment can overwrite this variable in the \"setenv\"\nsection.\n\nChange-Id: I51564920cec916031aaebf8151b015a3d87e9471\n(cherry picked from commit 2ed128014898c357063427e9347124835ff5384c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/cc45eb8268f570d3df744663ee889ea6f6a7288b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/cc45eb8268f570d3df744663ee889ea6f6a7288b"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
