)]}'
{"id":"openstack%2Fovsdbapp~890328","triplet_id":"openstack%2Fovsdbapp~master~I8d70d24beefb2813c977e9f89aa0535c3494ee85","project":"openstack/ovsdbapp","branch":"master","topic":"bug/2015471","attention_set":{},"removed_from_attention_set":{"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2023-11-16 20:37:14.000000000","reason":"Change was submitted"},"5756":{"account":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"last_update":"2023-11-16 17:56:11.000000000","reason":"\u003cGERRIT_ACCOUNT_5756\u003e replied on the change","reason_account":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}}},"hashtags":[],"change_id":"I8d70d24beefb2813c977e9f89aa0535c3494ee85","subject":"TAAS: Add commands for creating Mirrors","status":"MERGED","created":"2023-08-02 11:56:43.000000000","updated":"2023-11-16 20:38:14.000000000","submitted":"2023-11-16 20:37:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":42,"unresolved_comment_count":3,"has_review_started":true,"submission_id":"890328-bug/2015471","meta_rev_id":"513e9044c69ad6738896c400bd1f78dff7c460dc","_number":890328,"virtual_id_number":890328,"owner":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-16 20:37:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":2,"date":"2023-11-16 19:38:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":2,"date":"2023-11-16 17:56:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":1,"date":"2023-11-16 19:38:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-02 13:05:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-04 17:14:05.000000000","updated_by":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"reviewer":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"state":"REVIEWER"},{"updated":"2023-11-16 19:38:12.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"9d884bf8af0f63df11138e6ff2a07f3ccfd91982","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-08-02 11:56:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"79aaa3b30e0e2bb53bf9de8594608d417bf4085d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-08-02 12:04:12.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"b79a45e91cfef1c13e724f25a9470be96ffd3ade","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-02 13:05:11.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/97f7ef5672f048148ddf441c3c853f28\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/90a7de0449574fc69f28d86c534d0651 : SUCCESS in 3m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d118b52d84374ce8bacdcbc5d14259c3 : SUCCESS in 3m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/26140b315a4b47b68abf5d4961463e39 : SUCCESS in 2m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7726a6644cf842a1acb94d7e9aa34a91 : SUCCESS in 3m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/89d109fcf5e244c794e0bdc2bf7f0704 : SUCCESS in 4m 57s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/16414aa6c2ce43ae9c2131156605daf7 : SUCCESS in 5m 19s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/ab52d51abebe4216b514df28d12c0e20 : SUCCESS in 59m 22s","accounts_in_message":[],"_revision_number":2},{"id":"dc240e85eb145458463d67e842ad476e96c3563d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-08-02 13:12:37.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"e831f86c9210677dc8d0006eb71a5033fdfd960f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-02 14:13:29.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a013ccec430345caad277c6f5363d1c4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/928b0f3c82aa49628a7c596bfd96dde3 : SUCCESS in 4m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3b09328d68f54f489d5d6ff1a0abfd5e : SUCCESS in 3m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fb1ec5f1e70a4a6a9036e28218fdb910 : SUCCESS in 2m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2c0652ab04a249749472e24edb76aaed : SUCCESS in 3m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06a33cc9cde340e595ed7e5dfb31abf0 : SUCCESS in 4m 58s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/71f66682a3024ba3be663149eaf82154 : FAILURE in 5m 40s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/05c7df99a737483b87648cad64de8a2d : SUCCESS in 1h 00m 16s","accounts_in_message":[],"_revision_number":3},{"id":"4153ad64eb2fe8e185d23a50582bc775b65a99e2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-08-04 15:16:37.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"379bf7234f240abf3132dc5ff25901e0afae19c8","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-08-04 15:18:01.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"d698b5f6ef85224cb95e84f40926a0a5ce41965a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-08-04 15:19:32.000000000","message":"Patch Set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"51d0aa27a87e8668eb5e0ba75d2e191919b012f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-04 16:30:41.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a99543c7124a44669e77d8f6308cc331\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9835eb78ebc74edcbc7c1d11eb207323 : SUCCESS in 3m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6df1f94ab91242d79f0e93f9e21919ae : SUCCESS in 3m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9cd05ab3f76a4f1a9cc669eb185bd1b6 : SUCCESS in 3m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/656d7750c4714abea493b1c0c52e5344 : SUCCESS in 3m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a40fa8305ed4b22a45a9d1350fa0bf2 : SUCCESS in 4m 21s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/4d402efa4d1f4b64bcf48b98b8dd54ed : FAILURE in 5m 12s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/f662581e771c40699d876cf79a9fa97e : FAILURE in 1h 03m 01s","accounts_in_message":[],"_revision_number":5},{"id":"fcf125fdedd58178f7d2b2baf3b3715d3a8770f1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-08-10 11:23:19.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"7ddba9bf08668cc3f8cdd9ee05ad949dd07f9ee1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-10 12:41:19.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3163dc20acbb47c0b18bf25fade441af\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2e42ffe1bab04ac7832e02c8dd42b116 : SUCCESS in 3m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/885cb04035114899af7ebe099f6094dc : SUCCESS in 2m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/996a0dea76d5471e81bf6a7b954e6c1a : SUCCESS in 2m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/92e1cfbc716d4dd99aeaade917075135 : SUCCESS in 2m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/354cfe84e1d749af87acc0cc5c218c56 : SUCCESS in 4m 21s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/703abaabae964ca5ac430cec097dae57 : SUCCESS in 5m 05s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/c23d715144ef45fa8dd4d7baaee6f01a : SUCCESS in 1h 13m 03s","accounts_in_message":[],"_revision_number":6},{"id":"15f7272bbad3711c10ea133b6ce327cbdaff86fb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-09-01 12:41:45.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"3850637ec1d1bfd513e5dc20609a468532e5b695","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-01 13:59:28.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/47b14c3c0c4b4dc8be6bb881b4da345d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3512814dafef4cbeae0d457c8a8a937f : SUCCESS in 3m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/43cd71fd2a4548c7adfd84bbf53cdc87 : SUCCESS in 2m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1631ff9cff744b62b607fe93aea6ecc4 : SUCCESS in 3m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/80f1b742c4dd454699513e0550719772 : SUCCESS in 2m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/11391c8aca92425ba8c49e5590341f4e : SUCCESS in 3m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0c546f0f5324a67ae0c9b72ba224367 : SUCCESS in 4m 40s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/287146e33ff7497486be561fdf14160a : FAILURE in 5m 16s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/bbfcd674f78f42f781ea5f30e2c34f71 : SUCCESS in 1h 11m 55s","accounts_in_message":[],"_revision_number":7},{"id":"759f1d3a7abdb3369e4b015a9c92cbab097d92ff","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-09-13 14:55:16.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"6ee8a5c5b865ea98a294d4a3789ea47f8318b69d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-13 16:17:49.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3840d9e3c0f14c9eb5bd53feb4568092\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2dfcfc68c52049efa05c093b6b86e0ae : SUCCESS in 3m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9b97340816b2466bacc1d25e0c765f39 : SUCCESS in 3m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/672a58dfedac4aac902564449dff3348 : SUCCESS in 3m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fff57c2255fd4a41a85f400e0ce532a2 : SUCCESS in 5m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2612b1d079834f12944ac437a7006e60 : SUCCESS in 3m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cca347c7fc2e4ca3acafff8e976f14dc : SUCCESS in 4m 47s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/d923ecfcfe824829852507e196be3745 : SUCCESS in 6m 20s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/5084658de0c145ea8d4c190756438f01 : SUCCESS in 1h 19m 32s","accounts_in_message":[],"_revision_number":8},{"id":"8aedc26b439d6be17518c00e4e4af6e2dcd191e0","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2023-10-04 17:14:05.000000000","message":"Patch Set 8: Code-Review-1\n\n(13 comments)","accounts_in_message":[],"_revision_number":8},{"id":"4b4fb5917d78579512a7e7746c9da8b821d475dd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-10-10 11:48:26.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":9},{"id":"5bd37a256688e5ad6630817f35cb3193c8a6e94f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-10-10 12:20:30.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"0ddaf57a44609eda1668409b32f2a0cf6ca2c279","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-10-10 12:25:02.000000000","message":"Patch Set 10:\n\n(11 comments)","accounts_in_message":[],"_revision_number":10},{"id":"052cf0a9fc8a79d3b70361c9e25e9f9597807fed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-10 13:39:08.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/833f309ac73543f6a3747f1f4b0e0c34\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6d7ebee8c5994dfcb3d3ebb103e48e1e : SUCCESS in 2m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0550549c05bb43088a583013703ee97f : SUCCESS in 3m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/eefb46cccdbf4756bc4533db22cdc7ae : SUCCESS in 2m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/adb22dff8d624f198f1f0922b2fe720d : SUCCESS in 2m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f0eba1f0a4e7419f8b35d4e34c1d738f : SUCCESS in 3m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/80dc3b5d95204971b2863643ae9eb21c : SUCCESS in 4m 31s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/8989683c87074a28bc9421af1bc0b109 : SUCCESS in 4m 43s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/7a0a1c9e42a9432f90413635c2a19173 : SUCCESS in 1h 18m 01s","accounts_in_message":[],"_revision_number":10},{"id":"dac3bacc84c5df21e80b97d9484323775e6f7d31","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2023-10-20 21:26:37.000000000","message":"Patch Set 10: Code-Review-1\n\n(5 comments)","accounts_in_message":[],"_revision_number":10},{"id":"3ffc3d26e2af8375d4c84c8b8a3c271d8306d53c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-10-31 15:13:52.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":11},{"id":"34940d2b9c2517847b5d7747926ddc7a539a62d8","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-10-31 15:27:18.000000000","message":"Patch Set 11:\n\n(3 comments)","accounts_in_message":[],"_revision_number":11},{"id":"5fc30e2350b2c17685c0b1d8ea8d73f8560ab621","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-31 16:46:19.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/29e90dd1d537432f8c8dc10d12dd8821\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6eee3a26980e43e4ae52b2c6c430ce2a : SUCCESS in 3m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/321057f839da402da6c59256a14f744d : SUCCESS in 3m 02s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5d8a6a17c08742f1891960eb5ecf905e : SUCCESS in 3m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4644724883ce48dc991ac1ec0d853674 : SUCCESS in 5m 09s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/2e580ebade5742a1b103a164dc8ae848 : SUCCESS in 5m 36s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/59447a67bec145a98914ea09e3af822d : SUCCESS in 1h 01m 48s","accounts_in_message":[],"_revision_number":11},{"id":"4e690b795881d6cdbdf032ca8a3f02f97eb39e88","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2023-11-16 17:56:11.000000000","message":"Patch Set 11: Code-Review+2\n\n(9 comments)","accounts_in_message":[],"_revision_number":11},{"id":"66030f96f91edcf4ff31780907519a8f61d540d8","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-11-16 19:38:12.000000000","message":"Patch Set 11: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":11},{"id":"8994ae2550f25771dfc111f7e5eba626a3f31200","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-16 19:38:47.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"1e42e32c442a712dc2c4de92c3fe48456d6d8c65","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-16 20:37:14.000000000","message":"Patch Set 11: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d1f5f84cb5894773983be06ea746d92b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2a051edb9246422089c3a47e36f6472b : SUCCESS in 3m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/45af3ee979824b64a9cfe31eb10ac179 : SUCCESS in 3m 08s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e4588357ed394cd7bf46ca0213f3b363 : SUCCESS in 3m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/546ea43b9651423d9131805e470c8169 : SUCCESS in 6m 00s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/bfcd28c190344189acf040d971c94ea8 : SUCCESS in 5m 05s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/8773fd9a203a48e3b3b6138413d64d35 : SUCCESS in 57m 04s","accounts_in_message":[],"_revision_number":11},{"id":"60d9a3188c3146b00add7c5e4326f75b95e82695","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-16 20:37:14.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":11},{"id":"513e9044c69ad6738896c400bd1f78dff7c460dc","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-16 20:38:14.000000000","message":"Patch Set 11:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/875bc2ea718b48bd9747a3448e9e67e9\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe64945f0ed846d0b530efaaaf576516 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":11}],"current_revision_number":11,"current_revision":"02a7b2b7a881e89ee2c1327ba2330f055e0ea4b3","revisions":{"065e0ba15943228ddf9e85c01695b5ee4df48d33":{"kind":"REWORK","_number":1,"created":"2023-08-02 11:56:43.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/28/890328/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/28/890328/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/28/890328/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/28/890328/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/28/890328/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/28/890328/1"}}},"commit":{"parents":[{"commit":"6f7aed70982ba33ccf5a76a02bf9f9fa659e505c","subject":"Add some basic usage documentation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/6f7aed70982ba33ccf5a76a02bf9f9fa659e505c"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2023-07-28 07:42:26.000000000","tz":120},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2023-08-02 11:55:41.000000000","tz":120},"subject":"TAAS: Add commands for creating Mirrors","message":"TAAS: Add commands for creating Mirrors\n\nChange-Id: I8d70d24beefb2813c977e9f89aa0535c3494ee85\nRelated-Bug: #2015471\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/065e0ba15943228ddf9e85c01695b5ee4df48d33"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/065e0ba15943228ddf9e85c01695b5ee4df48d33"}]},"branch":"refs/heads/master"},"cf168144b57231344d0f15d29752a12285194984":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2023-08-02 12:04:12.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/28/890328/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/28/890328/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/28/890328/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/28/890328/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/28/890328/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/28/890328/2"}}},"commit":{"parents":[{"commit":"6f7aed70982ba33ccf5a76a02bf9f9fa659e505c","subject":"Add some basic usage documentation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/6f7aed70982ba33ccf5a76a02bf9f9fa659e505c"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2023-07-28 07:42:26.000000000","tz":120},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2023-08-02 12:03:49.000000000","tz":120},"subject":"TAAS: Add commands for creating Mirrors","message":"TAAS: Add commands for creating Mirrors\n\nTODO: tests\n\nChange-Id: I8d70d24beefb2813c977e9f89aa0535c3494ee85\nRelated-Bug: #2015471\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/cf168144b57231344d0f15d29752a12285194984"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/cf168144b57231344d0f15d29752a12285194984"}]},"branch":"refs/heads/master"},"95616447798270f723b06aef6ecbb18ccd6c7d7d":{"kind":"REWORK","_number":3,"created":"2023-08-02 13:12:37.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/28/890328/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/28/890328/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/28/890328/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/28/890328/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/28/890328/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/28/890328/3"}}},"commit":{"parents":[{"commit":"6f7aed70982ba33ccf5a76a02bf9f9fa659e505c","subject":"Add some basic usage documentation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/6f7aed70982ba33ccf5a76a02bf9f9fa659e505c"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2023-07-28 07:42:26.000000000","tz":120},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2023-08-02 13:12:29.000000000","tz":120},"subject":"TAAS: Add commands for creating Mirrors","message":"TAAS: Add commands for creating Mirrors\n\nTODO: tests\n\nChange-Id: I8d70d24beefb2813c977e9f89aa0535c3494ee85\nRelated-Bug: #2015471\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/95616447798270f723b06aef6ecbb18ccd6c7d7d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/95616447798270f723b06aef6ecbb18ccd6c7d7d"}]},"branch":"refs/heads/master"},"97cdd77f06ef95ea94213b4a2d59d4c3f078b33d":{"kind":"REWORK","_number":4,"created":"2023-08-04 15:16:37.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/28/890328/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/28/890328/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/28/890328/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/28/890328/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/28/890328/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/28/890328/4"}}},"commit":{"parents":[{"commit":"6f7aed70982ba33ccf5a76a02bf9f9fa659e505c","subject":"Add some basic usage documentation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/6f7aed70982ba33ccf5a76a02bf9f9fa659e505c"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2023-07-28 07:42:26.000000000","tz":120},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2023-08-04 15:16:26.000000000","tz":120},"subject":"TAAS: Add commands for creating Mirrors","message":"TAAS: Add commands for creating Mirrors\n\nTODO: tests\n\nChange-Id: I8d70d24beefb2813c977e9f89aa0535c3494ee85\nRelated-Bug: #2015471\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/97cdd77f06ef95ea94213b4a2d59d4c3f078b33d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/97cdd77f06ef95ea94213b4a2d59d4c3f078b33d"}]},"branch":"refs/heads/master"},"0e8be3055e4151f0eb706f1e18d3af14a030e4f4":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2023-08-04 15:19:32.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/28/890328/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/28/890328/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/28/890328/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/28/890328/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/28/890328/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/28/890328/5"}}},"commit":{"parents":[{"commit":"6f7aed70982ba33ccf5a76a02bf9f9fa659e505c","subject":"Add some basic usage documentation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/6f7aed70982ba33ccf5a76a02bf9f9fa659e505c"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2023-07-28 07:42:26.000000000","tz":120},"committer":{"name":"Lajos Katona","email":"katonalala@gmail.com","date":"2023-08-04 15:19:32.000000000","tz":0},"subject":"TAAS: Add commands for creating Mirrors","message":"TAAS: Add commands for creating Mirrors\n\nNote: mirrors are available only from OVN v22.12.0\n\nChange-Id: I8d70d24beefb2813c977e9f89aa0535c3494ee85\nRelated-Bug: #2015471\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/0e8be3055e4151f0eb706f1e18d3af14a030e4f4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/0e8be3055e4151f0eb706f1e18d3af14a030e4f4"}]},"branch":"refs/heads/master","description":"Edit commit message"},"6c7ed6d88b5dc6a2280098af4227ce2bac383d66":{"kind":"REWORK","_number":6,"created":"2023-08-10 11:23:19.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/28/890328/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/28/890328/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/28/890328/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/28/890328/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/28/890328/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/28/890328/6"}}},"commit":{"parents":[{"commit":"6f7aed70982ba33ccf5a76a02bf9f9fa659e505c","subject":"Add some basic usage documentation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/6f7aed70982ba33ccf5a76a02bf9f9fa659e505c"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2023-07-28 07:42:26.000000000","tz":120},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2023-08-10 11:22:47.000000000","tz":120},"subject":"TAAS: Add commands for creating Mirrors","message":"TAAS: Add commands for creating Mirrors\n\nNote: mirrors are available only from OVN v22.12.0\n\nChange-Id: I8d70d24beefb2813c977e9f89aa0535c3494ee85\nRelated-Bug: #2015471\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/6c7ed6d88b5dc6a2280098af4227ce2bac383d66"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/6c7ed6d88b5dc6a2280098af4227ce2bac383d66"}]},"branch":"refs/heads/master"},"e83e6dfe495ff5c40788f53f4c748cf1ad2da5c0":{"kind":"REWORK","_number":7,"created":"2023-09-01 12:41:45.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/28/890328/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/28/890328/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/28/890328/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/28/890328/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/28/890328/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/28/890328/7"}}},"commit":{"parents":[{"commit":"6f7aed70982ba33ccf5a76a02bf9f9fa659e505c","subject":"Add some basic usage documentation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/6f7aed70982ba33ccf5a76a02bf9f9fa659e505c"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2023-07-28 07:42:26.000000000","tz":120},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2023-09-01 12:37:01.000000000","tz":120},"subject":"TAAS: Add commands for creating Mirrors","message":"TAAS: Add commands for creating Mirrors\n\nNote: mirrors are available only from OVN v22.12.0\n\nChange-Id: I8d70d24beefb2813c977e9f89aa0535c3494ee85\nRelated-Bug: #2015471\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/e83e6dfe495ff5c40788f53f4c748cf1ad2da5c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/e83e6dfe495ff5c40788f53f4c748cf1ad2da5c0"}]},"branch":"refs/heads/master"},"32245c5527fa7439b260c73ceb65acdff801f2f6":{"kind":"REWORK","_number":8,"created":"2023-09-13 14:55:16.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/28/890328/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/28/890328/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/28/890328/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/28/890328/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/28/890328/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/28/890328/8"}}},"commit":{"parents":[{"commit":"6f7aed70982ba33ccf5a76a02bf9f9fa659e505c","subject":"Add some basic usage documentation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/6f7aed70982ba33ccf5a76a02bf9f9fa659e505c"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2023-07-28 07:42:26.000000000","tz":120},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2023-09-13 14:55:05.000000000","tz":120},"subject":"TAAS: Add commands for creating Mirrors","message":"TAAS: Add commands for creating Mirrors\n\nNote: mirrors are available only from OVN v22.12.0\n\nChange-Id: I8d70d24beefb2813c977e9f89aa0535c3494ee85\nRelated-Bug: #2015471\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/32245c5527fa7439b260c73ceb65acdff801f2f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/32245c5527fa7439b260c73ceb65acdff801f2f6"}]},"branch":"refs/heads/master"},"9cea2476ead014fdb915c77b23d0ec03c2e708ea":{"kind":"REWORK","_number":9,"created":"2023-10-10 11:48:26.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/28/890328/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/28/890328/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/28/890328/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/28/890328/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/28/890328/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/28/890328/9"}}},"commit":{"parents":[{"commit":"6f7aed70982ba33ccf5a76a02bf9f9fa659e505c","subject":"Add some basic usage documentation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/6f7aed70982ba33ccf5a76a02bf9f9fa659e505c"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2023-07-28 07:42:26.000000000","tz":120},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2023-10-10 11:41:14.000000000","tz":120},"subject":"TAAS: Add commands for creating Mirrors","message":"TAAS: Add commands for creating Mirrors\n\nNote: mirrors are available only from OVN v22.12.0\n\nChange-Id: I8d70d24beefb2813c977e9f89aa0535c3494ee85\nRelated-Bug: #2015471\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/9cea2476ead014fdb915c77b23d0ec03c2e708ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/9cea2476ead014fdb915c77b23d0ec03c2e708ea"}]},"branch":"refs/heads/master"},"225505202c77cfa9973442a078e3abdd7e95c8ef":{"kind":"REWORK","_number":10,"created":"2023-10-10 12:20:30.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/28/890328/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/28/890328/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/28/890328/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/28/890328/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/28/890328/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/28/890328/10"}}},"commit":{"parents":[{"commit":"6f7aed70982ba33ccf5a76a02bf9f9fa659e505c","subject":"Add some basic usage documentation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/6f7aed70982ba33ccf5a76a02bf9f9fa659e505c"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2023-07-28 07:42:26.000000000","tz":120},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2023-10-10 12:20:21.000000000","tz":120},"subject":"TAAS: Add commands for creating Mirrors","message":"TAAS: Add commands for creating Mirrors\n\nNote: mirrors are available only from OVN v22.12.0\n\nChange-Id: I8d70d24beefb2813c977e9f89aa0535c3494ee85\nRelated-Bug: #2015471\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/225505202c77cfa9973442a078e3abdd7e95c8ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/225505202c77cfa9973442a078e3abdd7e95c8ef"}]},"branch":"refs/heads/master"},"02a7b2b7a881e89ee2c1327ba2330f055e0ea4b3":{"kind":"REWORK","_number":11,"created":"2023-10-31 15:13:52.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/28/890328/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/28/890328/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/28/890328/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/28/890328/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/28/890328/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/28/890328/11"}}},"commit":{"parents":[{"commit":"6f7aed70982ba33ccf5a76a02bf9f9fa659e505c","subject":"Add some basic usage documentation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/6f7aed70982ba33ccf5a76a02bf9f9fa659e505c"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2023-07-28 07:42:26.000000000","tz":120},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2023-10-31 15:13:10.000000000","tz":60},"subject":"TAAS: Add commands for creating Mirrors","message":"TAAS: Add commands for creating Mirrors\n\nNote: mirrors are available only from OVN v22.12.0\n\nChange-Id: I8d70d24beefb2813c977e9f89aa0535c3494ee85\nRelated-Bug: #2015471\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/02a7b2b7a881e89ee2c1327ba2330f055e0ea4b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/02a7b2b7a881e89ee2c1327ba2330f055e0ea4b3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
