)]}'
{"id":"openstack%2Fovsdbapp~892459","triplet_id":"openstack%2Fovsdbapp~stable%2Ftrain~If5bb7e03cb5098f34c29ee620867c341d3ec078f","project":"openstack/ovsdbapp","branch":"stable/train","topic":"zuul-config-errors","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2023-08-24 07:23:37.000000000","reason":"\u003cGERRIT_ACCOUNT_16688\u003e replied on the change","reason_account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2023-08-24 07:33:46.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If5bb7e03cb5098f34c29ee620867c341d3ec078f","subject":"Remove ovsdbapp-tempest-dsvm-networking-ovn-ovs-release job definition","status":"MERGED","created":"2023-08-23 07:11:58.000000000","updated":"2023-08-24 07:33:46.000000000","submitted":"2023-08-24 07:33:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"892459-zuul-config-errors","meta_rev_id":"728d4247c7bee448dc48e9dd90f4ce96ea02f029","_number":892459,"virtual_id_number":892459,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-24 07:33:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2023-08-24 07:23:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2023-08-23 12:25:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2023-08-24 07:23:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-23 07:17:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-23 07:36:40.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2023-08-23 12:25:09.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"}],"messages":[{"id":"2b9f92aaa6242a8f5c2d36994d3f09eeb1d355e8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-08-23 07:11:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d2ad28394b8e6c23c9f8d3f85eb1a8bd8453ae55","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-23 07:17:35.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9b8becb1d2a443f984d730d27be2ee3a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d5714e3d9d64ec3a43e4764663d8e81 : SUCCESS in 3m 32s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/015d0c84ce1f454f859592e0560edcce : SUCCESS in 3m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/59b2058e4e2a4d0682239a5819d944d7 : SUCCESS in 3m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/12e1993ca6e54f028acecf762425ee86 : SUCCESS in 3m 14s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/83d3cde60ce24749bdd08e5681057c36 : FAILURE in 5m 00s","accounts_in_message":[],"_revision_number":1},{"id":"f42d7589a040489ff3f5b994386790b55ba08d77","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-08-23 07:36:40.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bda5b8052b40b8a33ea29473f46d494876450a1b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-08-23 09:27:49.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"e420a91b24b8f3fe5632b612f66abb01e061878b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-23 09:38:17.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f9d90344357047b2ace77494c9caa73c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c83219adefcb45db96d2914943ed901e : SUCCESS in 3m 10s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a5280f00db3d4b829cff053f6c26b5b7 : SUCCESS in 3m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6a7a3a4fc7e34659aee9657b95e4812f : SUCCESS in 3m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3ea0fee8f0f54ac98899def88bcc1da3 : SUCCESS in 4m 13s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/50c1aea7eec74622843a9ed0422e1b7e : SUCCESS in 5m 31s","accounts_in_message":[],"_revision_number":2},{"id":"df4b9656c83087992dc1cdc05aae70db9ded022d","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-08-23 12:25:09.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"bc726d525dacb44ca4df2ff7f64dea28f7fd90db","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-08-24 07:23:37.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"6f29b265e488a4cf9e878183f643e7000d637c76","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-24 07:24:02.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"6f02aac3d6eebcabd716c6d09682f945253193e7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-24 07:33:46.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/68a6aee28fa4498ab84581454f45d747\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/22dcc26991d64f7ea5239e836168c299 : SUCCESS in 3m 10s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6cb103b429a849c5b39ef2a3c976ed6b : SUCCESS in 4m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a426e964428d4fc5b2f5277ae9278000 : SUCCESS in 3m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9b9e14c0832b4227abe38bd09e5f4c3e : SUCCESS in 3m 28s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/dcb5d3927288407290285941794529f4 : SUCCESS in 7m 21s","accounts_in_message":[],"_revision_number":2},{"id":"728d4247c7bee448dc48e9dd90f4ce96ea02f029","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-24 07:33:46.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"fdfd76e420e3d2c55e86d1a9f8a0a01a641137e8","revisions":{"1f181db820e01b3c4051c2b9206ef1f47f2fc966":{"kind":"REWORK","_number":1,"created":"2023-08-23 07:11:58.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/59/892459/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/59/892459/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/59/892459/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/59/892459/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/59/892459/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/59/892459/1"}}},"commit":{"parents":[{"commit":"f22405c28722f105451b9acb12d37c036e1e1ef5","subject":"Fix logging exception while handling transaction exception","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/f22405c28722f105451b9acb12d37c036e1e1ef5"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2023-08-23 07:10:08.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2023-08-23 07:11:56.000000000","tz":120},"subject":"Remove ovsdbapp-tempest-dsvm-networking-ovn-ovs-release job definition","message":"Remove ovsdbapp-tempest-dsvm-networking-ovn-ovs-release job definition\n\nParent of this job (networking-ovn-tempest-dsvm-ovs-release) was defined\nin the networking-ovn repo in stable/train branch which is now EOL.\nBecause of that this is now causing zuul config error as\nnetworking-ovn-tempest-dsvm-ovs-release don\u0027t exists anymore.\nLets delete ovsdbapp-tempest-dsvm-networking-ovn-ovs-release job\ndefinition and keep just basic jobs in stable/train branch.\n\nChange-Id: If5bb7e03cb5098f34c29ee620867c341d3ec078f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/1f181db820e01b3c4051c2b9206ef1f47f2fc966"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/1f181db820e01b3c4051c2b9206ef1f47f2fc966"}]},"branch":"refs/heads/stable/train"},"fdfd76e420e3d2c55e86d1a9f8a0a01a641137e8":{"kind":"REWORK","_number":2,"created":"2023-08-23 09:27:49.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/59/892459/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/59/892459/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/59/892459/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/59/892459/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/59/892459/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/59/892459/2"}}},"commit":{"parents":[{"commit":"f22405c28722f105451b9acb12d37c036e1e1ef5","subject":"Fix logging exception while handling transaction exception","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/f22405c28722f105451b9acb12d37c036e1e1ef5"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2023-08-23 07:10:08.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2023-08-23 09:27:09.000000000","tz":120},"subject":"Remove ovsdbapp-tempest-dsvm-networking-ovn-ovs-release job definition","message":"Remove ovsdbapp-tempest-dsvm-networking-ovn-ovs-release job definition\n\nParent of this job (networking-ovn-tempest-dsvm-ovs-release) was defined\nin the networking-ovn repo in stable/train branch which is now EOL.\nBecause of that this is now causing zuul config error as\nnetworking-ovn-tempest-dsvm-ovs-release don\u0027t exists anymore.\nLets delete ovsdbapp-tempest-dsvm-networking-ovn-ovs-release job\ndefinition and keep just basic jobs in stable/train branch.\n\nAdditionally this patch pinned OVS and OVN branch in the functional\ntests job to the same versions as were used in the networking-ovn jobs\nin stable/train branch.\n\nChange-Id: If5bb7e03cb5098f34c29ee620867c341d3ec078f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/fdfd76e420e3d2c55e86d1a9f8a0a01a641137e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/fdfd76e420e3d2c55e86d1a9f8a0a01a641137e8"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
