)]}'
{"id":"openstack%2Fovsdbapp~896285","triplet_id":"openstack%2Fovsdbapp~master~I15b46e3457bf60c50c1efce260d60b8a1ad3771f","project":"openstack/ovsdbapp","branch":"master","attention_set":{},"removed_from_attention_set":{"5756":{"account":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"last_update":"2023-09-22 20:39:36.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I15b46e3457bf60c50c1efce260d60b8a1ad3771f","subject":"Don\u0027t silently fail events w/ broken definitions","status":"ABANDONED","created":"2023-09-22 18:16:12.000000000","updated":"2023-09-22 20:39:36.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"123f26af03a0f02d870a00fa29a730329946be90","_number":896285,"virtual_id_number":896285,"owner":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-09-22 19:09:57.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2023-09-22 18:17:16.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-22 19:09:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"486b9fa333c43150374f5f0e4c0f2ef0e44b139e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2023-09-22 18:16:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"adb402d4c1fbde716670e17d4430202e94dbe3ad","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2023-09-22 18:17:16.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e98b0fd34bf8d10a5cf1817c6f209d31c7c7a322","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-22 19:09:57.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3b4c17f73d71477eb4bfb0e81a2a06af\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/70429db575ef419996ed231cabc0ceff : SUCCESS in 3m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b3c75d31eb7c48d8b4f50bf334d69ad2 : SUCCESS in 2m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/56a0916004494689bc6aea2ec07fd532 : SUCCESS in 2m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a97d69ea8d944c868be216d2cf6b3125 : SUCCESS in 2m 54s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ea3c762409dc4ddfb7f14f2ccfdec68e : SUCCESS in 3m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c89126cdaec14a21acd4389beb9a02ea : SUCCESS in 5m 36s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/a2e6d4f57e2e4302bd55963ac97aba09 : SUCCESS in 5m 20s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/d7341feac24f4daaa514e5e18a3bcb30 : FAILURE in 48m 15s","accounts_in_message":[],"_revision_number":1},{"id":"123f26af03a0f02d870a00fa29a730329946be90","tag":"autogenerated:gerrit:abandon","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2023-09-22 20:39:36.000000000","message":"Abandoned\n\nThis won\u0027t work because external code checks things like `event \u003d\u003d self.ROW_CREATE` and the types would no longer match.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"cbd49acbc20ce0ce798e9e6a30fda207bd0bdc38","revisions":{"cbd49acbc20ce0ce798e9e6a30fda207bd0bdc38":{"kind":"REWORK","_number":1,"created":"2023-09-22 18:16:12.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/85/896285/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/85/896285/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/85/896285/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/85/896285/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/85/896285/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/85/896285/1"}}},"commit":{"parents":[{"commit":"6f7aed70982ba33ccf5a76a02bf9f9fa659e505c","subject":"Add some basic usage documentation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/6f7aed70982ba33ccf5a76a02bf9f9fa659e505c"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2023-09-22 18:09:18.000000000","tz":-300},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2023-09-22 18:09:18.000000000","tz":-300},"subject":"Don\u0027t silently fail events w/ broken definitions","message":"Don\u0027t silently fail events w/ broken definitions\n\nIt is a fairly common error to define a RowEvent with either\n\nevents\u003d(RowEvent.ROW_CREATE)  # missing comma, so a string\n\nor\n\nevents\u003dtupe(RowEvent.ROW_CREATE)  # (\u0027c\u0027, \u0027r\u0027, \u0027e\u0027, \u0027a\u0027, \u0027t\u0027, \u0027e\u0027)\n\nUsing an Enum makes it easy to distinguish between a single value\nbeing passed, as they are not iterable like strings, so we can\njust handle that case. If someone tried to just pass as string\ninstead of the defined values, we can throw an AttributeError and\nmake them fix their code.\n\nChange-Id: I15b46e3457bf60c50c1efce260d60b8a1ad3771f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/cbd49acbc20ce0ce798e9e6a30fda207bd0bdc38"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/cbd49acbc20ce0ce798e9e6a30fda207bd0bdc38"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Workflow\u003dMIN"],"failing_atoms":["label:Workflow\u003dMAX"],"atom_explanations":{}}}]}
