)]}'
{"id":"openstack%2Fovsdbapp~896645","triplet_id":"openstack%2Fovsdbapp~master~I84f234c6c5bbf4d318893166943de62fb87fff69","project":"openstack/ovsdbapp","branch":"master","topic":"gerrit/del-ecmp-route","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2023-10-05 10:40:26.000000000","reason":"\u003cGERRIT_ACCOUNT_16688\u003e replied on the change","reason_account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},"5756":{"account":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"last_update":"2023-10-05 13:23:55.000000000","reason":"\u003cGERRIT_ACCOUNT_5756\u003e replied on the change","reason_account":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}},"33871":{"account":{"_account_id":33871,"name":"Anton Vazhnetsov","display_name":"Anton Vazhnetsov","email":"dragen15051@gmail.com","username":"0x5b"},"last_update":"2023-10-11 15:33:29.000000000","reason":"Change was submitted"},"23567":{"account":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"last_update":"2023-10-11 15:33:29.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I84f234c6c5bbf4d318893166943de62fb87fff69","subject":"nb: add \u0027nexthop\u0027 argument to \u0027lr_route_del\u0027","status":"MERGED","created":"2023-09-27 11:52:27.000000000","updated":"2023-10-11 15:34:37.000000000","submitted":"2023-10-11 15:33:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":19,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"896645-gerrit/del-ecmp-route","meta_rev_id":"1180b66b65629392cbda77d0505e8d44d230d110","_number":896645,"virtual_id_number":896645,"owner":{"_account_id":33871,"name":"Anton Vazhnetsov","display_name":"Anton Vazhnetsov","email":"dragen15051@gmail.com","username":"0x5b"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-11 15:33:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2023-10-11 14:12:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-10-05 13:23:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2023-10-11 14:12:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-27 12:56:12.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2023-09-27 13:03:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-27 14:43:24.000000000","updated_by":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"reviewer":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"state":"REVIEWER"},{"updated":"2023-09-28 06:07:14.000000000","updated_by":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"reviewer":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"state":"REVIEWER"}],"messages":[{"id":"17c2c17a6ee18019ffd230e0972a588b6528e7d0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33871,"name":"Anton Vazhnetsov","display_name":"Anton Vazhnetsov","email":"dragen15051@gmail.com","username":"0x5b"},"date":"2023-09-27 11:52:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a8da098ed138d00524e400de6a6864b8413ee85d","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-09-27 12:56:12.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"54414fdea5fe43a5d62dd651d1e51ceb96bd3711","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-27 13:03:10.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0bc9b33be3f7435fa7f1b95da9036479\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/73ea451868e44e7ca36fcbf6f5fc370b : FAILURE in 3m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/44798e1c455941179fa65b7a6001e4a6 : SUCCESS in 3m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9bc4ab1ad38a46f2b3511c43f956e4ee : SUCCESS in 3m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a7c1d9e02a5747599ddd773b13837576 : SUCCESS in 2m 59s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2d041c9ffefc4a50b5354c326f5598ba : SUCCESS in 3m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/abdef99869924b28bd0034665a9b7553 : SUCCESS in 6m 23s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/970e9b245a5340e793e9ba6630660749 : SUCCESS in 5m 14s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/312757e8ae634e51a17f2f35b65d5203 : SUCCESS in 1h 09m 12s","accounts_in_message":[],"_revision_number":1},{"id":"8fe9f0922903b505ad983069dfe609caca07afed","author":{"_account_id":33871,"name":"Anton Vazhnetsov","display_name":"Anton Vazhnetsov","email":"dragen15051@gmail.com","username":"0x5b"},"date":"2023-09-27 14:26:09.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ec3edacadeed6c18506e1692392187f672c7680a","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2023-09-27 14:43:24.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c96a9ff00756c1828b08e90355d63715a8b4ee3f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33871,"name":"Anton Vazhnetsov","display_name":"Anton Vazhnetsov","email":"dragen15051@gmail.com","username":"0x5b"},"date":"2023-09-27 14:55:05.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":2},{"id":"7b30ea39433f5d7d891e79521c5e461f03d3837e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-27 15:54:29.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/64286936994f47c59fc080cf46e3cd4b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cfc6ca48342343568085bc1ba0bca556 : SUCCESS in 3m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7190c69f1dbf47358505d2ca2ea75535 : SUCCESS in 3m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e976c1b1f193426c98b9bd24f76fc14c : SUCCESS in 3m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9a951e2b047f4a9881209f8da0db7d38 : SUCCESS in 3m 33s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a0db2210b5234e1690db642333228864 : SUCCESS in 3m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b583bb14ad4f4af292a1ed3520e4f8ec : SUCCESS in 5m 01s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/54715771a96f40379ff535698d481d77 : SUCCESS in 7m 06s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/39dee83db8414c0eaecb4f40a0d627ca : SUCCESS in 57m 42s","accounts_in_message":[],"_revision_number":2},{"id":"5db31182c36c7bc1fac2a71427133d6d0e4173c1","author":{"_account_id":33871,"name":"Anton Vazhnetsov","display_name":"Anton Vazhnetsov","email":"dragen15051@gmail.com","username":"0x5b"},"date":"2023-09-27 21:19:26.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d4856110e0f97d7f48afb9c656b834996c6e3a96","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2023-09-28 06:07:14.000000000","message":"Patch Set 2: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"a6972a201ed4f303beaf4d4ad947a9f70758417d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33871,"name":"Anton Vazhnetsov","display_name":"Anton Vazhnetsov","email":"dragen15051@gmail.com","username":"0x5b"},"date":"2023-09-28 08:37:57.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"4460a22512dcc6359f77c55308feb5236c8ca9d2","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2023-09-28 09:41:31.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"f47b09518fcbca6c5ced656627abad409a251ff7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-28 09:52:32.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4288122a926c4535bb96c08985fa5ba8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/27fa9495473a4d0c9f13ec741d4b05f2 : SUCCESS in 3m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8337e8c9af5f48d58ab9fac1a8216219 : SUCCESS in 3m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3d46dc886d174d6ea2bf3d9f4d6357f0 : SUCCESS in 3m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/605dee5eecf940a9815556cab5e0860d : SUCCESS in 3m 57s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b5895daf83904b30bce6a4cc700ea08a : SUCCESS in 3m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ef1a5f444de4f4580435a69342a93f7 : SUCCESS in 4m 47s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/b4759a3926b4478aa3e234ece548f355 : SUCCESS in 6m 33s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/8039a6c931b54526a10bf36d9d96ca6d : SUCCESS in 1h 09m 27s","accounts_in_message":[],"_revision_number":3},{"id":"23058fc6e975f87fd2b62560e406b9c4742848e7","author":{"_account_id":33871,"name":"Anton Vazhnetsov","display_name":"Anton Vazhnetsov","email":"dragen15051@gmail.com","username":"0x5b"},"date":"2023-09-28 13:24:49.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"5ef90a14ad7ba3560465a574fce81f4516e74f89","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33871,"name":"Anton Vazhnetsov","display_name":"Anton Vazhnetsov","email":"dragen15051@gmail.com","username":"0x5b"},"date":"2023-10-01 07:18:03.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"ba5577262d2cab605e264fca30c7e3d82ddd6d34","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-01 08:07:52.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/947028e393464eb8b59f464a78cdf25e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/96cd3f65fce34cddafccb9c12694d591 : SUCCESS in 5m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0a5a300940154c908009f8a277f1651a : SUCCESS in 5m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/400140206b694d2590032acc9808ebfd : SUCCESS in 3m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d1252cdfb7a34ad6bedd3370c52a8bbe : SUCCESS in 3m 06s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/78ed2ae7e4f0490b98c594096f71a0b7 : SUCCESS in 3m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/092068efc86744cfa53df596b22a3e67 : SUCCESS in 7m 56s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/90e9e61a6ed84f0f82b0b4b4277d7120 : SUCCESS in 5m 12s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/e551db1a46d4437e8a316a63266a5634 : SUCCESS in 43m 43s","accounts_in_message":[],"_revision_number":4},{"id":"595628391f45d8409e66d60764ddd3577eb9de36","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2023-10-02 05:40:46.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"4ca6259ff9b8d3d1ae44c58a0be839657be14e1d","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-10-03 14:29:00.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"94c298647a10f0dae84db48538c1e0fd998ca928","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33871,"name":"Anton Vazhnetsov","display_name":"Anton Vazhnetsov","email":"dragen15051@gmail.com","username":"0x5b"},"date":"2023-10-03 15:28:31.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"aa6dfeb70e4ad0d571200e098ed007227501427f","author":{"_account_id":33871,"name":"Anton Vazhnetsov","display_name":"Anton Vazhnetsov","email":"dragen15051@gmail.com","username":"0x5b"},"date":"2023-10-03 15:28:54.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"377a7a03c5c8ea0387dad8b8f48151c8726872ed","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-10-03 15:32:30.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"274dad86c26b5ea1cf1e9b20c346374bd3e1dcd5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-03 16:46:50.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/98a44a6ecf194340b97d97ee2a03c79c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af39bb2921744d01887210feb89bc2ba : SUCCESS in 3m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9c0e8888409f40bc94bf5b33d785315d : SUCCESS in 2m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d3936d0ff29241f1a4fb60d267125eea : SUCCESS in 3m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d4eeba0ea2b54422b87593a37e608b0b : SUCCESS in 3m 10s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9af10bff3f1e4f779675cfa116943571 : SUCCESS in 3m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a71552551fc4150bbde2811bea7c4c5 : SUCCESS in 4m 53s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/5cf0c56d55e04a84a8a8c06a48641f65 : SUCCESS in 5m 33s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/ab353f7245f74ccda364c44919538fc1 : SUCCESS in 1h 10m 53s","accounts_in_message":[],"_revision_number":5},{"id":"42fc7ef778046ad5167c0902e7174648285a9efa","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2023-10-04 06:36:01.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"df65e16324c45f94278a2357625b1b7a525da749","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33871,"name":"Anton Vazhnetsov","display_name":"Anton Vazhnetsov","email":"dragen15051@gmail.com","username":"0x5b"},"date":"2023-10-04 15:09:58.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"d25c1569586ac7406de4462f34ac706b05abf176","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-10-04 15:12:58.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"c3961d41773bbbf97bf97df2e208c4fa54a55191","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-10-04 15:13:09.000000000","message":"Patch Set 6: Code-Review-1","accounts_in_message":[],"_revision_number":6},{"id":"183fd8450da923b736ca681ab51cab01a191c5e2","author":{"_account_id":33871,"name":"Anton Vazhnetsov","display_name":"Anton Vazhnetsov","email":"dragen15051@gmail.com","username":"0x5b"},"date":"2023-10-04 15:38:47.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"3cbc4f44cd9179f56306db95b1dabd68ddd1d5df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-04 16:09:41.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1e6a75de4aab4262bd3f4088faf9b1da\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/93a41495254c4fa38a7b179054acf9b9 : SUCCESS in 3m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8fee17b00c294bbe90d104868d224d26 : SUCCESS in 3m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3c25c4b2ec8d4ebf8d3c36bf8dd20db1 : SUCCESS in 3m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/97382598f7d3462ba7fcff3dd237412f : SUCCESS in 4m 19s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/30794b6562364f4d8b7feb3a12598fc0 : SUCCESS in 3m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dde290df697a4994a620e922e82287cb : SUCCESS in 6m 08s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/c5e2a7e1e3f24ae9999418f5e42b9f7c : SUCCESS in 5m 34s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/f125718a309543d89fa8f4d8973b4498 : SUCCESS in 58m 29s","accounts_in_message":[],"_revision_number":6},{"id":"ef48299abc50db1bf9b52bd0e34cecdeb5954d01","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2023-10-04 17:52:16.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"10879314af97d724605cb90b4371bb3bfef5fbfe","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2023-10-04 18:13:57.000000000","message":"Patch Set 6:\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"2e97c138d90ec15de3dab5cc4bc58f335b0df501","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2023-10-04 19:58:19.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"6f505b259b8bf34aa560559135ec129fc2a3359e","author":{"_account_id":33871,"name":"Anton Vazhnetsov","display_name":"Anton Vazhnetsov","email":"dragen15051@gmail.com","username":"0x5b"},"date":"2023-10-04 21:20:51.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"271554e43d02a3a2bfb718edfd192495290338d2","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-10-05 10:40:26.000000000","message":"Patch Set 6: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"5a68993232a6cd07ab0cc0be20b9e5167c194707","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2023-10-05 13:23:55.000000000","message":"Patch Set 6: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"3683a8816e250a86046d4304accbaa822b9649f8","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-10-11 14:12:07.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"5ccff4060fcc771d16c4759988b15c967d8a799d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-11 14:12:38.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"d78842fdb9039776c0793fa8c6dd8fa46a27bf16","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-11 15:33:29.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f1061ef3f1094cfcb9e882aa838b83a6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd04afd05b57455f90cf35c8faaf7cb5 : SUCCESS in 3m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b14b99434f294a66a8cfcc3f9786963f : SUCCESS in 3m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/893f92e9da9f474c9a1bde13a8225174 : SUCCESS in 3m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c0ef861eba1240deb21796f53cddfc52 : SUCCESS in 3m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bcc0d14592a54f5f84f235053e1b52c3 : SUCCESS in 5m 14s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/43d38b90bf67480aa5e137729feb0b8d : SUCCESS in 5m 53s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/0f784372850542d7be7d2a8628eb2baa : SUCCESS in 1h 12m 54s","accounts_in_message":[],"_revision_number":6},{"id":"177a8e21b31cd6d115414e754b8614378b89a91c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-11 15:33:29.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"1180b66b65629392cbda77d0505e8d44d230d110","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-11 15:34:37.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/982762e13bec4ce4b94b5c9b5c490753\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7eb03737cde4f129acb95a01f5104e3 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"033b6e3bd217c3e424e13f94893d1cb112abf87c","revisions":{"281cc82124ef0ce00a1a1a7786c68d1e419a8fc4":{"kind":"REWORK","_number":1,"created":"2023-09-27 11:52:27.000000000","uploader":{"_account_id":33871,"name":"Anton Vazhnetsov","display_name":"Anton Vazhnetsov","email":"dragen15051@gmail.com","username":"0x5b"},"ref":"refs/changes/45/896645/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/45/896645/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/45/896645/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/45/896645/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/45/896645/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/45/896645/1"}}},"commit":{"parents":[{"commit":"a39b4f8e62bcef958cc962adfad0103a3a41c1ae","subject":"Update master for stable/2023.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/a39b4f8e62bcef958cc962adfad0103a3a41c1ae"}]}],"author":{"name":"Anton Vazhnetsov","email":"dragen15051@gmail.com","date":"2023-09-26 17:01:59.000000000","tz":180},"committer":{"name":"Anton Vazhnetsov","email":"dragen15051@gmail.com","date":"2023-09-27 10:09:36.000000000","tz":180},"subject":"nb: add \u0027nexthop\u0027 argument to \u0027lr_route_del\u0027","message":"nb: add \u0027nexthop\u0027 argument to \u0027lr_route_del\u0027\n\nIn request https://review.opendev.org/c/openstack/ovsdbapp/+/882268, the ability to create ECMP routes was added. This PR adds the ability to delete these routes.\n\nCloses-bug: 2037536\n\nChange-Id: I84f234c6c5bbf4d318893166943de62fb87fff69\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/281cc82124ef0ce00a1a1a7786c68d1e419a8fc4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/281cc82124ef0ce00a1a1a7786c68d1e419a8fc4"}]},"branch":"refs/heads/master"},"a33db3562dec703a38ff8a9a70fc7b7eda6c152b":{"kind":"REWORK","_number":2,"created":"2023-09-27 14:55:05.000000000","uploader":{"_account_id":33871,"name":"Anton Vazhnetsov","display_name":"Anton Vazhnetsov","email":"dragen15051@gmail.com","username":"0x5b"},"ref":"refs/changes/45/896645/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/45/896645/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/45/896645/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/45/896645/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/45/896645/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/45/896645/2"}}},"commit":{"parents":[{"commit":"a39b4f8e62bcef958cc962adfad0103a3a41c1ae","subject":"Update master for stable/2023.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/a39b4f8e62bcef958cc962adfad0103a3a41c1ae"}]}],"author":{"name":"Anton Vazhnetsov","email":"dragen15051@gmail.com","date":"2023-09-26 17:01:59.000000000","tz":180},"committer":{"name":"Anton Vazhnetsov","email":"dragen15051@gmail.com","date":"2023-09-27 14:52:16.000000000","tz":180},"subject":"nb: add \u0027nexthop\u0027 argument to \u0027lr_route_del\u0027","message":"nb: add \u0027nexthop\u0027 argument to \u0027lr_route_del\u0027\n\nIn request https://review.opendev.org/c/openstack/ovsdbapp/+/882268, the ability to create ECMP routes was added. This PR adds the ability to delete these routes.\n\nCloses-bug: 2037536\n\nChange-Id: I84f234c6c5bbf4d318893166943de62fb87fff69\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/a33db3562dec703a38ff8a9a70fc7b7eda6c152b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/a33db3562dec703a38ff8a9a70fc7b7eda6c152b"}]},"branch":"refs/heads/master"},"d30e80475048585273c460612a1655d828434c6f":{"kind":"REWORK","_number":3,"created":"2023-09-28 08:37:57.000000000","uploader":{"_account_id":33871,"name":"Anton Vazhnetsov","display_name":"Anton Vazhnetsov","email":"dragen15051@gmail.com","username":"0x5b"},"ref":"refs/changes/45/896645/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/45/896645/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/45/896645/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/45/896645/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/45/896645/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/45/896645/3"}}},"commit":{"parents":[{"commit":"a39b4f8e62bcef958cc962adfad0103a3a41c1ae","subject":"Update master for stable/2023.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/a39b4f8e62bcef958cc962adfad0103a3a41c1ae"}]}],"author":{"name":"Anton Vazhnetsov","email":"dragen15051@gmail.com","date":"2023-09-26 17:01:59.000000000","tz":180},"committer":{"name":"Anton Vazhnetsov","email":"dragen15051@gmail.com","date":"2023-09-28 08:37:40.000000000","tz":180},"subject":"nb: add \u0027nexthop\u0027 argument to \u0027lr_route_del\u0027","message":"nb: add \u0027nexthop\u0027 argument to \u0027lr_route_del\u0027\n\nIn request https://review.opendev.org/c/openstack/ovsdbapp/+/882268, the ability to create ECMP routes was added. This PR adds the ability to delete these routes.\n\nCloses-bug: 2037536\n\nChange-Id: I84f234c6c5bbf4d318893166943de62fb87fff69\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/d30e80475048585273c460612a1655d828434c6f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/d30e80475048585273c460612a1655d828434c6f"}]},"branch":"refs/heads/master"},"61cf813855f3ab780158e8ef2d41848209641375":{"kind":"REWORK","_number":4,"created":"2023-10-01 07:18:03.000000000","uploader":{"_account_id":33871,"name":"Anton Vazhnetsov","display_name":"Anton Vazhnetsov","email":"dragen15051@gmail.com","username":"0x5b"},"ref":"refs/changes/45/896645/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/45/896645/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/45/896645/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/45/896645/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/45/896645/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/45/896645/4"}}},"commit":{"parents":[{"commit":"a39b4f8e62bcef958cc962adfad0103a3a41c1ae","subject":"Update master for stable/2023.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/a39b4f8e62bcef958cc962adfad0103a3a41c1ae"}]}],"author":{"name":"Anton Vazhnetsov","email":"dragen15051@gmail.com","date":"2023-09-26 17:01:59.000000000","tz":180},"committer":{"name":"Anton Vazhnetsov","email":"dragen15051@gmail.com","date":"2023-10-01 07:17:41.000000000","tz":180},"subject":"nb: add \u0027nexthop\u0027 argument to \u0027lr_route_del\u0027","message":"nb: add \u0027nexthop\u0027 argument to \u0027lr_route_del\u0027\n\nIn request https://review.opendev.org/c/openstack/ovsdbapp/+/882268, the ability to create ECMP routes was added. This PR adds the ability to delete these routes.\n\nCloses-bug: 2037536\n\nChange-Id: I84f234c6c5bbf4d318893166943de62fb87fff69\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/61cf813855f3ab780158e8ef2d41848209641375"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/61cf813855f3ab780158e8ef2d41848209641375"}]},"branch":"refs/heads/master"},"2e8575d56bd168aaba3c9481058eb89396ea45b2":{"kind":"REWORK","_number":5,"created":"2023-10-03 15:28:31.000000000","uploader":{"_account_id":33871,"name":"Anton Vazhnetsov","display_name":"Anton Vazhnetsov","email":"dragen15051@gmail.com","username":"0x5b"},"ref":"refs/changes/45/896645/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/45/896645/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/45/896645/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/45/896645/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/45/896645/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/45/896645/5"}}},"commit":{"parents":[{"commit":"a39b4f8e62bcef958cc962adfad0103a3a41c1ae","subject":"Update master for stable/2023.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/a39b4f8e62bcef958cc962adfad0103a3a41c1ae"}]}],"author":{"name":"Anton Vazhnetsov","email":"dragen15051@gmail.com","date":"2023-09-26 17:01:59.000000000","tz":180},"committer":{"name":"Anton Vazhnetsov","email":"dragen15051@gmail.com","date":"2023-10-03 15:27:45.000000000","tz":180},"subject":"nb: add \u0027nexthop\u0027 argument to \u0027lr_route_del\u0027","message":"nb: add \u0027nexthop\u0027 argument to \u0027lr_route_del\u0027\n\nIn request https://review.opendev.org/c/openstack/ovsdbapp/+/882268, the ability to create ECMP routes was added. This PR adds the ability to delete these routes.\n\nCloses-bug: 2037536\n\nChange-Id: I84f234c6c5bbf4d318893166943de62fb87fff69\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/2e8575d56bd168aaba3c9481058eb89396ea45b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/2e8575d56bd168aaba3c9481058eb89396ea45b2"}]},"branch":"refs/heads/master"},"033b6e3bd217c3e424e13f94893d1cb112abf87c":{"kind":"REWORK","_number":6,"created":"2023-10-04 15:09:58.000000000","uploader":{"_account_id":33871,"name":"Anton Vazhnetsov","display_name":"Anton Vazhnetsov","email":"dragen15051@gmail.com","username":"0x5b"},"ref":"refs/changes/45/896645/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/45/896645/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/45/896645/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/45/896645/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/45/896645/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/45/896645/6"}}},"commit":{"parents":[{"commit":"a39b4f8e62bcef958cc962adfad0103a3a41c1ae","subject":"Update master for stable/2023.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/a39b4f8e62bcef958cc962adfad0103a3a41c1ae"}]}],"author":{"name":"Anton Vazhnetsov","email":"dragen15051@gmail.com","date":"2023-09-26 17:01:59.000000000","tz":180},"committer":{"name":"Anton Vazhnetsov","email":"dragen15051@gmail.com","date":"2023-10-04 15:06:31.000000000","tz":180},"subject":"nb: add \u0027nexthop\u0027 argument to \u0027lr_route_del\u0027","message":"nb: add \u0027nexthop\u0027 argument to \u0027lr_route_del\u0027\n\nIn request https://review.opendev.org/c/openstack/ovsdbapp/+/882268, the ability to create ECMP routes was added. This PR adds the ability to delete these routes.\n\nCloses-bug: 2037536\n\nChange-Id: I84f234c6c5bbf4d318893166943de62fb87fff69\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/033b6e3bd217c3e424e13f94893d1cb112abf87c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/033b6e3bd217c3e424e13f94893d1cb112abf87c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
