)]}'
{"id":"openstack%2Fovsdbapp~896686","triplet_id":"openstack%2Fovsdbapp~master~I2a7a110551ee302dc9602ef6b01153262ea637d4","project":"openstack/ovsdbapp","branch":"master","topic":"gerrit/check-ic-learned-routes","attention_set":{},"removed_from_attention_set":{"27623":{"account":{"_account_id":27623,"name":"Zuul","email":"test.openstack.test@gmail.com","username":"openstack.test.test"},"last_update":"2023-11-16 19:18:52.000000000","reason":"Change was submitted"},"5756":{"account":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"last_update":"2023-11-16 17:51:02.000000000","reason":"\u003cGERRIT_ACCOUNT_5756\u003e replied on the change","reason_account":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}},"33871":{"account":{"_account_id":33871,"name":"Anton Vazhnetsov","display_name":"Anton Vazhnetsov","email":"dragen15051@gmail.com","username":"0x5b"},"last_update":"2023-11-16 19:18:52.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I2a7a110551ee302dc9602ef6b01153262ea637d4","subject":"nb: allow to add route even if learned route exists","status":"MERGED","created":"2023-09-27 16:30:53.000000000","updated":"2023-11-16 19:20:04.000000000","submitted":"2023-11-16 19:18:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"896686-gerrit/check-ic-learned-routes","meta_rev_id":"6d7aff31d3f128a9883392b2d360e9333c7c083a","_number":896686,"virtual_id_number":896686,"owner":{"_account_id":33871,"name":"Anton Vazhnetsov","display_name":"Anton Vazhnetsov","email":"dragen15051@gmail.com","username":"0x5b"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-16 19:18:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":33871,"name":"Anton Vazhnetsov","display_name":"Anton Vazhnetsov","email":"dragen15051@gmail.com","username":"0x5b"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":2,"date":"2023-11-15 20:38:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-11-16 17:51:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":33871,"name":"Anton Vazhnetsov","display_name":"Anton Vazhnetsov","email":"dragen15051@gmail.com","username":"0x5b"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"all":[{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-11-16 17:51:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":33871,"name":"Anton Vazhnetsov","display_name":"Anton Vazhnetsov","email":"dragen15051@gmail.com","username":"0x5b"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":33871,"name":"Anton Vazhnetsov","display_name":"Anton Vazhnetsov","email":"dragen15051@gmail.com","username":"0x5b"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":27623,"name":"Zuul","email":"test.openstack.test@gmail.com","username":"openstack.test.test"}],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27623,"name":"Zuul","email":"test.openstack.test@gmail.com","username":"openstack.test.test"},{"_account_id":33871,"name":"Anton Vazhnetsov","display_name":"Anton Vazhnetsov","email":"dragen15051@gmail.com","username":"0x5b"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-27 16:32:25.000000000","updated_by":{"_account_id":33871,"name":"Anton Vazhnetsov","display_name":"Anton Vazhnetsov","email":"dragen15051@gmail.com","username":"0x5b"},"reviewer":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"state":"REVIEWER"},{"updated":"2023-09-27 16:32:25.000000000","updated_by":{"_account_id":33871,"name":"Anton Vazhnetsov","display_name":"Anton Vazhnetsov","email":"dragen15051@gmail.com","username":"0x5b"},"reviewer":{"_account_id":27623,"name":"Zuul","email":"test.openstack.test@gmail.com","username":"openstack.test.test"},"state":"REVIEWER"},{"updated":"2023-09-27 17:47:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-13 18:12:11.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2023-11-15 20:38:52.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"83a711332354b6417eb8a1a3865ee94547634f9d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33871,"name":"Anton Vazhnetsov","display_name":"Anton Vazhnetsov","email":"dragen15051@gmail.com","username":"0x5b"},"date":"2023-09-27 16:30:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5fde11e217bcfab156326c88d3f2f040c00ab599","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-27 17:47:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/353098339bbe4efa8f27ad4a0e65f963\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2c8e96a500a94adeaf64dda3d6f11c5f : SUCCESS in 3m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c8207e818b904d65a7498dc78d97b15b : SUCCESS in 3m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/327032d975d948e5aad153a8e0647ebe : SUCCESS in 3m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c76988f7260a4511bb7cc6744e2c1816 : SUCCESS in 3m 36s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2c4b9fa2bf4f4e3ba7ad85a90eaec6bc : SUCCESS in 4m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d66c4551d665434496c5fc57432c7fb5 : SUCCESS in 5m 54s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/86ff0e3a0fa2426f9fd412659a39f4e4 : SUCCESS in 5m 39s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/9c6988fac6a64f8aa3217a70bb9292b9 : SUCCESS in 1h 10m 50s","accounts_in_message":[],"_revision_number":1},{"id":"e550bf27afe74ebefcb9b0657118d612a8a0f4eb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33871,"name":"Anton Vazhnetsov","display_name":"Anton Vazhnetsov","email":"dragen15051@gmail.com","username":"0x5b"},"date":"2023-10-11 20:05:38.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"dece95e614762393f2d13e39a712cb72884a6ea8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-11 21:25:09.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3f669c0534164408901d3d3bd120f961\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0dc5188ae1114803a410e0d0d9d461dc : SUCCESS in 3m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2a05553dab9b427081a6fa8db5a4b711 : SUCCESS in 3m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/57d296a9a4e047c99422c16c8a07c589 : SUCCESS in 3m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c0b8d8d5026c4da7812d99aba5b33c7f : SUCCESS in 3m 37s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4a5c211aa3fc499dbd847852dd3f5703 : SUCCESS in 3m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b3cec047013d43eaa78cc67b7c952343 : SUCCESS in 4m 57s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/4894808e0734411fb3db6ea4b95bf23a : SUCCESS in 5m 32s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/c4f0aefdcbf04991ab94da30a6dbbd88 : SUCCESS in 1h 11m 51s","accounts_in_message":[],"_revision_number":2},{"id":"1441db77d72f4b9efa3e241ba43389e453c77402","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2023-10-20 21:39:34.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"5c6cc4c19a09bcfbb4097522e7ec300d655b62be","author":{"_account_id":33871,"name":"Anton Vazhnetsov","display_name":"Anton Vazhnetsov","email":"dragen15051@gmail.com","username":"0x5b"},"date":"2023-11-03 12:04:06.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"55dfc9f1d1aa56f4f44dc823fca2329ba1b7c692","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-11-13 18:12:11.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"eaa127083e07f26d03e197c9fb30913ad4696fbb","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2023-11-15 14:45:39.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1df06fe08e9ebd0b63e58cee1fb1a300eb0067cd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33871,"name":"Anton Vazhnetsov","display_name":"Anton Vazhnetsov","email":"dragen15051@gmail.com","username":"0x5b"},"date":"2023-11-15 17:21:30.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"2cb5dee6a09d91648a02dc48286396886367fbd9","author":{"_account_id":33871,"name":"Anton Vazhnetsov","display_name":"Anton Vazhnetsov","email":"dragen15051@gmail.com","username":"0x5b"},"date":"2023-11-15 17:24:43.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"530c88c8d566edbdd7f14d04007ca9b6cd4478ca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-15 18:28:04.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f4d4005dcd364793baa186b4f0309c11\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e15f05e541f242fe81d3c6aed9c6f09c : SUCCESS in 4m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cf98b0305886494c94fc530f048aa45c : SUCCESS in 4m 13s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/37b4b21c7a104ac0807478ffd5418765 : SUCCESS in 3m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/facee192dac442c48ee765214e93f324 : SUCCESS in 5m 45s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/a5d0610333134028b4102fbdfe7b95fd : SUCCESS in 5m 33s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/2e1e5811dd8b41f2828df47122fea6ce : SUCCESS in 1h 05m 15s","accounts_in_message":[],"_revision_number":3},{"id":"634208908619f948251570d8b36607d55908b801","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-11-15 20:38:52.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"05fb538eb43d49ccf8ee01c5b18b7c3e305a9bbf","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2023-11-16 17:51:02.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"d4bddd220bc186f2d5853eb5395d4a30dffe11e3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-16 17:51:30.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"6b4d651527fb03ab4a93aedd0ef7851e8fe7c27d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-16 19:18:52.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e76467441ca846b8a25f873796977000\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/85bba9ca6bf841639fc62815e4b84244 : SUCCESS in 3m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/499afc0edf8d435abd7b773719297c47 : SUCCESS in 3m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2daee6fd3344489996141136df55c167 : SUCCESS in 3m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60a89b29538e489ab24777dba0eb35b2 : SUCCESS in 6m 39s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/9d3d54f28e2a4d48a7caa96ef04664bc : SUCCESS in 6m 28s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/e162382293884463925d3de5663c1892 : SUCCESS in 1h 21m 57s","accounts_in_message":[],"_revision_number":3},{"id":"7bb5c7baf8c44783002f1cacee944533315d4df1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-16 19:18:52.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"6d7aff31d3f128a9883392b2d360e9333c7c083a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-16 19:20:04.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/44f03c8b172a433ab7acff2c9084c9ab\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f720428baf1145b79d8f87c0148285fe : SUCCESS in 52s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"ef564d64aa0dec3eb230cccd5464de944b497eed","revisions":{"297c9f58c782c87e3cf30ee0cf45edfc064beae8":{"kind":"REWORK","_number":1,"created":"2023-09-27 16:30:53.000000000","uploader":{"_account_id":33871,"name":"Anton Vazhnetsov","display_name":"Anton Vazhnetsov","email":"dragen15051@gmail.com","username":"0x5b"},"ref":"refs/changes/86/896686/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/86/896686/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/86/896686/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/86/896686/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/86/896686/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/86/896686/1"}}},"commit":{"parents":[{"commit":"a39b4f8e62bcef958cc962adfad0103a3a41c1ae","subject":"Update master for stable/2023.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/a39b4f8e62bcef958cc962adfad0103a3a41c1ae"}]}],"author":{"name":"Anton Vazhnetsov","email":"AVazhnetsov@croc.ru","date":"2021-09-29 12:31:13.000000000","tz":180},"committer":{"name":"Anton Vazhnetsov","email":"dragen15051@gmail.com","date":"2023-09-27 16:25:03.000000000","tz":180},"subject":"nb: allow to add route even if learned route exists","message":"nb: allow to add route even if learned route exists\n\nAllow to install static route even if static route with same CIDR has\nbeen learned by ovn-ic from another AZ.\n\nSee this commit for more information: https://github.com/ovn-org/ovn/commit/68105f68c3f15fd1ae678f62b225730ba79a0745\n\nCloses-bug: 2037573\n\nChange-Id: I2a7a110551ee302dc9602ef6b01153262ea637d4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/297c9f58c782c87e3cf30ee0cf45edfc064beae8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/297c9f58c782c87e3cf30ee0cf45edfc064beae8"}]},"branch":"refs/heads/master"},"75e49f69a4af15693e493b96106f77c069820f9a":{"kind":"REWORK","_number":2,"created":"2023-10-11 20:05:38.000000000","uploader":{"_account_id":33871,"name":"Anton Vazhnetsov","display_name":"Anton Vazhnetsov","email":"dragen15051@gmail.com","username":"0x5b"},"ref":"refs/changes/86/896686/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/86/896686/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/86/896686/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/86/896686/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/86/896686/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/86/896686/2"}}},"commit":{"parents":[{"commit":"dae455b6eea3dcae931d07d4a81c03190e8259ed","subject":"nb: provide \u0027route_table\u0027 in lr-route-{add,del,list}","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/dae455b6eea3dcae931d07d4a81c03190e8259ed"}]}],"author":{"name":"Anton Vazhnetsov","email":"AVazhnetsov@croc.ru","date":"2021-09-29 12:31:13.000000000","tz":180},"committer":{"name":"Anton Vazhnetsov","email":"dragen15051@gmail.com","date":"2023-10-11 19:54:29.000000000","tz":180},"subject":"nb: allow to add route even if learned route exists","message":"nb: allow to add route even if learned route exists\n\nAllow to install static route even if static route with same CIDR has\nbeen learned by ovn-ic from another AZ.\n\nSee this commit for more information: https://github.com/ovn-org/ovn/commit/68105f68c3f15fd1ae678f62b225730ba79a0745\n\nCloses-bug: 2037573\n\nChange-Id: I2a7a110551ee302dc9602ef6b01153262ea637d4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/75e49f69a4af15693e493b96106f77c069820f9a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/75e49f69a4af15693e493b96106f77c069820f9a"}]},"branch":"refs/heads/master"},"ef564d64aa0dec3eb230cccd5464de944b497eed":{"kind":"REWORK","_number":3,"created":"2023-11-15 17:21:30.000000000","uploader":{"_account_id":33871,"name":"Anton Vazhnetsov","display_name":"Anton Vazhnetsov","email":"dragen15051@gmail.com","username":"0x5b"},"ref":"refs/changes/86/896686/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/86/896686/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/86/896686/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/86/896686/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/86/896686/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/86/896686/3"}}},"commit":{"parents":[{"commit":"dae455b6eea3dcae931d07d4a81c03190e8259ed","subject":"nb: provide \u0027route_table\u0027 in lr-route-{add,del,list}","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/dae455b6eea3dcae931d07d4a81c03190e8259ed"}]}],"author":{"name":"Anton Vazhnetsov","email":"AVazhnetsov@croc.ru","date":"2021-09-29 12:31:13.000000000","tz":180},"committer":{"name":"Anton Vazhnetsov","email":"dragen15051@gmail.com","date":"2023-11-15 17:20:40.000000000","tz":180},"subject":"nb: allow to add route even if learned route exists","message":"nb: allow to add route even if learned route exists\n\nAllow to install static route even if static route with same CIDR has\nbeen learned by ovn-ic from another AZ.\n\nSee this commit for more information: https://github.com/ovn-org/ovn/commit/68105f68c3f15fd1ae678f62b225730ba79a0745\n\nCloses-bug: 2037573\n\nChange-Id: I2a7a110551ee302dc9602ef6b01153262ea637d4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/ef564d64aa0dec3eb230cccd5464de944b497eed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/ef564d64aa0dec3eb230cccd5464de944b497eed"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
