)]}'
{"id":"openstack%2Fovsdbapp~896793","triplet_id":"openstack%2Fovsdbapp~master~Ic965ae098adb6db2e5a057eee74250e3b6331b01","project":"openstack/ovsdbapp","branch":"master","topic":"pr/provide-route-table-arg","attention_set":{},"removed_from_attention_set":{"27623":{"account":{"_account_id":27623,"name":"Zuul","email":"test.openstack.test@gmail.com","username":"openstack.test.test"},"last_update":"2023-10-11 15:33:30.000000000","reason":"Change was submitted"},"5756":{"account":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"last_update":"2023-10-04 22:56:23.000000000","reason":"\u003cGERRIT_ACCOUNT_5756\u003e replied on the change","reason_account":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}},"33871":{"account":{"_account_id":33871,"name":"Anton Vazhnetsov","display_name":"Anton Vazhnetsov","email":"dragen15051@gmail.com","username":"0x5b"},"last_update":"2023-10-11 15:33:30.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic965ae098adb6db2e5a057eee74250e3b6331b01","subject":"nb: provide \u0027route_table\u0027 in lr-route-{add,del,list}","status":"MERGED","created":"2023-09-28 13:43:34.000000000","updated":"2023-10-11 15:35:33.000000000","submitted":"2023-10-11 15:33:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":18,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"896793-pr/provide-route-table-arg","meta_rev_id":"d59bb5391aa97b4ca031c2fc0317e53aeed03628","_number":896793,"virtual_id_number":896793,"owner":{"_account_id":33871,"name":"Anton Vazhnetsov","display_name":"Anton Vazhnetsov","email":"dragen15051@gmail.com","username":"0x5b"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-11 15:33:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2023-10-09 09:46:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-10-04 22:56:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2023-10-09 09:46:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":27623,"name":"Zuul","email":"test.openstack.test@gmail.com","username":"openstack.test.test"}],"reviewers":{"REVIEWER":[{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27623,"name":"Zuul","email":"test.openstack.test@gmail.com","username":"openstack.test.test"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-28 13:45:14.000000000","updated_by":{"_account_id":33871,"name":"Anton Vazhnetsov","display_name":"Anton Vazhnetsov","email":"dragen15051@gmail.com","username":"0x5b"},"reviewer":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"state":"REVIEWER"},{"updated":"2023-09-28 13:45:14.000000000","updated_by":{"_account_id":33871,"name":"Anton Vazhnetsov","display_name":"Anton Vazhnetsov","email":"dragen15051@gmail.com","username":"0x5b"},"reviewer":{"_account_id":27623,"name":"Zuul","email":"test.openstack.test@gmail.com","username":"openstack.test.test"},"state":"REVIEWER"},{"updated":"2023-09-28 14:39:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-09 09:46:57.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"}],"messages":[{"id":"48e8fafdaa8ae1c26779d052261468cb98330e9e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33871,"name":"Anton Vazhnetsov","display_name":"Anton Vazhnetsov","email":"dragen15051@gmail.com","username":"0x5b"},"date":"2023-09-28 13:43:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2c327c7d1e4db6aac44e6db38514ff8769179de2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-28 14:39:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8c0505aeafd24636bf0a80b372f41cc6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/580803b4938c441aabad3f2364c302a5 : SUCCESS in 3m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7588418db80a4979a38383e0ba1930e9 : SUCCESS in 3m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1a08124bbf7448ad9f2699c7972178fb : SUCCESS in 3m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/89764cffd9ba4415bd4363f0f7308a6b : SUCCESS in 3m 36s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b8d646463909450cbd782774b3775cd8 : SUCCESS in 3m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a180325ea77485fa8295303e3520934 : SUCCESS in 6m 15s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/ce41b218b21a4b5bb31d5105456c2548 : SUCCESS in 5m 30s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/69eba811b3704c00b933f7200b7ab666 : SUCCESS in 53m 50s","accounts_in_message":[],"_revision_number":1},{"id":"ff32b468783c57bf743da763da534ab53089a356","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2023-09-28 19:08:24.000000000","message":"Patch Set 1:\n\n(6 comments)","accounts_in_message":[],"_revision_number":1},{"id":"f6e6170a059a2582220e1e8713a7f140f4963365","author":{"_account_id":33871,"name":"Anton Vazhnetsov","display_name":"Anton Vazhnetsov","email":"dragen15051@gmail.com","username":"0x5b"},"date":"2023-09-28 21:34:33.000000000","message":"Patch Set 1:\n\n(6 comments)","accounts_in_message":[],"_revision_number":1},{"id":"681675502da9e0a8fefef65495efeb65fe02f78c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33871,"name":"Anton Vazhnetsov","display_name":"Anton Vazhnetsov","email":"dragen15051@gmail.com","username":"0x5b"},"date":"2023-09-28 22:49:35.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"98ba05a14985d18120130cfa4ccd38eedd7421f2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-29 00:00:49.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3b222318c389474695e846121b24d9ea\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba70dc9dffe043e99411e21d63bfc048 : SUCCESS in 6m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fd32a7c5dff44b818839eeb8bcb9402b : SUCCESS in 2m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/408781a6ffd84575b88cb5822798fc90 : SUCCESS in 4m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/aa36ffb403eb42a69d72ff717d7cee08 : SUCCESS in 3m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7723273d00654239a7f2e93facee5b37 : SUCCESS in 3m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f196001e00bf4b8c886c774e12860d29 : SUCCESS in 8m 32s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/73d7ffebe99b4c3db2123f814cf271b2 : SUCCESS in 5m 49s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/b681e5e0ab9640cb885c82d7020bf8a9 : SUCCESS in 1h 10m 37s","accounts_in_message":[],"_revision_number":2},{"id":"e37430f416be187ff3a4730692f9e797a61b4b9b","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2023-09-29 22:17:11.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"732dac4e38df721073057559168b81b40438f2dc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33871,"name":"Anton Vazhnetsov","display_name":"Anton Vazhnetsov","email":"dragen15051@gmail.com","username":"0x5b"},"date":"2023-10-01 07:25:27.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"02ec180eab0b2062a419bb422e8c1e450a826ec4","author":{"_account_id":33871,"name":"Anton Vazhnetsov","display_name":"Anton Vazhnetsov","email":"dragen15051@gmail.com","username":"0x5b"},"date":"2023-10-01 07:44:45.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5a6fff14dc80c2ea96ad11b88a13d2d6a51bb9f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-01 08:08:15.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9e0f60ee0efb49bcaaca36119fcc2786\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/710803f8f6dc4e20a18dc86701c084ce : SUCCESS in 2m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8d5ea8c1f9af48a09b56e6048a4335f9 : SUCCESS in 3m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2a08d9b6ea4544a3a89ff2d4899ffc5a : SUCCESS in 2m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2e13db687f5f469a9adac6e779ce4034 : SUCCESS in 2m 54s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a1ef2bad6a0f4afba42073ae5c468bf5 : SUCCESS in 2m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1281f7f3b1b649cc8099c2592f718bd8 : SUCCESS in 5m 08s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/c32ab3f3f9b142ff9109995aa202b3d6 : SUCCESS in 5m 17s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/c95bf183b7664143917f34aef43143ea : SUCCESS in 41m 53s","accounts_in_message":[],"_revision_number":3},{"id":"d275689fb314ec73c2ddb807a55c62da9a952312","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2023-10-04 13:50:54.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"5d5df8f95a39677fbbd23eff3cc0890077690b27","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33871,"name":"Anton Vazhnetsov","display_name":"Anton Vazhnetsov","email":"dragen15051@gmail.com","username":"0x5b"},"date":"2023-10-04 15:19:35.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"b6ce23eddc3e48371802e640339702fc14043e2d","author":{"_account_id":33871,"name":"Anton Vazhnetsov","display_name":"Anton Vazhnetsov","email":"dragen15051@gmail.com","username":"0x5b"},"date":"2023-10-04 15:34:35.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"1995b0272e77e5d9b412bb7b5163b9abb81cd84e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-04 16:39:53.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/227c5b0ab9db4406b6c7bc88f390925a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/953fa698e8b347f389fed40901f7c8e1 : SUCCESS in 3m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9b787c086199407298b704afaa1686ba : SUCCESS in 3m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8f8e8e73c6e8446d9435605ced4e9c4c : SUCCESS in 3m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6e73586498d54bdebce06b3db0fced6b : SUCCESS in 3m 25s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d733eef2128f4a738c8df8ed70fe456d : SUCCESS in 3m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a4e46c42c9fe44088a47b71f3a976cf1 : SUCCESS in 4m 31s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/bbd7c4ea37554d50915ec989c821fd97 : SUCCESS in 5m 44s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/9ff2710b497d4b378a0dfb1bdc8fa1d1 : SUCCESS in 1h 14m 51s","accounts_in_message":[],"_revision_number":4},{"id":"2075f02b949a1ec4c421527a7d10365674ac37ce","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2023-10-04 22:56:23.000000000","message":"Patch Set 4: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"b843adb5ec961e629665e05df4205c974601124e","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-10-09 09:46:57.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"1f17945bffdd83ce487b88ab1657df2dbe78c887","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-11 14:12:40.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"121d74d612beae5091a8da47143cb11d01ef5712","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-11 15:33:30.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5c5bea8dd09b453f83032ec5b607000a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0cddb47e51924265a839c6b22b79fed5 : SUCCESS in 5m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a1f47ff2317a44afa3b8806dd067c3fc : SUCCESS in 3m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f85df4e85dbb4e1698ae369726daa606 : SUCCESS in 3m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/38a4830f33734186b9a3091369482312 : SUCCESS in 3m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e7bd51acd7046fb9f40734e549a21a5 : SUCCESS in 5m 22s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/af1d97797aa5487a95ca400a08d0b6be : SUCCESS in 5m 33s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/ca9e246d0b634f7988f59caf46952950 : SUCCESS in 1h 02m 16s","accounts_in_message":[],"_revision_number":4},{"id":"5576642a6be5460ed5edfe39086ef953b2e4e154","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-11 15:33:30.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"d59bb5391aa97b4ca031c2fc0317e53aeed03628","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-11 15:35:33.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/32bf45ddc5f946dc849d76008fd8b6c9\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bfc0841473454e0b97f67c4da80db6a9 : SUCCESS in 48s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"dae455b6eea3dcae931d07d4a81c03190e8259ed","revisions":{"1727c23eaadd721c629ff36a5dfa23aa1d7e42d2":{"kind":"REWORK","_number":1,"created":"2023-09-28 13:43:34.000000000","uploader":{"_account_id":33871,"name":"Anton Vazhnetsov","display_name":"Anton Vazhnetsov","email":"dragen15051@gmail.com","username":"0x5b"},"ref":"refs/changes/93/896793/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/93/896793/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/93/896793/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/93/896793/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/93/896793/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/93/896793/1"}}},"commit":{"parents":[{"commit":"d30e80475048585273c460612a1655d828434c6f","subject":"nb: add \u0027nexthop\u0027 argument to \u0027lr_route_del\u0027","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/d30e80475048585273c460612a1655d828434c6f"}]}],"author":{"name":"Anton Vazhnetsov","email":"AVazhnetsov@croc.ru","date":"2021-07-28 16:29:51.000000000","tz":180},"committer":{"name":"Anton Vazhnetsov","email":"dragen15051@gmail.com","date":"2023-09-28 13:36:31.000000000","tz":180},"subject":"nb: provide \u0027route_table\u0027 in lr-route-{add,del,list}","message":"nb: provide \u0027route_table\u0027 in lr-route-{add,del,list}\n\nAdds possibility to specify \u0027route_table\u0027 in methods for static\nroutes.\n\nSee this commit for more information:\nhttps://github.com/ovn-org/ovn/commit/1655a6c146cab4518ed5086df6c6f5e46cc35a2d\n\nCloses-bug: 2037652\n\nChange-Id: Ic965ae098adb6db2e5a057eee74250e3b6331b01\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/1727c23eaadd721c629ff36a5dfa23aa1d7e42d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/1727c23eaadd721c629ff36a5dfa23aa1d7e42d2"}]},"branch":"refs/heads/master"},"92cd6321118bbda9a4540290c2cd322121d67401":{"kind":"REWORK","_number":2,"created":"2023-09-28 22:49:35.000000000","uploader":{"_account_id":33871,"name":"Anton Vazhnetsov","display_name":"Anton Vazhnetsov","email":"dragen15051@gmail.com","username":"0x5b"},"ref":"refs/changes/93/896793/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/93/896793/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/93/896793/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/93/896793/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/93/896793/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/93/896793/2"}}},"commit":{"parents":[{"commit":"d30e80475048585273c460612a1655d828434c6f","subject":"nb: add \u0027nexthop\u0027 argument to \u0027lr_route_del\u0027","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/d30e80475048585273c460612a1655d828434c6f"}]}],"author":{"name":"Anton Vazhnetsov","email":"AVazhnetsov@croc.ru","date":"2021-07-28 16:29:51.000000000","tz":180},"committer":{"name":"Anton Vazhnetsov","email":"dragen15051@gmail.com","date":"2023-09-28 22:43:41.000000000","tz":180},"subject":"nb: provide \u0027route_table\u0027 in lr-route-{add,del,list}","message":"nb: provide \u0027route_table\u0027 in lr-route-{add,del,list}\n\nAdds possibility to specify \u0027route_table\u0027 in methods for static\nroutes.\n\nSee this commit for more information:\nhttps://github.com/ovn-org/ovn/commit/1655a6c146cab4518ed5086df6c6f5e46cc35a2d\n\nCloses-bug: 2037652\n\nChange-Id: Ic965ae098adb6db2e5a057eee74250e3b6331b01\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/92cd6321118bbda9a4540290c2cd322121d67401"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/92cd6321118bbda9a4540290c2cd322121d67401"}]},"branch":"refs/heads/master"},"ed39ea6204117a5434b86bf36bb1cad0d375c5f8":{"kind":"REWORK","_number":3,"created":"2023-10-01 07:25:27.000000000","uploader":{"_account_id":33871,"name":"Anton Vazhnetsov","display_name":"Anton Vazhnetsov","email":"dragen15051@gmail.com","username":"0x5b"},"ref":"refs/changes/93/896793/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/93/896793/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/93/896793/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/93/896793/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/93/896793/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/93/896793/3"}}},"commit":{"parents":[{"commit":"61cf813855f3ab780158e8ef2d41848209641375","subject":"nb: add \u0027nexthop\u0027 argument to \u0027lr_route_del\u0027","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/61cf813855f3ab780158e8ef2d41848209641375"}]}],"author":{"name":"Anton Vazhnetsov","email":"AVazhnetsov@croc.ru","date":"2021-07-28 16:29:51.000000000","tz":180},"committer":{"name":"Anton Vazhnetsov","email":"dragen15051@gmail.com","date":"2023-10-01 07:24:00.000000000","tz":180},"subject":"nb: provide \u0027route_table\u0027 in lr-route-{add,del,list}","message":"nb: provide \u0027route_table\u0027 in lr-route-{add,del,list}\n\nAdds possibility to specify \u0027route_table\u0027 in methods for static\nroutes.\n\nSee this commit for more information:\nhttps://github.com/ovn-org/ovn/commit/1655a6c146cab4518ed5086df6c6f5e46cc35a2d\n\nCloses-bug: 2037652\n\nChange-Id: Ic965ae098adb6db2e5a057eee74250e3b6331b01\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/ed39ea6204117a5434b86bf36bb1cad0d375c5f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/ed39ea6204117a5434b86bf36bb1cad0d375c5f8"}]},"branch":"refs/heads/master"},"dae455b6eea3dcae931d07d4a81c03190e8259ed":{"kind":"REWORK","_number":4,"created":"2023-10-04 15:19:35.000000000","uploader":{"_account_id":33871,"name":"Anton Vazhnetsov","display_name":"Anton Vazhnetsov","email":"dragen15051@gmail.com","username":"0x5b"},"ref":"refs/changes/93/896793/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/93/896793/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/93/896793/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/93/896793/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/93/896793/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/93/896793/4"}}},"commit":{"parents":[{"commit":"033b6e3bd217c3e424e13f94893d1cb112abf87c","subject":"nb: add \u0027nexthop\u0027 argument to \u0027lr_route_del\u0027","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/033b6e3bd217c3e424e13f94893d1cb112abf87c"}]}],"author":{"name":"Anton Vazhnetsov","email":"AVazhnetsov@croc.ru","date":"2021-07-28 16:29:51.000000000","tz":180},"committer":{"name":"Anton Vazhnetsov","email":"dragen15051@gmail.com","date":"2023-10-04 15:16:03.000000000","tz":180},"subject":"nb: provide \u0027route_table\u0027 in lr-route-{add,del,list}","message":"nb: provide \u0027route_table\u0027 in lr-route-{add,del,list}\n\nAdds possibility to specify \u0027route_table\u0027 in methods for static\nroutes.\n\nSee this commit for more information:\nhttps://github.com/ovn-org/ovn/commit/1655a6c146cab4518ed5086df6c6f5e46cc35a2d\n\nCloses-bug: 2037652\n\nChange-Id: Ic965ae098adb6db2e5a057eee74250e3b6331b01\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/dae455b6eea3dcae931d07d4a81c03190e8259ed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/dae455b6eea3dcae931d07d4a81c03190e8259ed"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
