)]}'
{"id":"openstack%2Fovsdbapp~905044","triplet_id":"openstack%2Fovsdbapp~master~I6bfd486bf037da5b4d2dd61c84a50d6a7fbfb39d","project":"openstack/ovsdbapp","branch":"master","topic":"bug/1982287","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2024-01-12 10:18:58.000000000","reason":"Change was submitted"},"6773":{"account":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"last_update":"2024-01-12 09:00:40.000000000","reason":"\u003cGERRIT_ACCOUNT_6773\u003e replied on the change","reason_account":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"}},"28056":{"account":{"_account_id":28056,"name":"Liushy","email":"liuxie_11@163.com","username":"liuxie_11"},"last_update":"2024-01-12 10:18:58.000000000","reason":"Change was submitted"},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2024-01-12 10:18:58.000000000","reason":"Change was submitted"},"5756":{"account":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"last_update":"2024-01-11 15:24:25.000000000","reason":"\u003cGERRIT_ACCOUNT_5756\u003e replied on the change","reason_account":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}}},"hashtags":[],"change_id":"I6bfd486bf037da5b4d2dd61c84a50d6a7fbfb39d","subject":"Support add/remove CIDR type address in Address_Set cmds","status":"MERGED","created":"2024-01-09 03:45:37.000000000","updated":"2024-01-12 10:19:58.000000000","submitted":"2024-01-12 10:18:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"905044-bug/1982287","meta_rev_id":"27fa4a9ad19a6d9bc7dde10b993bd22a7f36d722","_number":905044,"virtual_id_number":905044,"owner":{"_account_id":28056,"name":"Liushy","email":"liuxie_11@163.com","username":"liuxie_11"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":28056,"name":"Liushy","email":"liuxie_11@163.com","username":"liuxie_11"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-01-12 10:18:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":34125,"name":"huanghailun","display_name":"huanghailun","email":"hailun.huang@easystack.cn","username":"lnun"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2024-01-12 09:00:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":28056,"name":"Liushy","email":"liuxie_11@163.com","username":"liuxie_11"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-01-11 15:24:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":1,"date":"2024-01-12 08:11:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":34125,"name":"huanghailun","display_name":"huanghailun","email":"hailun.huang@easystack.cn","username":"lnun"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2024-01-12 09:00:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":28056,"name":"Liushy","email":"liuxie_11@163.com","username":"liuxie_11"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":34125,"name":"huanghailun","display_name":"huanghailun","email":"hailun.huang@easystack.cn","username":"lnun"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":28056,"name":"Liushy","email":"liuxie_11@163.com","username":"liuxie_11"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":34125,"name":"huanghailun","display_name":"huanghailun","email":"hailun.huang@easystack.cn","username":"lnun"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28056,"name":"Liushy","email":"liuxie_11@163.com","username":"liuxie_11"},{"_account_id":34125,"name":"huanghailun","display_name":"huanghailun","email":"hailun.huang@easystack.cn","username":"lnun"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-09 04:50:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-01-09 07:37:06.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2024-01-09 10:02:27.000000000","updated_by":{"_account_id":28056,"name":"Liushy","email":"liuxie_11@163.com","username":"liuxie_11"},"reviewer":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"state":"REVIEWER"},{"updated":"2024-01-09 12:43:45.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2024-01-10 18:33:02.000000000","updated_by":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"reviewer":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"state":"CC"},{"updated":"2024-01-11 15:24:25.000000000","updated_by":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"reviewer":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"state":"REVIEWER"},{"updated":"2024-01-12 08:11:35.000000000","updated_by":{"_account_id":34125,"name":"huanghailun","display_name":"huanghailun","email":"hailun.huang@easystack.cn","username":"lnun"},"reviewer":{"_account_id":34125,"name":"huanghailun","display_name":"huanghailun","email":"hailun.huang@easystack.cn","username":"lnun"},"state":"REVIEWER"}],"messages":[{"id":"7e6153ef869eec20edcc5384b35b1ea3f3b43503","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28056,"name":"Liushy","email":"liuxie_11@163.com","username":"liuxie_11"},"date":"2024-01-09 03:45:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a5141fab61d1d2e57ad1a4bb4cd96d2ddc23c2bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-09 04:50:47.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d647e97467cc45ff9d32fa5ea38cde95\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/818e9f2f472646fca64ebcd31f69f865 : SUCCESS in 3m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/36156486196f4b50a9a7995dacd2e342 : SUCCESS in 3m 21s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3ffd6ab7257844eb8a37ddf13964ccbc : SUCCESS in 5m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/23e12177fe62434a83a9645c4f779315 : SUCCESS in 5m 49s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/0f60cf55a24342e5aac6996ed7707694 : FAILURE in 5m 15s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/199339c136064ec58a11febaba59ff96 : SUCCESS in 1h 03m 22s","accounts_in_message":[],"_revision_number":1},{"id":"ff1cc0eb07cda42dded1dfb0300d2b824db26c04","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28056,"name":"Liushy","email":"liuxie_11@163.com","username":"liuxie_11"},"date":"2024-01-09 06:14:26.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"dd3e7c1d0a509063beff6dc498aa71309f3b5df1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-09 07:32:35.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cde281e55f72490e9aeefb45e5784c95\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bee2a9960a804a10b17ccbe258ca3833 : SUCCESS in 3m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dd10dce5d6d1449bb3fc8c3fd9230a77 : SUCCESS in 3m 10s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7644f3cf5f4d41d1ab7165c97aa7fdca : SUCCESS in 3m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7782026b9dae45d8b55a52773864c41d : SUCCESS in 4m 57s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/265bf34e2ce145b4ba6c473ff43fd382 : SUCCESS in 5m 14s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/35e986cba364460586489756d0222d7a : SUCCESS in 1h 12m 12s","accounts_in_message":[],"_revision_number":2},{"id":"0a6499cc575d23a254a88bf925e1ad1c9ae1a643","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-01-09 07:37:06.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3b90788872ec442a83155b9a5f6163e2c9dca224","author":{"_account_id":28056,"name":"Liushy","email":"liuxie_11@163.com","username":"liuxie_11"},"date":"2024-01-09 08:13:53.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8a62bac656fd49645af71a75d08b1802e2e398a2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28056,"name":"Liushy","email":"liuxie_11@163.com","username":"liuxie_11"},"date":"2024-01-09 09:52:53.000000000","message":"Uploaded patch set 3: Commit message was updated.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"549a8ec0a2971cd8a4f73b3cab06bbb1e164ccfe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-09 11:12:22.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cf017c1d51bc4de69cda53e8a8668273\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8d52cf5b08c440f9ab4f910d9bb83ec0 : SUCCESS in 3m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e3595b7110254a7bb8f303785ba20b12 : SUCCESS in 2m 59s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/840fd6d239d244debeef66ffe6eb54d0 : SUCCESS in 3m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9481d8b9f9374502b4b835f47a484253 : SUCCESS in 5m 16s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/00dceab28c7542bdb6d19ebd14177099 : SUCCESS in 5m 04s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/273207601ab24eb9ab9ba697236eb4b5 : SUCCESS in 1h 02m 28s","accounts_in_message":[],"_revision_number":3},{"id":"92ec27d84c58111bffda34372277c7902baf2ccd","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-01-09 12:43:45.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"55fe54d94ae8eea6d2109b6af649e4e0e3f1ba14","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2024-01-09 14:44:58.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"a18f55ebc00f245876601fb1c9cde39a352f389d","author":{"_account_id":28056,"name":"Liushy","email":"liuxie_11@163.com","username":"liuxie_11"},"date":"2024-01-10 03:30:25.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"2b20716edef4e0103ed570f61984c29e7ce00d52","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2024-01-10 18:33:02.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"01a0e9ffff590a465b160d8827d46154e7ec8a36","author":{"_account_id":28056,"name":"Liushy","email":"liuxie_11@163.com","username":"liuxie_11"},"date":"2024-01-11 03:27:31.000000000","message":"Patch Set 3: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"f52a524c19021257274f7e22454a0d5decd776cf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28056,"name":"Liushy","email":"liuxie_11@163.com","username":"liuxie_11"},"date":"2024-01-11 03:57:08.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"c7f7e265f4a487500de2396f88cdc67094d942aa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-11 04:58:56.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cb2ae1240ced4d1d83f1c7856fad0607\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/54e4a974e1a3461f9826a2fa1c28731c : SUCCESS in 4m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/63139a5adb29499cbe1fde6f51d82b0e : SUCCESS in 3m 09s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e32dead81f03410286fdfe0df95fcf0d : SUCCESS in 3m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83afe40ea829457098a634305f1e38e9 : SUCCESS in 5m 11s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/5cfb3ad96d1847699bcc24bc20398d90 : SUCCESS in 6m 11s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/fcc1a098bb494cff8617be7885f55ac9 : SUCCESS in 1h 00m 10s","accounts_in_message":[],"_revision_number":4},{"id":"f5d426c135e5d9d3886663989b5e032553aa8db8","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2024-01-11 15:24:25.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"00f2a214029ad28e936b9eb37c919464ecb48c94","author":{"_account_id":34125,"name":"huanghailun","display_name":"huanghailun","email":"hailun.huang@easystack.cn","username":"lnun"},"date":"2024-01-12 08:11:35.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"d8db6f51afeb859e16a0238530bc81b969863f95","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2024-01-12 09:00:40.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"9999d09bd4b23c223ec85268d411733c1a0f0dbc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-12 09:01:08.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"95a893140138305c5bb88e009ab1315e8e7931fe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-12 10:18:57.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8d741783ef0a4fcea2fe196c03731667\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a14acba41ab04d748ba31142b6efeaca : SUCCESS in 3m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/85bbee8a81b64900b01939d5383825a3 : SUCCESS in 3m 15s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/30399d11f9b340a58e2ed29465755f7d : SUCCESS in 3m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7fac6dec8b244c798cf3b819ee924a7c : SUCCESS in 4m 35s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/81838fd9eb3742f6b76e6e8b39ecea29 : SUCCESS in 4m 56s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/cdc3f790e969411595fb87187ec2ebc7 : SUCCESS in 1h 12m 39s","accounts_in_message":[],"_revision_number":4},{"id":"1f808c5e1c2744225c05e20a3098adf053de860f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-12 10:18:58.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"27fa4a9ad19a6d9bc7dde10b993bd22a7f36d722","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-12 10:19:58.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c9b32a301d6b4b1391f9f856f79a92de\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/43a6577b46c745ad9b6f72b25a98a3b7 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"374bf64068418e5111f6e2f3e2bf1bd4e3d84b15","revisions":{"fc8df55357daee8b31d1c8307b4f172dc3c2478b":{"kind":"REWORK","_number":1,"created":"2024-01-09 03:45:37.000000000","uploader":{"_account_id":28056,"name":"Liushy","email":"liuxie_11@163.com","username":"liuxie_11"},"ref":"refs/changes/44/905044/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/44/905044/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/44/905044/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/44/905044/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/44/905044/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/44/905044/1"}}},"commit":{"parents":[{"commit":"1cb9acc015414ba1eda822fcce886b678117c554","subject":"Merge \"add Port_Binding to lookup_table\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/1cb9acc015414ba1eda822fcce886b678117c554"}]}],"author":{"name":"liushy","email":"liuxie_11@163.com","date":"2024-01-09 03:27:51.000000000","tz":480},"committer":{"name":"liushy","email":"liuxie_11@163.com","date":"2024-01-09 03:45:30.000000000","tz":480},"subject":"Support add/remove CIDR type address in Address_Set cmds","message":"Support add/remove CIDR type address in Address_Set cmds\n\nCloses-Bug: #1982287\nChange-Id: I6bfd486bf037da5b4d2dd61c84a50d6a7fbfb39d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/fc8df55357daee8b31d1c8307b4f172dc3c2478b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/fc8df55357daee8b31d1c8307b4f172dc3c2478b"}]},"branch":"refs/heads/master"},"81b03e3f8ee87437921b7510bb0eb4131401d9a2":{"kind":"REWORK","_number":2,"created":"2024-01-09 06:14:26.000000000","uploader":{"_account_id":28056,"name":"Liushy","email":"liuxie_11@163.com","username":"liuxie_11"},"ref":"refs/changes/44/905044/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/44/905044/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/44/905044/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/44/905044/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/44/905044/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/44/905044/2"}}},"commit":{"parents":[{"commit":"1cb9acc015414ba1eda822fcce886b678117c554","subject":"Merge \"add Port_Binding to lookup_table\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/1cb9acc015414ba1eda822fcce886b678117c554"}]}],"author":{"name":"liushy","email":"liuxie_11@163.com","date":"2024-01-09 03:27:51.000000000","tz":480},"committer":{"name":"liushy","email":"liuxie_11@163.com","date":"2024-01-09 06:14:03.000000000","tz":480},"subject":"Support add/remove CIDR type address in Address_Set cmds","message":"Support add/remove CIDR type address in Address_Set cmds\n\nCloses-Bug: #1982287\nChange-Id: I6bfd486bf037da5b4d2dd61c84a50d6a7fbfb39d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/81b03e3f8ee87437921b7510bb0eb4131401d9a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/81b03e3f8ee87437921b7510bb0eb4131401d9a2"}]},"branch":"refs/heads/master"},"8e957c84bb29e0285a10b96c32003f6bcd514789":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2024-01-09 09:52:53.000000000","uploader":{"_account_id":28056,"name":"Liushy","email":"liuxie_11@163.com","username":"liuxie_11"},"ref":"refs/changes/44/905044/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/44/905044/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/44/905044/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/44/905044/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/44/905044/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/44/905044/3"}}},"commit":{"parents":[{"commit":"1cb9acc015414ba1eda822fcce886b678117c554","subject":"Merge \"add Port_Binding to lookup_table\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/1cb9acc015414ba1eda822fcce886b678117c554"}]}],"author":{"name":"liushy","email":"liuxie_11@163.com","date":"2024-01-09 03:27:51.000000000","tz":480},"committer":{"name":"liushy","email":"liuxie_11@163.com","date":"2024-01-09 09:51:43.000000000","tz":480},"subject":"Support add/remove CIDR type address in Address_Set cmds","message":"Support add/remove CIDR type address in Address_Set cmds\n\nRelated-Bug: #1982287\nChange-Id: I6bfd486bf037da5b4d2dd61c84a50d6a7fbfb39d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/8e957c84bb29e0285a10b96c32003f6bcd514789"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/8e957c84bb29e0285a10b96c32003f6bcd514789"}]},"branch":"refs/heads/master"},"374bf64068418e5111f6e2f3e2bf1bd4e3d84b15":{"kind":"REWORK","_number":4,"created":"2024-01-11 03:57:08.000000000","uploader":{"_account_id":28056,"name":"Liushy","email":"liuxie_11@163.com","username":"liuxie_11"},"ref":"refs/changes/44/905044/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/44/905044/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/44/905044/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/44/905044/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/44/905044/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/44/905044/4"}}},"commit":{"parents":[{"commit":"1cb9acc015414ba1eda822fcce886b678117c554","subject":"Merge \"add Port_Binding to lookup_table\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/1cb9acc015414ba1eda822fcce886b678117c554"}]}],"author":{"name":"liushy","email":"liuxie_11@163.com","date":"2024-01-09 03:27:51.000000000","tz":480},"committer":{"name":"liushy","email":"liuxie_11@163.com","date":"2024-01-11 03:55:14.000000000","tz":480},"subject":"Support add/remove CIDR type address in Address_Set cmds","message":"Support add/remove CIDR type address in Address_Set cmds\n\nRelated-Bug: #1982287\nChange-Id: I6bfd486bf037da5b4d2dd61c84a50d6a7fbfb39d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/374bf64068418e5111f6e2f3e2bf1bd4e3d84b15"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/374bf64068418e5111f6e2f3e2bf1bd4e3d84b15"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
